)]}'
{"id":"openstack%2Fkolla~858312","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I46f03329853a2025bff20b7b60e35c06aba619eb","project":"openstack/kolla","branch":"stable/ussuri","topic":"zuul-queue-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-19 16:01:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46f03329853a2025bff20b7b60e35c06aba619eb","subject":"[CI] Move queue setting to project level","status":"MERGED","created":"2022-09-19 11:56:16.000000000","updated":"2022-09-19 16:04:01.000000000","submitted":"2022-09-19 16:01:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858312-zuul-queue-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":858311,"cherry_pick_of_patch_set":1,"meta_rev_id":"381c12b292c901cbc8a8157477704bd52279ec23","_number":858312,"virtual_id_number":858312,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 16:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-09-19 15:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-09-19 11:56:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-09-19 11:56:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 13:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 14:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 15:01:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7703cb9574d3b21a3a254be2a038aa3f61a84dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-19 11:56:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"f2ce539e210233d2ad30e89ba7eda7f3b4b158eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-19 11:56:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8feb6e2a04f51fd67120982bce6fada22307a8ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 13:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5297dfc71f3f49d8bb47913a06d829da\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30f03efe0d1f4f7a9b797eae152ba51d : SUCCESS in 1h 21m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd4ed5778e6846dfa3200892dc86d75a : FAILURE in 1h 05m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13706c608e24a342b5a4da8a43c71946bf3f977a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 14:44:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b390ed834b486ba0c91f9c71f74211\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1dfa4d5a97ea45ee945f963703725c54 : SUCCESS in 1h 12m 09s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7045a6a278a947468f0bb1b4e7f31d37 : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316c695300f64a49bcc2b9ad2aa0775d : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/049d139050ea45f19fb5afedacb808e0 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fba5e3c8285b411daf791707cd501c80 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c6daa11c744ef1ae811a396e8bd45f : SUCCESS in 6m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bab8eb83d69543629f5ff1e9e42a1ec6 : SUCCESS in 26m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bbeebc260714fcbafd8332daa339805 : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7a3c631e7464f499cc4bd09ad36955c : SUCCESS in 59m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db5d8e8ffd164154bb82d00e2e5593b6 : SUCCESS in 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62f207eaa2133273d65052efe49426ec935c784e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 14:44:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"349903d3dc1fc1cb89bf7e557c8bb332ae186609","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-19 15:01:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d275d967f78914f696e1ee323dba64ef3575a53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:01:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390d830daefe49538118659dd449e316\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/129d4d0937b74b8fa3d6166afb58b199 : SUCCESS in 1h 09m 19s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5495e67509374ea0a0359afd111a2fc9 : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ed6b70744444fc863325108543ddbd : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3043eae0b064a639ce566074b3315e8 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7ceaee9fd9a45b08ce6cacd58e9f582 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4352ac95b63f44eaa1e4107d1e52a9ba : SUCCESS in 5m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1eebb7a3f074cef86175821a469fb93 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70be52f2ab454036b65b81a087dfa660 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ad78d5573b20a9c9269c5ee53f6b50cf50d05cd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:01:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"381c12b292c901cbc8a8157477704bd52279ec23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:04:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddd144ae085409a9add68b1c1d6d308\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63902c48e4594a24ba268d190ceb3d51 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9107c5f9ed360176909cc7322faf2d3a9464c675","revisions":{"9107c5f9ed360176909cc7322faf2d3a9464c675":{"kind":"REWORK","_number":1,"created":"2022-09-19 11:56:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/858312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/858312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/858312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/858312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/858312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/858312/1"}}},"commit":{"parents":[{"commit":"89c1391619fd0983676b819d55dfb8f3d9b8eeba","subject":"mariadb: Fix repository URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89c1391619fd0983676b819d55dfb8f3d9b8eeba"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 07:49:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-19 11:56:16.000000000","tz":0},"subject":"[CI] Move queue setting to project level","message":"[CI] Move queue setting to project level\n\nPer [1].\n\n[1] http://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I46f03329853a2025bff20b7b60e35c06aba619eb\n(cherry picked from commit 3967e5c24bd3de931ee3e3b19f98c72edf4d1e61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9107c5f9ed360176909cc7322faf2d3a9464c675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9107c5f9ed360176909cc7322faf2d3a9464c675"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
