)]}'
{"id":"openstack%2Fkolla~859400","triplet_id":"openstack%2Fkolla~master~I7af51d07b8a939c908aa072294061def57dd67de","project":"openstack/kolla","branch":"master","topic":"rl9","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-30 20:50:45.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-09-29 08:39:03.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-30 19:31:14.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I7af51d07b8a939c908aa072294061def57dd67de","subject":"Add RockyLinux 9 support","status":"MERGED","created":"2022-09-27 10:18:35.000000000","updated":"2022-09-30 20:53:09.000000000","submitted":"2022-09-30 20:50:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"859400-rl9","meta_rev_id":"a7a4185ba390aed1a868f6fb31ebb644d05eb032","_number":859400,"virtual_id_number":859400,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-30 20:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-09-30 14:40:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-09-30 19:31:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-09-30 19:31:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 10:18:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-27 10:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-27 10:23:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-09-27 12:20:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-09-29 08:43:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"01c4bfab137fd4f834353d81d4cbc4d88d06aaf7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 10:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a021db19699c1fb21a6af1d68537875c768002a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 10:18:55.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-rockylinux-9-aarch64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-rocky9-aarch64\n      parent: kolla-build-rocky9\n      nodeset: kolla-rockylinux-9-aarch64\n      voting: false\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 30, column 3","accounts_in_message":[],"_revision_number":1},{"id":"4e1f20e136bb7a39db8530f49bc46c1d4a60bb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 10:18:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-rockylinux-9-aarch64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-rocky9-aarch64\n      parent: kolla-build-rocky9\n      nodeset: kolla-rockylinux-9-aarch64\n      voting: false\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 30, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3841a2f6cbba2354348cb760a692b81c9facf575","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 10:20:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de5361249612a7da620527ff2ae7b21b1006e9dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 10:22:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fb7c30d9b47a31420fab26abd69358537bce34a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-27 10:23:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2af19e173dd889054f56824d6c9bb8d32c7b2d34","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-27 10:26:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f49f522446aaf28c6d6031362493868e1ba6dbf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 10:32:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2dac328ca20b5e5d8ea0bc6b1059e501cd9c506e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 10:50:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0f21fe4f1f3fd083b924bca05a76e94878f2ee40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:01:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd639f86da747599db9b3ac0de991cc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/df114968fd7e45939a828c2e752212a9 : FAILURE in 8m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57f9c3c6a31c43509f42d342d856ee9b : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/588c990ccb3d4b63bf7875b6f3a52c51 : SUCCESS in 6m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dd7ee02a888c4fc9bcd7e80247bbfe84 : FAILURE in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed3199db4efe9410d68a046df0c9e5785d10fa64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:02:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b4d78eb6d7a4f9982c10bdbd9f8f443\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/132d7881f71944c1b95c9384643baf31 : FAILURE in 4m 40s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3c9837f57f3945bcb17476df3182c2b3 : FAILURE in 6m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75712fcffa14509a03ece0162347af1 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220b4676121042fe8ad234808e05df8f : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8727e2c6061f40f69805cefaa62ff10f : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2ee4232cf0a4670b94db3f55cc6953e : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b36f0087a44d368e384747676b5886 : SUCCESS in 5m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7721e46f4654700af3219b2f7c72a59 : FAILURE in 3m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b188f2068ca94a91a2020b46bc18c0db : FAILURE in 5m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2a2b9808c3c44c4288a9fd5d6f899da4 : FAILURE in 2m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4de5d812c3a14d949a431dcc40b9cf3b : FAILURE in 5m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64697c0614f14d2a867878ead2a60805 : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"81d3a1534eaabe4a8df939ddafecc65779437625","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 11:38:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76ea3132f2a5cff72f3cbf46b0fb6da413a39110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:48:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be70f01ddde43acae6895d0517b8eec\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/46422a724bcd44f5a58425915630cce8 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7256fef582bc4bffa3aeb9dc2cc5882d : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f533767b0d245828cc3f518b5c8e6f7 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/04636f278a0d462790e57cadf98677b7 : FAILURE in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"95b2fa99f6a47fc5ac55429d5e39a38c73384c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:51:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20c46614390a4f9390a4c9299c662601\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c7d895f47fc043c287f101380e1f3a85 : FAILURE in 4m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/882f43f7ec3648e3b09111b6266ff309 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f794e9a9a047079a3aba2d1f774b01 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60f056d89f4475b901e769b73ea4c32 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/132f086ed0674360b1f851e5d4d7ba82 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97f3e01e8eb24a2498390d4d0bdde947 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce4571770a84a59823837dddc7fe07f : SUCCESS in 6m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f14afad90fa5436db1d59e312ce93c21 : FAILURE in 4m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3f446736a6c41338b1e23a4be156aca : FAILURE in 5m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/73549fda4cc04271b8b309d8f20bcdf0 : FAILURE in 3m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a64f16605e6e4f2997e14e9d7afda508 : FAILURE in 6m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d4bb33adc4843429f24340d7ee7768b : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":6},{"id":"bce8b542105b8567c93d0c8fe6e3aad928e67479","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 12:20:49.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f53429a097bb003d2724398118323fb6a350a087","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 15:03:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4a2164d837e99547fef3c7191458f5c79acbd061","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 15:33:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f2447bc1591fa04e73b1bdb766f3543565d0db25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-27 15:43:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d0c7b2fc74d23b12dc1921a8e9ec7df855a932b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 17:14:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4215f22ef654bbb9a9891548cb98844\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/82a05228dfac40f9bd780acdf7ba59b3 : SUCCESS in 27m 26s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/42537c395ceb40638d0ceaaa68d8a295 : SUCCESS in 1h 18m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5deecc17c87417dae6bf38b22064e08 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd697bfbdb0454bbccb9f29bc811f74 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc548ab9ec5b409bb1244fb91daee528 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5b115b69f7142cb810f98985c6f726b : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f130c0da263347cab6b1b31e29fc5609 : SUCCESS in 6m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/61f72b66e9f44dbd960b2084130657be : FAILURE in 30m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/12ef8842513a45829f1c3ec4bf988da1 : SUCCESS in 1h 25m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b13a890be5a54009a1872ccb69525413 : SUCCESS in 27m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/94c39cf94c5c44a2867cfceff25d019f : SUCCESS in 1h 00m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d48179e06dc4dfe8220f994509867bf : SUCCESS in 1h 10m 04s","accounts_in_message":[],"_revision_number":9},{"id":"8ad0a92f13b29c93eb0d432a87670f606ec8438a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 17:28:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c43a5b188d49089611b4051c91e22b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0dc4d639cb0e462e9fb7b66770c416ca : FAILURE in 1h 41m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbda2189b8c3420db027d182ab0ae35a : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e96fb435a11455ca7c629279b2e0e87 : SUCCESS in 8m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d63052480b8d4afda3dcee36fbf10784 : SUCCESS in 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bde655e8fd43c275bc0e6787e27e14905664eb62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 07:19:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a4d455978d4baa41a3c053ad353339e6c1cbf195","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 07:51:00.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"488195cdc27739efad229a25920af36c9d8dbc18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 08:31:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe1a4f2eac944c292a8a2558193da6f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4e2e51854aad4f0eb782bae5499b53ac : FAILURE in 1h 08m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a773a7427faa471a90e9484e2a353ee2 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c56c82da6a294187b6e6ad878e6f641a : SUCCESS in 8m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2bd75afb72d74397a98e99cb558bee3c : SUCCESS in 54m 36s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c38b9bd049af0f86a245196b842f5d4f54fdd719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 08:51:28.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/239965fb899242009d6ff0f8cbe9d75b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/be25009bebbb4d3ca1244334cee52987 : SUCCESS in 21m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/9faa2f4bc4624d19b0d9e41a9b142941 : SUCCESS in 1h 18m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23a7262b93a4d3faf67f31bf339f459 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab7fd7eec983465ba7bd583ff27ce30c : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddd975df093449afa49f59b1d1da6b83 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e21ee822ad14ddb9237ffbd3fe823ba : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978e84241b9a4d879f4d79f8b6b26ac0 : SUCCESS in 7m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5fa7a0e5d824a01932660d84cfd70ca : FAILURE in 25m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a390cce9143d4206940b85a5cd39554c : SUCCESS in 53m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8a06901aaea1483488a4d579c075c03a : SUCCESS in 20m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4dc1eb121363452bbeee7ba26beb2ddd : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d881ed764a24bb78bd8b4911d367a93 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":10},{"id":"6267e48f261bedb5845d02790be762324a7d34d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 09:05:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"34df4567044760df953e0da6d9b3a405f1eda7b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 09:18:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"158452ac01effe76aa8a6d8404b955ce133f46bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 09:20:05.000000000","message":"Patch Set 12: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d09a759592ce7d7df25307b1de5b07f5d3d7ff81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 09:20:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"286a1a489bf89642d9c245d1edb1d4be294fdf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 10:36:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f3fb4a12ef54e58829035de4881968a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4522acff9aa540e7b7894058792e5a2a : FAILURE in 18m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/735e8d7a9106480caa592b7927fe7160 : FAILURE in 16m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2825d9642ce3470391d4ee2f927a6ab0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19de908eccc04acd97d4745f6f83fd77 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4b0dd1e6d094eb8b45c0f6c7bbb117f : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6bfe43bb9ff4b5d8b8fcf96110bbc2d : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1187d00eebf44c82b7ed66bc36cc0c6a : SUCCESS in 6m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/65389905eb4b4076a554523cf5773a43 : FAILURE in 17m 25s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2eec6e35244f4936b054305dea996c69 : FAILURE in 17m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40da943ff6de45ab84a40e30606f4e48 : SUCCESS in 19m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee0f51ebc5a44c72947e2d07bb5bad2e : SUCCESS in 1h 02m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/225bd341f4ef4f4091c097902f3aef28 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":13},{"id":"9fb47ff3761a474a6392038aded27c568349501b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 10:40:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d583dbee728b4ac1912a8b339061383d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/963caa06f5e64e1f9505e575741f63ce : FAILURE in 26m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7450d9d074d34ab79cdb44e235412d73 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a22c26c800e41ae915c019d3733e52d : SUCCESS in 5m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7047f8155e244b1594e109f605ee4deb : SUCCESS in 1h 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0143750f4df8ef950c97e356b4f22c853ec70a2d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 10:41:36.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2d8645be4b149aeec3d0c8b4a62f0187f606acfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 11:56:24.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cffad4d9ef4145df9ff0ffe728bdddf1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dd45bd3855ad497cb46caa83aaea065b : SUCCESS in 25m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/1aaab048134043bdb2aae9011c9933dc : SUCCESS in 1h 00m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4087b6f60df458888a1a64b1044a3e4 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80fa529cdf704c44b4ccb21c17c3bd12 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07237b92b77248f5bedaac30fc4bf4a6 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39e9a2046b0942b5974531be9e15caa4 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c8667460674432aab361bce30a8840 : SUCCESS in 5m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ae74a13375c044d58ef7b46dd899cfc0 : SUCCESS in 22m 01s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fde817801d244bfaa50c66735b001888 : FAILURE in 30m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/208389b8fe4645efa16fe01e31deb110 : SUCCESS in 19m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90b8e97a6f134e59bc31308e70b2beef : SUCCESS in 56m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8c76f0e8c8f4d189aead17687ba9709 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":14},{"id":"3fc74a2bf0dcea1f45b6d2f65cfb85d43cbd015b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 12:05:30.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"2f24865de9814fc7a61892046b6e818cb696294a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:52:06.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859570 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"3cfdcfb788ba84e42b12f82b7a3fe2ff8ae1510e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:52:08.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859570 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"12c511801a7d20a6aa2587565f5079452c681560","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 12:55:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":15},{"id":"96d76b865c65f8970d4c901bfe62dffdf126bd9e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 12:55:22.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b3bfeee138eb6fbef9482c69b1c5ebf548046aea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 13:00:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5f2d0e602ccab64aed93139dd71dcbb1f1fdcf0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 13:39:04.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"5787bc8f07a3eca27ba7bb7adecc280ba3ee6087","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:45:47.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859570 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"ecb746da4f285b0838b4aae03653cab4ec5d7b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:45:54.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859570 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"373d376ae3cddcfe3ce23578c124727a42ba274b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 14:15:04.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"12a140e4688d13479c487d3c68a319367524d439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 15:48:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a47c9dd97174d42b88cf3f339ba64bf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f7d08ad08e914abda977cbf25ee8699a : SUCCESS in 25m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/90fafba3c3ef4df486ea20631b7ab8a9 : SUCCESS in 1h 23m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2e369e818e4177a1232711a2d9d9c8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ea6f134a194f9892ada55ca4c4bb84 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b19581cd5436486baa0dd3d028de9c22 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef56f2db2ff84d8f8680377b31d9f61d : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06b630112b347018bbfa7ad52ffe3c1 : SUCCESS in 6m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/81f3eb9bf31c4f97b065cb3d7fec4968 : SUCCESS in 25m 12s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3e758056883454abe3b2f6aa6bb9812 : SUCCESS in 1h 23m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0e8eb3ad7be849af9dcca798e6edbb94 : SUCCESS in 28m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67f01a57c5c54395b3b5cdb3c0f11cc0 : SUCCESS in 1h 18m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/637fdee119fb463892d659754c208b0f : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":16},{"id":"bd220aac1e5f0507c2a2e0ef39034667c5e2b293","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 16:18:26.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4cbb32bfa944868fe4da5fbba8686f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/36a47521850f4e6ead5f61c32cec54b5 : FAILURE in 2h 00m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c88404403dfd44bb9740b18d945e60f5 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc2ec990a60a4eca81557c80a1dacc6f : SUCCESS in 8m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c6533a5d9cbd4d7f8c641348355f4040 : SUCCESS in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"50f7277e4f525e091510e667522e2cb5daaa9c91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 08:00:12.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"1d13e4116c334a95a2104ef14648173a9b601f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 08:00:39.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-rockylinux-9-aarch64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-rocky9-aarch64\n      parent: kolla-build-rocky9\n      nodeset: kolla-rockylinux-9-aarch64\n      voting: false\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 31, column 3","accounts_in_message":[],"_revision_number":17},{"id":"524c2ba63ade3f5e718a5612a6074e1c7c2c2efd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 08:00:43.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  The nodeset \"kolla-rockylinux-9-aarch64\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: kolla-build-rocky9-aarch64\n      parent: kolla-build-rocky9\n      nodeset: kolla-rockylinux-9-aarch64\n      voting: false\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 31, column 3","accounts_in_message":[],"_revision_number":17},{"id":"e864736b557e774359d04c81c2dae9057c181af8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 08:02:29.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"8cd66b11127d9ebfc5454a05431e46a2113cdf05","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 08:39:03.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"d5416fed9d87bfb9bb6ddc234cc4fca6a0ffb4dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 08:43:34.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"4ad0f9d2db9bf1a0d1069a42085468e475823f08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 08:47:16.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0f14542049d46c191500d2aeb640990\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/39fa3219a11e49029b57879ddd553e89 : FAILURE in 36m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1910cc82a764914be13f41d81c045c4 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53b3891a35dc4d19b4f950d44da56c84 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0812cc63427f4750844ac48c0ed536b2 : FAILURE in 29m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/48bc751e76f74143b45a262d746d66bf : SUCCESS in 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5fa26bb56be517d3db74e19a454311be8d1d8003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 09:13:46.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d453155b7844249854ee063a17116f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8788d774f8024643bf5112a2b7103b31 : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ffb3fdcce5124e5280e6217bada95129 : FAILURE in 17m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657d87a5ce964fc7821e3c83aaaed6e4 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a83200621584d33a4730d67c91908a1 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50e59f3fc7294ec7b6bc96dd1336829b : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4550dd0527a4ac2bb13510749e93ccc : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e40de9ec5c457d9465421df0286bb3 : SUCCESS in 6m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d6aa56906f0042c1a6fa3ddc575646a0 : SUCCESS in 27m 51s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ba53579bb8ec4846a32c8f646a4b7b52 : SUCCESS in 1h 03m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cb2bd1541c20455fbcd6150bc6fd9936 : SUCCESS in 20m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2d86f6693224a9da8787004d675eade : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7232864d06154274a67d0f57bbe3729e : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":18},{"id":"4a9c6a6eaba86c04a773bc98843d3c999bd0b49c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 09:30:12.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"509c1fd145b140bd2deaad667d5674145f9bfd2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 09:32:43.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"931618ccd07469f0fe61f0f36ef970407df30a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 09:33:16.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"4dba71097d9952ad274d0ad4d4da206ab34bf737","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 09:56:31.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"b4b70c956d4d399ff6ff429d5cf36237fefd58c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:21:02.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb1b33ef028c454b96c31375e637528c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aa551ab2e0104243a6e46768a2c0d631 : SUCCESS in 24m 26s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/1b849c81a9f947d38d64f6608aec7bd9 : SUCCESS in 1h 21m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e01494a93474f528136ca1d7ed80d9f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f886bfe69dd64cb6acec1e51a3961226 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58b4f5820fbe4879bf43e06b1643045b : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/394755f820274702adf790684d9749a8 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9babf8fb86ed4bca9760a71221463a8e : SUCCESS in 6m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6d22c0d325d545d69e6edfd3f60b20df : SUCCESS in 29m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3204ecfbdda34cc4a407009944acf7b7 : SUCCESS in 1h 05m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ba5ff7063a34a268267982256bf2c75 : SUCCESS in 21m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff19997ddf504c03bf68b527ff9a7431 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0bd8385c2774742bea2e5a8f130a323 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":21},{"id":"db925ac37480b0feeffdee0988f94ff50838b31c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:45:07.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a034524e0804a48a37f919a350125f7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2490b71fd114474ea376f4abaa20797e : FAILURE in 1h 48m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/131f3b52026b445eb9f40f775bfaa68f : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3e145a70215409486c300526bd6c4b7 : SUCCESS in 7m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/935f925bfc86461a99325f8bac1f385e : FAILURE in 1h 20m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dd175c8377d246b096822662287f5b9d : SUCCESS in 1h 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"1ea1e68e6dd78d4cd5d082b867b101ac8795b1d5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 13:25:27.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"f25f3fcfb93453ceb0fbb6db2d1d3d8c0f9a1112","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-30 08:13:48.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"079bc26ab89a11a19fcf919f0c3f8cd818140259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 08:15:10.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"d11eb88552e3a87f974644324a84fcbbde00bd04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 08:15:14.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"126b8638d655bb940cdccb0cbd98aaf5bb0f2ae2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 08:15:24.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"a11fb51ed7da9104c1967265358110ae7cea0be2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 08:56:07.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f6adf2b84b4bba9bdc66fc1aab0341\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d0f741e8c4674441865f10a850af8d4e : SUCCESS in 32m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc6f44972cb747fb9eb0c9b3dda95943 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a237584bbbbd44aa9c748c01e076981c : SUCCESS in 3m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/43d0f03945e24cbd917bd4551a01f159 : FAILURE in 24m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ff083aacb1ec43c49c465ba502a11dc4 : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"94323804baaf846444491a0da52b890ad6e7cca6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 09:08:41.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"5e7c42ce6e5fcf54bbdf4b658f1ad7541f93bd97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 09:46:57.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"dd91c95b53ae37d42d4356679726fc65d75a1c85","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-30 10:14:39.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"2366aa34143a61d4fb0e21c413f686523b2c5b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:08:53.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331d2150cc364d6db7785d0fb5fbd87a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8a105e89865b49519bda6bd52a0807b8 : SUCCESS in 30m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/92d0a3f89e09451d87f15b82a4baf175 : SUCCESS in 1h 17m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27f18c4b4b441f384d5fc8c6d900fd3 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81864b02e7334b13b02890968d361fd0 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e1e56747bd4cfb845be3ba3bc2675f : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99cd3f8ee1c54391bb9276ccce43e78c : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c17c13d71fe4bd7b72ed0a2a23aef8c : SUCCESS in 5m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/05f62a72f0064c609b1fc353194c45a9 : SUCCESS in 24m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6ac273f58f534a838d4c9cc1515d2e98 : SUCCESS in 1h 09m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/792563ae4e8a4d8e9e7dd4ff44d3fde0 : SUCCESS in 19m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/147745f8f3344070b63c1994330cbb04 : SUCCESS in 1h 00m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0857171603c249b2bed62a22a8e8f73e : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":25},{"id":"45bac0f97555044f1c3cc97978288715a9b6fdcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:17:35.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23c957b009814b60b5a521d6769bf999\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/62cdd962b4ae4dba93e51725480f73a6 : SUCCESS in 27m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8e680197fa14aab9d3bd08885909178 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bb8fff7d83f4198944640797737251a : SUCCESS in 3m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/afc137397352478693f69a4e57171565 : FAILURE in 24m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f7f34b8ed046428da562a130dc240088 : SUCCESS in 1h 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"84791371c1aef7e6efd1f983fb7d98eba65a89af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 13:37:45.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"ff9757d582bdf44190977d994d062181d8d674b7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-30 14:00:44.000000000","message":"Patch Set 26: Code-Review+2","accounts_in_message":[],"_revision_number":26},{"id":"529e924f661f5e245be420f211bc3af8412c1b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 14:14:32.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"5e0eb3a0d50dc66d6174882f2701ddba3d43006d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 14:15:41.000000000","message":"Patch Set 27:\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"f0e7f35f6ce5fcff57f8f4baed8e9d10d4768d1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 14:17:35.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"def498ce607c4c18bfb573674958b365b91a5ec0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-30 14:40:58.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"408c870e3f17132b0e85f83ddbea1129515a1ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 15:30:09.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6cfbbdeb35d47d7a33229fe1a5e342e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ee01b5932e2a4aa488b6d9d70ac44905 : SUCCESS in 23m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7ed6ca92cdad433fbd670b1a9f7f6396 : SUCCESS in 1h 11m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c91a73112d949e9b398d9cdd488ffef : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ca560453d61472a839622eed8b63f5a : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2d0b5a819cd499a819857a0880f8ce0 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e392b3b52a9471b9358ca273e191f0c : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e5d352cc18488c9bc1e6e0a9b270cf : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1a214f09c0468b9f5b61b7ef6cd8e7 : SUCCESS in 6m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a8d2abb6f5d424eaeb20920ee4f0d3b : SUCCESS in 26m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/07c58e937eef4b03896ead6162d64731 : FAILURE in 1h 01m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8b2571b8dfc64b6fa48589ff14ef7cac : SUCCESS in 20m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ebcf18225af4286a19cff29755f6cb9 : SUCCESS in 56m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37d4fcf41c4a4bacb67c17c1638ca6da : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":28},{"id":"f5f8d5a4e9259592ea7788622632b20b0cc95856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:34:29.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd573d5ce46a49798d5d269f71efca45\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f3e67e801714678adf1485f3e8baf9c : FAILURE in 41m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2703b48a6e24fe6955331e2b916ff1e : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af6e0580808c42a8a29bec39e2a4bf07 : SUCCESS in 10m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/70bcbc13514e4cc98b01d03b50820b3c : SUCCESS in 36m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bf0bd5dfb9cb49228e440aa03483356e : SUCCESS in 2h 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"419fc2307dadf90d3366147f29f8f636d276f168","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-30 17:49:58.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"5cc3815b033940f01ff31f136295d012eba0c57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 18:57:22.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec3da1138852488c9d9ea85a44b986c0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/786e9729127946f89a54169b15ff3a3b : SUCCESS in 19m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/36758b02e60f4ac9a526de48424723da : SUCCESS in 1h 01m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c97146e2db4153baad04fee733492d : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bf05597bf04ec8849c2f23c06014aa : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/810d08e8e9e74825b4ff65731994a675 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79c7d591921a49f086e78df7f09d142b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81666c915be644d1ab805d7b41a432a4 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7378a945b44d56930df109d650d2e0 : SUCCESS in 6m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a3d0ac0f14f54ef6a8a752943cf3d240 : SUCCESS in 20m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9553346d0c60465a84958e95db7f2351 : SUCCESS in 1h 01m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/591f888634aa42dd8627ccc7d1f64b18 : SUCCESS in 18m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b1fb2c693b04587a01bcf6bac90b190 : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/640887aaee994c11a3631d18b2364669 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":28},{"id":"e3005f3752a442de1e83b5e142ddfcaffa378f23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 19:26:59.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084d7fca95c14ce99fd33082dbe47a38\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3b0adb7424ca4857b169dc5c71e12e0d : SUCCESS in 1h 34m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0513c9362fbc40da9e98290df3e8153e : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62f3f331c6a94ad6ad1b0a634cb7184d : SUCCESS in 7m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4692332988a04e58afe5d099be708a84 : SUCCESS in 51m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f51548cf58dd423b9ebd81a0ae1799d6 : SUCCESS in 57m 44s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"0b0917f7f56da7c8f2543f4b32da58dca9ea4301","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-30 19:31:14.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"d23712d93de94181ff47abdcd196c269f2858c8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 19:31:43.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"9c07f0fac07f602cfa9b66f3b3bbe352641f275b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 20:50:45.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8c04ec97424beeb96bcff7bc862bcb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/706c8219fd874b5681138b063fe1a589 : SUCCESS in 24m 45s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b6b37ba240fe4be383095c3ce3cb7ab3 : SUCCESS in 1h 15m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465a557005764e50a71d9672557dc379 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc4ea3cb1424a26abe728215b63f61f : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f332073de94171a462ffe3bfdc9374 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdb700b94904d5c9c46eb9e7394e9a2 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/306239655fcf473bb8eda4895646e0ca : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ae93a206dc22429c80c927f8dc58c42c : SUCCESS in 24m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/649886e3736642cd86fed1d0550540ed : SUCCESS in 1h 10m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a21f3cfd8a3647d19b93246308ca3a8a : SUCCESS in 22m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bf8d3a23fdf488aadaf2ca79f45f652 : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c3adf708ff0a4e94bc0c3de559148e11 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":28},{"id":"d9268de2e32b94bab53a47a763d378471717e60a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 20:50:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"a7a4185ba390aed1a868f6fb31ebb644d05eb032","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 20:53:09.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97d018dc797a41d3a44465feea42592f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb955850a2fd40519da9cb5dd1b660fd : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de798699b5874d58ba18d63c8bf4aeb0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"0883532434fff341c444d8fce8e28fcff3f5e769","revisions":{"9f70eef9f6c127b71f9cd11eef47d2fd814c5376":{"kind":"REWORK","_number":1,"created":"2022-09-27 10:18:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/1"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:34.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f70eef9f6c127b71f9cd11eef47d2fd814c5376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f70eef9f6c127b71f9cd11eef47d2fd814c5376"}]},"branch":"refs/heads/master"},"aa8ebdec234ccec95479555035eaec9215c07724":{"kind":"REWORK","_number":2,"created":"2022-09-27 10:20:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/2"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:19:59.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa8ebdec234ccec95479555035eaec9215c07724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa8ebdec234ccec95479555035eaec9215c07724"}]},"branch":"refs/heads/master"},"69c5fa566987944e8f74afca8098ad04d78a9ed9":{"kind":"REWORK","_number":3,"created":"2022-09-27 10:22:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/3"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:22:52.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69c5fa566987944e8f74afca8098ad04d78a9ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69c5fa566987944e8f74afca8098ad04d78a9ed9"}]},"branch":"refs/heads/master"},"71e5e59b135184eddd68f6d755faa352dab3d276":{"kind":"REWORK","_number":4,"created":"2022-09-27 10:32:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/4"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:31:59.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71e5e59b135184eddd68f6d755faa352dab3d276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71e5e59b135184eddd68f6d755faa352dab3d276"}]},"branch":"refs/heads/master"},"97a19a62c2255fea2e7d1f5b626854ce43f714bb":{"kind":"REWORK","_number":5,"created":"2022-09-27 10:50:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/5"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:49:55.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a19a62c2255fea2e7d1f5b626854ce43f714bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97a19a62c2255fea2e7d1f5b626854ce43f714bb"}]},"branch":"refs/heads/master"},"f4f2ab0c24ca4f4040c465af73379b2c828adbf4":{"kind":"REWORK","_number":6,"created":"2022-09-27 11:38:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/6"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 11:37:54.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f2ab0c24ca4f4040c465af73379b2c828adbf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f2ab0c24ca4f4040c465af73379b2c828adbf4"}]},"branch":"refs/heads/master"},"42a010d195b1fd20524c883781eb7a33194f0c5e":{"kind":"REWORK","_number":7,"created":"2022-09-27 15:03:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/7"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 15:03:26.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a010d195b1fd20524c883781eb7a33194f0c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a010d195b1fd20524c883781eb7a33194f0c5e"}]},"branch":"refs/heads/master"},"76eb6cff7c24655541e65fb33841d109b32c42d3":{"kind":"REWORK","_number":8,"created":"2022-09-27 15:33:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/8"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 15:33:50.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76eb6cff7c24655541e65fb33841d109b32c42d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76eb6cff7c24655541e65fb33841d109b32c42d3"}]},"branch":"refs/heads/master"},"75f72d5782448e5e6fa9f58a6ed181444c5cce3c":{"kind":"REWORK","_number":9,"created":"2022-09-27 15:43:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/9"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 15:43:07.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f72d5782448e5e6fa9f58a6ed181444c5cce3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75f72d5782448e5e6fa9f58a6ed181444c5cce3c"}]},"branch":"refs/heads/master"},"e373c9bf9966c45dec39b4e3858c9df78129a150":{"kind":"REWORK","_number":10,"created":"2022-09-28 07:19:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/10"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 07:19:42.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e373c9bf9966c45dec39b4e3858c9df78129a150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e373c9bf9966c45dec39b4e3858c9df78129a150"}]},"branch":"refs/heads/master"},"509fbd99f8990289fd98c0699e2a6ed39c0c92df":{"kind":"REWORK","_number":11,"created":"2022-09-28 09:05:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/11"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 09:05:16.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/509fbd99f8990289fd98c0699e2a6ed39c0c92df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/509fbd99f8990289fd98c0699e2a6ed39c0c92df"}]},"branch":"refs/heads/master"},"6b223235208cedc67d7c05a53344d32733c4f7fb":{"kind":"REWORK","_number":12,"created":"2022-09-28 09:18:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/12"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 09:17:59.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b223235208cedc67d7c05a53344d32733c4f7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b223235208cedc67d7c05a53344d32733c4f7fb"}]},"branch":"refs/heads/master"},"b0e37bfea90983bd7a4c334eff4cd0c5dd3d5a40":{"kind":"REWORK","_number":13,"created":"2022-09-28 09:20:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/13"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 09:20:36.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0e37bfea90983bd7a4c334eff4cd0c5dd3d5a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0e37bfea90983bd7a4c334eff4cd0c5dd3d5a40"}]},"branch":"refs/heads/master"},"7938d1e8e3af1bc27a9a514bfe23a3f77cef8a26":{"kind":"REWORK","_number":14,"created":"2022-09-28 10:41:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/14"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 10:41:28.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7938d1e8e3af1bc27a9a514bfe23a3f77cef8a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7938d1e8e3af1bc27a9a514bfe23a3f77cef8a26"}]},"branch":"refs/heads/master"},"fc58bf8b42d217d8424d5917327b6cdcdd940b43":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2022-09-28 12:05:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/15"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 12:05:03.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc58bf8b42d217d8424d5917327b6cdcdd940b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc58bf8b42d217d8424d5917327b6cdcdd940b43"}]},"branch":"refs/heads/master"},"73a59604063e088e9efdd62fb6f480a6549e24f8":{"kind":"REWORK","_number":16,"created":"2022-09-28 13:39:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/16"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-28 13:38:57.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73a59604063e088e9efdd62fb6f480a6549e24f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73a59604063e088e9efdd62fb6f480a6549e24f8"}]},"branch":"refs/heads/master"},"88a56d1f1820446d376f67d03a40b3b561b82e58":{"kind":"REWORK","_number":17,"created":"2022-09-29 08:00:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/17"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 08:00:01.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88a56d1f1820446d376f67d03a40b3b561b82e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88a56d1f1820446d376f67d03a40b3b561b82e58"}]},"branch":"refs/heads/master"},"144525cd707156c0927982bb786f2a93030b14b8":{"kind":"REWORK","_number":18,"created":"2022-09-29 08:02:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/18"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 08:02:23.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/144525cd707156c0927982bb786f2a93030b14b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/144525cd707156c0927982bb786f2a93030b14b8"}]},"branch":"refs/heads/master"},"250d96ec152214d81160f9d1859c3494790f28e9":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-09-29 09:32:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/19"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 09:32:43.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250d96ec152214d81160f9d1859c3494790f28e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250d96ec152214d81160f9d1859c3494790f28e9"}]},"branch":"refs/heads/master","description":"Rebase"},"accc5cf420ab091de9edf2b9aee0bf13801d23c1":{"kind":"REWORK","_number":20,"created":"2022-09-29 09:33:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/20"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 09:33:16.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/accc5cf420ab091de9edf2b9aee0bf13801d23c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/accc5cf420ab091de9edf2b9aee0bf13801d23c1"}]},"branch":"refs/heads/master"},"59f49d9268f959fb200fc26ba680b11a0e051bb0":{"kind":"REWORK","_number":21,"created":"2022-09-29 09:56:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/21"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 09:56:30.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59f49d9268f959fb200fc26ba680b11a0e051bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59f49d9268f959fb200fc26ba680b11a0e051bb0"}]},"branch":"refs/heads/master"},"657d8dca117385a6663ee1e19f4ea8fb0bb5f049":{"kind":"REWORK","_number":22,"created":"2022-09-30 08:15:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/22"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 08:15:10.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657d8dca117385a6663ee1e19f4ea8fb0bb5f049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657d8dca117385a6663ee1e19f4ea8fb0bb5f049"}]},"branch":"refs/heads/master"},"50834444cfd42ecf63a45bf365b003413384e81f":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-09-30 08:15:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/23"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 08:15:14.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50834444cfd42ecf63a45bf365b003413384e81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50834444cfd42ecf63a45bf365b003413384e81f"}]},"branch":"refs/heads/master","description":"Rebase"},"9d63d0cfee6721779a0797d0f4c0ec864b5a58f0":{"kind":"REWORK","_number":24,"created":"2022-09-30 09:08:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/24"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 09:08:41.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d63d0cfee6721779a0797d0f4c0ec864b5a58f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d63d0cfee6721779a0797d0f4c0ec864b5a58f0"}]},"branch":"refs/heads/master"},"201a340cf5053943e9dae57431888110661c7423":{"kind":"REWORK","_number":25,"created":"2022-09-30 09:46:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/25"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 09:46:56.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201a340cf5053943e9dae57431888110661c7423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/201a340cf5053943e9dae57431888110661c7423"}]},"branch":"refs/heads/master"},"69fbd6ac53a033ea0cc7db4769d0ec798f008b22":{"kind":"REWORK","_number":26,"created":"2022-09-30 13:37:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/26"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 13:37:45.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69fbd6ac53a033ea0cc7db4769d0ec798f008b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69fbd6ac53a033ea0cc7db4769d0ec798f008b22"}]},"branch":"refs/heads/master"},"f80e7c7956579f513cdcb7cd8a17e70d6ffbc04d":{"kind":"REWORK","_number":27,"created":"2022-09-30 14:14:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/27"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 14:14:32.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f80e7c7956579f513cdcb7cd8a17e70d6ffbc04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f80e7c7956579f513cdcb7cd8a17e70d6ffbc04d"}]},"branch":"refs/heads/master"},"0883532434fff341c444d8fce8e28fcff3f5e769":{"kind":"REWORK","_number":28,"created":"2022-09-30 14:17:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/859400/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/859400/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/859400/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/859400/28"}}},"commit":{"parents":[{"commit":"237013960425e4ff6cbc4814d4f9d61efcf601de","subject":"Merge \"debian: use Erlang 24 on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/237013960425e4ff6cbc4814d4f9d61efcf601de"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-27 10:18:01.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-30 14:17:34.000000000","tz":0},"subject":"Add RockyLinux 9 support","message":"Add RockyLinux 9 support\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859570\n\nChange-Id: I7af51d07b8a939c908aa072294061def57dd67de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0883532434fff341c444d8fce8e28fcff3f5e769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0883532434fff341c444d8fce8e28fcff3f5e769"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
