)]}'
{"id":"openstack%2Fkolla~859463","triplet_id":"openstack%2Fkolla~master~I04121fd5ca9894b6deeda3808808cbffd6b97811","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-05 13:44:48.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-04 13:01:36.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-05 13:44:48.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-10-05 13:44:48.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-05 13:44:48.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-10-05 12:17:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I04121fd5ca9894b6deeda3808808cbffd6b97811","subject":"CI/centos/rocky: use repos from opendev mirror","status":"MERGED","created":"2022-09-27 17:25:20.000000000","updated":"2022-10-05 13:45:44.000000000","submitted":"2022-10-05 13:44:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"859463","meta_rev_id":"c3fc5a19ada1f0e3d4f1a6d595054cdf86bc6652","_number":859463,"virtual_id_number":859463,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-04 13:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-05 12:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-10-05 12:17:50.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 18:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-27 19:38:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-09-27 19:38:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-09-27 19:38:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-27 19:38:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-27 19:38:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7770cb1267036c50c9721125dd54a0d8f7e09ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 17:25:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdf2e2ead0ac4cc778f8a9e800d84b8b3370ece1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 17:25:51.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8adcaab690a183b1faf14022be08838e1056c934","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 17:39:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea2d7ac87e23d87066b72919b4742f64f48f4f78","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 17:49:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2eab4e349f14b72aa7bb0ed85ffb427cc9181a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 18:00:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8106e0e28a364bf8acee51a719462cfe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/027574e8f9794c03815e185ed889a1d2 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70269413762a48c98b75351435553a4d : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e85801aa663a448da16ee0725d6d9600 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ddd854a9c4a43909b2d8b6a0ad17ef0 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/634dc1fe58ed4fb1979fa9fef468a3eb : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3d9083494c476b98830522ea9c0287 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"1bba08afb558d84f7b2e95c80fd872482ef2e74f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 18:00:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27b534f23eab4a989605d581fac0fb0d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97b998a620b94cfe9d946576561f7238 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c81ca6751c704039a9c79f90beea1daa : SUCCESS in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f31c23a4c1dffef0ea9416bcdb8a1df93b8c697d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 18:10:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e9b7574fca75c44affe1f80d791f49a6318bd8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 18:21:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f6872b8ddd472781a6d29e4ade2ff2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ff33e83cc1346998170840dd028a4e3 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/387e5bad21614db8963a34f8be71269f : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3561c366dbd9cd1a7247d902fb76cf96aa53a6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 18:43:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6ee9449d0b478e8a3cc21845f3c92a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/21c7fae172fe46118121d3b3d4de3c36 : FAILURE in 27m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b744f5a35a49379228b76f884879db : FAILURE in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8fa60bcfda24219a0d68338bd937134 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1136818d8854d81ac353fd6ec01d04c : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8b59c227ea444abbd815bb5d8eec17a : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f952608c9c094cf18cd2e1161fc9587e : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":4},{"id":"07dd40a5ce7a122506fa9122e21b1f50f7d82358","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 18:46:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1c9a6efdcc32f293257231fe272aa2b896ab8556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 18:55:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1938edced8bb435b9573f814eb311767\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4ab7612edd3406a84d849ee6bbf6a37 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b97321ff3cc4c7c867fbe6eed52cf4a : SUCCESS in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19b047be07dde18ca32707d07c83a40c21055c34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 18:58:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"de8a88acb4f65c4da55ff95ea48d2b8a2149e2b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 19:09:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45696e2ad7324742b930d2cfb51c638c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1eb14774b3dd4ab6a8f70eea47cbb6c0 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/962adf5fd1604937923589e92331ab7d : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1623be3ca30265508c585bed5b101672f64ea603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 19:29:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0b7b46ac48d432d8c9a6d7ac663c69d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/15b9e8db6d354435990920529a9dc5aa : SUCCESS in 25m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13fa64965daa4a1f841ea20cf3d8f57a : FAILURE in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0610ec1d62ee4b009080c83f73b7df98 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6a3a96b5d484c1f97ee46768edc3d78 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/904702c0c7c04055b2e648b75c070d5b : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4cd66ae26dd46c690c88946a403afc2 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":6},{"id":"db25f8307f7b0df4052b84aefeaa86d17302bb56","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 19:38:07.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"c9964d3dd9f607f53a16cb374645ed61a75866e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 10:31:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"18ead0e19ec9a498a93fb5a085bb893e45d1b411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 11:09:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18e58a791154bc68e5c24a63d1da832\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/477a1a6f68574da9a20b4bbc1699a03e : FAILURE in 32m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a0bcdf92c7244faa58305f9739cbcee : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1bc1645f0c34854bd0d3fd9c9259fd0 : SUCCESS in 7m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b8a47c6521e94ce58ce6143850c1791c : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1a55fabfa086ff475e20763e6f77e4ca8a2306c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 11:52:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216c4331d20341b094f5faa40d5f6dd7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2aabd605a77344348203aa8b1decc8dc : SUCCESS in 23m 32s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7098a413c86c492ea746539cec38f2c0 : SUCCESS in 1h 16m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f80812cf2f413e8ef01eb860ee94c3 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f4c17227b448d4beae365634048da7 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc1817dbd27e4b28a7395a3163ee5449 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/300a4e035450495f9b358bad824bab90 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eedc84e476b54c35a240e7c034c1b02b : SUCCESS in 5m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/41e00bb47d2a4b56950b4d0f846ec21b : SUCCESS in 19m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d38b13927deb4086bd8b80be01874075 : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f696a1297d454b27abe4794b53c3dbbb : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":7},{"id":"debf990c50bf8d32d4d8d937e914ed38053c0447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 12:09:56.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"3d286b51f044d7efb4a664ea8ceb8f3c4514b833","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:56:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9b30e25b2114f699c296880c84cc2cf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/63eb33afb7ff49fdb6059f9641174c99 : FAILURE in 37m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55c456ea376d45e487fe144c3bea1590 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24c137bb364a44d2b04d63cadbf2100f : SUCCESS in 3m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/291e0c790eca4ca5a877a17bb4c1be3a : SUCCESS in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e7e2c2f06a32298e6a423b35fe7752f96ac69b6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:36:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde6792660ad4ac38a9ff20c0cfc0a68\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/065edd0a647c4dcfa1b45be843e9d6f1 : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b4c3da88c60641f3ad543f63f806bede : SUCCESS in 1h 24m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566daea8cefe47e9becdc538e01d5fdb : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05683360dcbc46ed975fa8721739242b : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c9c9236b10d4efa82c6592ab68c11cc : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/108aa863001a4581afb75d1b3cf7915c : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373147c8d9864c918aa442f1b3bccb32 : SUCCESS in 6m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6abca1c360d14e019c7b0acdd28e3642 : SUCCESS in 18m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f38818e9568843c9a118d8b40bfda483 : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/303d15f409b147e39e17012b2c5d8c1f : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":8},{"id":"396c28e08e9507eafdae5274496014764291da94","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-28 15:39:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"37cea985f60ae8d2a38f2681175f12f47921d3a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-28 15:42:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b68104eeef773027fafd3f3f0081726ca88a2a9a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 15:44:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"058ffd055c1b3e8b989549b5d98154ccc3391496","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-28 15:48:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8c03ab16a00ac78f57013e72c632e8072f825f51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-28 16:28:11.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"86f7acbade35381a8b15118ddf7020517d595c00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-28 17:16:03.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1ee603f922c500467f908a6d01302836d5814a87","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 08:41:10.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c0bfdf618ef2cf4c9adf2953a266c40a75ab0025","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 08:58:05.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fa0c2737f0ad3a5c527232bbfcbd5ff3dae70d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 08:58:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"19d3db5d3a2cb14bf1a2c30755ca4beaa503b915","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 08:59:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"de10a3990ff7cff2ee4db1d7a4ef7b430984ea0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 09:00:21.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bd2da69535813a6bdcb34e0db3b65577d8f8877e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 09:00:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b79299d2ddcab3005972c0665f81f98aa3b94dfe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 09:02:27.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ef99dc19549b022a3fd6331df79cf4ae2eb47baf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 09:13:04.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d500307681c55aeae3ac155f9e0601ff9f8b2f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 10:14:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c6fb82800154b0fa954d3ded23a354f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/36486251cb984260a7ec25d779cc3dcc : FAILURE in 22m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f80907a855b34da286fd9c61dd7d8924 : FAILURE in 22m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1731b88b078c4c17a24c9d6ad3a6052e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4d1a6301e134ae886b7af409912f5c6 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fb49de1313548b98aebe7da3096df9b : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6408bdb689e74cf096a790fb5f411be0 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38580c9bc85e45e297be265ba5a88f71 : SUCCESS in 5m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/24a47b88fcbb47108ad6cc0343b85cd8 : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bc148886d6b46f89274a97943d2b690 : SUCCESS in 49m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06b80c07c1a543c484ad7778388fe641 : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":9},{"id":"55b17aa90e9dd164d1c866a962e824af77447e75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 10:20:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1103cd97d9ce26bd521be09dada255fca49cda47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:34:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce1c04cac3c41c894f7b18a0271d1b2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/31a941b3e3f746bca1a7bad264c31cdb : FAILURE in 23m 28s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6241177165794f898191d8b7fc27e41a : FAILURE in 22m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8eed0fe1db41eb9d3bc3a66a88b166 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5719ec05afd349798bd9945a1f82b0bd : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48ebd5fe76a34233805f4fd17776a088 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cc7ee0933ff41508354fbacb1a64321 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43163013b05548779dda6a89234d0388 : SUCCESS in 6m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec4acd87e89f4d388d3a4ee107e5483d : SUCCESS in 25m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c40ffd81020540bc8ed2e0309f5ab191 : SUCCESS in 56m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed7b5647fcf748509f84092342d2cd47 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":10},{"id":"27ece55903bebea88b92bbb9cd5dd81e93b9cd50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:56:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0251578ec05403d8365c17fa7a36383\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/204513c27d414cecaf8be28c93167cb1 : FAILURE in 1h 20m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c61fa627ddb040d39e58701930d7eb06 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63688160eba5432186acdabe65e5bb17 : SUCCESS in 8m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/de6cfbaa5c8845319f18f6f11c685c2b : SUCCESS in 1h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fefae5476b4a04e82da4ea40d7db8dd025833ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 12:16:04.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"3101c4caf067c6842fc1deb8f8ae581439a65b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 13:43:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25d1b056fc84d3580f6b94888b714f0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3413eb9a8ebf4eb3877ed74480cee001 : SUCCESS in 24m 42s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/59f0ba0f347f4de099962074705ae6d1 : SUCCESS in 1h 20m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fb8d456ca545dd8b19764e55423e86 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/094fa3a4e8fe4bd9b2b0e2c15b9db388 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b733ca9300c402498c49dc88a2db147 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d540c91fee7e4cf59a3ede474c5ce283 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32d0ca3a2ac4a869c0e4593765a65ae : SUCCESS in 8m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d2346564ce184b5eaefa4147b1769403 : SUCCESS in 22m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec17a888a2774b86b7039fb23604477a : SUCCESS in 54m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29b4c92162784bb2bca42fd976827aba : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":11},{"id":"939d2ac629d8a2ffe2fbc8b1b5f0f8f531582c47","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 13:52:44.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b54bedda4668f96b593c62020fe8df1852f0f023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 13:56:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1ef1ea9357a44b8b228c82a88d72c77\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a85b66b3ab514565afa24d45c7d89fef : FAILURE in 1h 36m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/083d1f4a473245fb8b969d8631b6c595 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49dbd0ccd37f49e6b81f350e28759a4d : SUCCESS in 8m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5585fc9cd40d49229d0d0cc6d94eca29 : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"137e3ab0c3114d546d0f95bcc31bce4dd5858233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 08:01:25.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f5c5ee958accdd292986a183b12416a66cee3a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 09:20:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a665bdf1614c108893c995324290ec\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/397740848dcb4d488fc40317eed441e0 : SUCCESS in 26m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/342fee5ce5fd4644a99db2fd4c298f9b : SUCCESS in 1h 18m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e1abe55de3419d850b234ba6b158be : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/488917c84ae94fc2b0b7727c9bde2156 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7580eaab6f6407f93e9ca783491213a : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365f215bcd5141e48b00bfadd2cb23fb : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8a972a4ec04ae9b11c5d88b5112315 : SUCCESS in 7m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1871dbec4420431885aa8a9f62ee333a : SUCCESS in 26m 14s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/06e41ab3bf854aa59ad912dc3b69c593 : SUCCESS in 1h 12m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/59363ce6268a48c888909d01477a70f0 : SUCCESS in 25m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/788c7a47753143149dd9659dd1185c1a : SUCCESS in 57m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5b5640151414c63b989a70c7fae5a8f : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":12},{"id":"6b408fe92e75fb4fe8913ed03fa91709d98bdac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 09:44:35.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"1ed8a3c30b39d3900eb4e066d2846f6375e883e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 09:59:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aadf1907766e4cc8906f48a7ae2ee925\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a792b4fd194149f287915dbdddfa1906 : FAILURE in 4m 56s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6782ba366ef64aeebdefeae2d46faf09 : FAILURE in 6m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382a427cdc864b228b9c56ee61c3c3f9 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc02456d852a424ca2f4c6c2ec40c4df : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb9b6d5b38849d3a6bc09072efe14bb : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/431bf37b20cd4842893daf0941b54063 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8922475441a64ef5a1c15991acd1656e : SUCCESS in 7m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47a7f026c8294541a44973eb532c50e0 : FAILURE in 5m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/45beee7e35f64de1843b50e2cade542b : FAILURE in 8m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af470f48e43b4f3ba0e4987914047e43 : FAILURE in 5m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6abd183ada7428e9b876e1c5d36149b : FAILURE in 4m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92bd59ee47bf44288ebfa92af458f313 : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":13},{"id":"5d33e600f54023ed83399a3082f291dec4cffdd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 10:14:09.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa67647d5023403ea2ba98553512f0e2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dadaa2b1ef434aa89628ec074a826495 : FAILURE in 9m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d99797cd2f6b4a37b106087370b87319 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edb44070b7a147e2aac4c09eb25848d1 : SUCCESS in 9m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5238b925f8b34880924c67132959f26f : FAILURE in 8m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/00c4d59f73594e8c8780d14e9d4e710d : FAILURE in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ef88ac61e126cb87378d4e30c6ce46a3303478b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 10:49:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b211bb13fef010d9a21161c5108a2161500d54ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 10:58:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6c16b627dddb43b98533f798f0ff4356a8e875d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 12:16:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9161e7ba2644ff898747485e3ff1eb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/156fa9a4d1a14c5da89b2e45042311e6 : SUCCESS in 32m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/21755fb8777a4e35bb0386faaf5b6ace : SUCCESS in 1h 16m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963a217dcf18472697a2fff7970e264a : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c8d0e85bf7c4f0db44b8648be5ca759 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55cc75977fac454ba33e30fde6d625b8 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d81e3b8904d04555b81c84e1a1487aed : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d9b48233a6401692e80dbdcd631a01 : SUCCESS in 8m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/30341dbe38ef4fe1881b4241046acb5f : SUCCESS in 31m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a459f3d0ea9742f2ac175b825764a676 : SUCCESS in 56m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/05787c37bd35406abe8bb7863b28fe3f : SUCCESS in 31m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff28aee99eb540809d7cca621c0e95da : SUCCESS in 51m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7604e7fd3b54d699cea23084df68074 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":15},{"id":"502d61e83e0ee492cce6b0bbe058a718cc508551","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 12:20:54.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0596435b744c07ab0c3fed57c478ee\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d2be312469ff4214b487f758ed9ec757 : FAILURE in 1h 20m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba8088e09af64514bf346b2050d424b6 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/361cfc08742f42b0a552d7cfc2cf85cd : SUCCESS in 8m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f12a5b3738b74563bb630f8a13bdfa2c : FAILURE in 1h 10m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f80d773077cc4a0cb7dc9493effb7627 : SUCCESS in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c305e5adea3abafa5b16a5c96bec08ecfc7fb645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 13:29:16.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c39242951bb5e48b9da167f4b94cb6553c761685","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:15:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8baad507676c402c9a985d0e567dc398\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/47fd3275e9104fad94eb6ee00a1dc595 : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5eb095b11c524355a55eb4ee5bd5d522 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e115a90113b4c9cad596f05be4e43ec : SUCCESS in 3m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b5863d1fa1384f9d9b2710666b6f4a77 : SUCCESS in 36m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ff4eba668e8549d9b8729ea4603c15e3 : SUCCESS in 34m 09s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"9aa76f2a47447ff2124cd627014523633378f9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:47:58.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1691a02c0e846158cbe891bca6c8fdc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/40184ac4299e44b9b1cf6a0fa94566bc : SUCCESS in 26m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3df53c13e926455c8fe0fb335bffae86 : SUCCESS in 1h 11m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a86fa1281d4bd9be4d7cbfb07555d5 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05a09c72ec34ae487f05941817704f8 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4b1c6519f61484089575f356e9add17 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a128a1dfc4224612adaa8d5dc7c3ffbe : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7dd150e2524a73b6cc8f9fdb59b41b : SUCCESS in 7m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/63606d469c374f3ab60bae33db099d18 : SUCCESS in 21m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/07de00b6460a465191504e836f5fa78d : SUCCESS in 1h 13m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/531511e6222a41b1adebdea94ff3f60b : SUCCESS in 24m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c52c562bf8954fa4a1cf7ec3065442cd : SUCCESS in 1h 01m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc52cf7b32584e8881fc91161a88eccb : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":16},{"id":"562a7d0d11750ecaadb523566bd62f3d1c92f4ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-04 13:01:36.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"e5b48fa1876b46b0d99b7e6000a1ad7763539729","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-05 12:17:50.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"3bf676543e312826bc15af4e63761eebdd6ab3ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 12:18:17.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"4da547473c4a2457de78d622c3092b5d85ef74ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:44:48.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385f3c12f34b4e0e99e0cc59dd9b8093\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2f4280fb06ae4e3681f8a7197f813fb7 : SUCCESS in 24m 51s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/40e1a41567f849fba18520744ecd28aa : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/abe8db3784e44a9980195b2d00c159a3 : SUCCESS in 22m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4c364c9052e48aab4e50df30e6caffa : SUCCESS in 1h 11m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e83ab2e0b38f49baa2f55024f2d5e008 : SUCCESS in 1h 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2867ab32fa5b458bb74c6bd92d714a50 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1cd6870b4ff4aaaba788bbf2dec229a : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9626b1212df4f96b13f8233859b1522 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd8890ef20b4441b6b503ef695c434c : SUCCESS in 5m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/619d4ea1e7c845d1998e9c84f3f319fb : SUCCESS in 24m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/075b038f3c5641d88456430988b616b1 : SUCCESS in 1h 07m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/31db36abed3c4d3faef8bc08b62298b0 : SUCCESS in 19m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2b8a8feff104c7fb808a5cd2bbd8b59 : SUCCESS in 51m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b463e89e90442efbde12df67175e0ff : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":16},{"id":"67a5bd760981032bb63e75db55bf39ae2139b72a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:44:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"c3fc5a19ada1f0e3d4f1a6d595054cdf86bc6652","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:45:44.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a9cb0cf878b46b6b2ccaa880363ca3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6532d4929a194bc2bd3c5a0c5f35ebf4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"0c523e7f1b10ab0b456c20f4e900a6c8c6f7dcea","revisions":{"e887096a06cfa0f28dd475f80aa34388b96eb00d":{"kind":"REWORK","_number":1,"created":"2022-09-27 17:25:20.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/1"}}},"commit":{"parents":[{"commit":"042a95be10514409323e2a3144f37e507359e3ce","subject":"repos: in CentOS Stream 9 haproxy is in appstream repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/042a95be10514409323e2a3144f37e507359e3ce"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e887096a06cfa0f28dd475f80aa34388b96eb00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e887096a06cfa0f28dd475f80aa34388b96eb00d"}]},"branch":"refs/heads/master"},"3d2ac99007dcf45be5603feeee196b47bbe74674":{"kind":"REWORK","_number":2,"created":"2022-09-27 17:39:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/2"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:39:00.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d2ac99007dcf45be5603feeee196b47bbe74674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d2ac99007dcf45be5603feeee196b47bbe74674"}]},"branch":"refs/heads/master"},"5b153fe03fb6b970feba2ab2d79ee251b4f50bbe":{"kind":"REWORK","_number":3,"created":"2022-09-27 17:49:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/3"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:49:32.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b153fe03fb6b970feba2ab2d79ee251b4f50bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b153fe03fb6b970feba2ab2d79ee251b4f50bbe"}]},"branch":"refs/heads/master"},"22348efd8917b806962a791a821ea9eed08a1089":{"kind":"REWORK","_number":4,"created":"2022-09-27 18:10:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/4"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 18:10:45.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22348efd8917b806962a791a821ea9eed08a1089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22348efd8917b806962a791a821ea9eed08a1089"}]},"branch":"refs/heads/master"},"43f409b4ed89db789ab70578d6418e5c6269ed0a":{"kind":"REWORK","_number":5,"created":"2022-09-27 18:46:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/5"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 18:46:33.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43f409b4ed89db789ab70578d6418e5c6269ed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43f409b4ed89db789ab70578d6418e5c6269ed0a"}]},"branch":"refs/heads/master"},"d0ae41011f67f4c7293b6177bd50d644a5706863":{"kind":"REWORK","_number":6,"created":"2022-09-27 18:58:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/6"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 18:58:28.000000000","tz":120},"subject":"[WIP] centos: use repos from opendev mirror","message":"[WIP] centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\nNOTE: removed Debian/Ubuntu CI jobs\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0ae41011f67f4c7293b6177bd50d644a5706863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0ae41011f67f4c7293b6177bd50d644a5706863"}]},"branch":"refs/heads/master"},"500770aa61d312a417c897c50112b1fb0d136a5f":{"kind":"REWORK","_number":7,"created":"2022-09-28 10:31:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/7"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 19:34:53.000000000","tz":120},"subject":"centos: use repos from opendev mirror","message":"centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/500770aa61d312a417c897c50112b1fb0d136a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/500770aa61d312a417c897c50112b1fb0d136a5f"}]},"branch":"refs/heads/master"},"2fef7fe05c152335f261389697badd93722d8a80":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-09-28 12:09:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/8"}}},"commit":{"parents":[{"commit":"2185f817baf57be9bdb39b3f0dcf3bd05462d3b5","subject":"Merge \"Don\u0027t ignore the remaining apt-get update errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2185f817baf57be9bdb39b3f0dcf3bd05462d3b5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-28 12:09:56.000000000","tz":0},"subject":"centos: on CI use repos from opendev mirror","message":"centos: on CI use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fef7fe05c152335f261389697badd93722d8a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fef7fe05c152335f261389697badd93722d8a80"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e703546da6021397a5c83ba2b1cbce4e422056e9":{"kind":"REWORK","_number":9,"created":"2022-09-29 08:58:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/9"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-29 08:58:34.000000000","tz":120},"subject":"centos: on CI use repos from opendev mirror","message":"centos: on CI use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e703546da6021397a5c83ba2b1cbce4e422056e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e703546da6021397a5c83ba2b1cbce4e422056e9"}]},"branch":"refs/heads/master"},"22ecf3c00442e4bc0fa7b2a01af0eb6c0d41678f":{"kind":"REWORK","_number":10,"created":"2022-09-29 10:20:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/10"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-29 10:20:19.000000000","tz":120},"subject":"CI/centos: use repos from opendev mirror","message":"CI/centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22ecf3c00442e4bc0fa7b2a01af0eb6c0d41678f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22ecf3c00442e4bc0fa7b2a01af0eb6c0d41678f"}]},"branch":"refs/heads/master"},"9361dea868cb99896b2bfade11e1a094a20a3e7e":{"kind":"REWORK","_number":11,"created":"2022-09-29 12:16:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/11"}}},"commit":{"parents":[{"commit":"4a94fb4d16da2f9e666cfd9dede38d1719bdf795","subject":"Merge \"repos: in CentOS Stream 9 haproxy is in appstream repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a94fb4d16da2f9e666cfd9dede38d1719bdf795"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-29 12:16:04.000000000","tz":0},"subject":"CI/centos: use repos from opendev mirror","message":"CI/centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9361dea868cb99896b2bfade11e1a094a20a3e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9361dea868cb99896b2bfade11e1a094a20a3e7e"}]},"branch":"refs/heads/master"},"01c6cf409d8df3bc98e4040689dd715de0e0e605":{"kind":"REWORK","_number":12,"created":"2022-10-03 08:01:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/12"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-03 08:00:42.000000000","tz":120},"subject":"CI/centos: use repos from opendev mirror","message":"CI/centos: use repos from opendev mirror\n\nCentOS Stream 9 repo definitions use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01c6cf409d8df3bc98e4040689dd715de0e0e605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01c6cf409d8df3bc98e4040689dd715de0e0e605"}]},"branch":"refs/heads/master"},"b18d48848d3487bc3b15ac8b7880444c1485ac8e":{"kind":"REWORK","_number":13,"created":"2022-10-03 09:44:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/13"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-03 09:42:56.000000000","tz":120},"subject":"CI/centos/rocky: use repos from opendev mirror","message":"CI/centos/rocky: use repos from opendev mirror\n\nRepo definitions in RHEL 9 rebuilds use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nSimilar is done for RockyLinux 9 builds but there RL9 repos are used\nfrom official mirrors while CS9/EPEL ones are from OpenDev mirrors.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b18d48848d3487bc3b15ac8b7880444c1485ac8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b18d48848d3487bc3b15ac8b7880444c1485ac8e"}]},"branch":"refs/heads/master"},"0b7fcd298098e000d0461707a43c801ae3fbf67b":{"kind":"REWORK","_number":14,"created":"2022-10-03 10:49:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/14"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-03 10:49:45.000000000","tz":120},"subject":"CI/centos/rocky: use repos from opendev mirror","message":"CI/centos/rocky: use repos from opendev mirror\n\nRepo definitions in RHEL 9 rebuilds use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nSimilar is done for RockyLinux 9 builds but there RL9 repos are used\nfrom official mirrors while CS9/EPEL ones are from OpenDev mirrors.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7fcd298098e000d0461707a43c801ae3fbf67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b7fcd298098e000d0461707a43c801ae3fbf67b"}]},"branch":"refs/heads/master"},"e9e23491c02e51464f1d490307f8f28fb62513f2":{"kind":"REWORK","_number":15,"created":"2022-10-03 10:58:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/15"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-03 10:57:53.000000000","tz":120},"subject":"CI/centos/rocky: use repos from opendev mirror","message":"CI/centos/rocky: use repos from opendev mirror\n\nRepo definitions in RHEL 9 rebuilds use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nSimilar is done for RockyLinux 9 builds but there RL9 repos are used\nfrom official mirrors while CS9/EPEL ones are from OpenDev mirrors.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e23491c02e51464f1d490307f8f28fb62513f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e23491c02e51464f1d490307f8f28fb62513f2"}]},"branch":"refs/heads/master"},"0c523e7f1b10ab0b456c20f4e900a6c8c6f7dcea":{"kind":"REWORK","_number":16,"created":"2022-10-03 13:29:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/859463/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/859463/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/859463/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/859463/16"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 17:21:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-03 13:29:05.000000000","tz":120},"subject":"CI/centos/rocky: use repos from opendev mirror","message":"CI/centos/rocky: use repos from opendev mirror\n\nRepo definitions in RHEL 9 rebuilds use metalink method for most of\nentries which makes our use of sed obsolete.\n\nThis patch adds file with definitions of all CS9 repos we use and\nmoves all official definitions a side at start of build. After build our\ndefinitions get removed and official ones are restored.\n\nSimilar is done for RockyLinux 9 builds but there RL9 repos are used\nfrom official mirrors while CS9/EPEL ones are from OpenDev mirrors.\n\nFinal images will have repos in wrong enabled/disabled state.\n\nChange-Id: I04121fd5ca9894b6deeda3808808cbffd6b97811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c523e7f1b10ab0b456c20f4e900a6c8c6f7dcea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c523e7f1b10ab0b456c20f4e900a6c8c6f7dcea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
