)]}'
{"id":"openstack%2Fkolla~859801","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Iad3c7270579ffb38299269481ac64d5582abaa22","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-07 12:00:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad3c7270579ffb38299269481ac64d5582abaa22","subject":"curl: Add retries","status":"MERGED","created":"2022-09-29 14:21:31.000000000","updated":"2022-10-07 12:01:49.000000000","submitted":"2022-10-07 12:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859801","cherry_pick_of_change":853398,"cherry_pick_of_patch_set":3,"meta_rev_id":"f5b4dd59c74cf197d0cd374de3fbff0b51977a63","_number":859801,"virtual_id_number":859801,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-07 12:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-09-29 15:48:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-09-29 14:22:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-09-29 15:48:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 14:22:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-29 15:48:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-09-29 15:57:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1795a4eaac4dc6293670d9d502530c16fd013bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 14:21:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b5886992ad5ef891179fb40f65a02ec65efe2650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 14:21:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd7a1039c7c17edcd00cd8d2e2235cbb7ce412d4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 14:22:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1f9d2e4c5c044095bb4a888e5a4978fe2d06d62","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 15:48:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4dcf05694a93ba0e9a7fa4cef8845f83dfb8333e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:57:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84d22da1d2d04ef98602d6db861e5c50\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/49e16f11d5364167a69b95d378040dd9 : SUCCESS in 30m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/577e36919b2a4940aca22ef11a648d38 : SUCCESS in 28m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8dd62ea0a39446e3bef76e9164b81179 : SUCCESS in 1h 08m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/40677accba6b4d0db7692df546a4ec5b : SUCCESS in 1h 06m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7276db4d6599412fb554490090e4985c : SUCCESS in 1h 30m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4f54c2b4d106423eb4fb8173667790e3 : FAILURE in 22m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/78386edacd5c44259d237ac21a4acbca : FAILURE in 14m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cbc9c34cd69f4163a05ef9e093406b48 : FAILURE in 34m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8c99f60d16bf4331b7ccab3ca6c95665 : FAILURE in 23m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8520ae80224485906ab13b7aee2245 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4fd448e4554831a577a72c2721ae48 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11cce70f738647da9ae4067998d146a3 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b356bb64d96a40449383cf6cd37c05db : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42407745008f4dd59b867fc9c1798b19 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5aa797416794e28869c5c1a75bf05ba : SUCCESS in 16m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/509fb9ed04a848d798121290b4516002 : SUCCESS in 18m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/193e499882654c13986c1ba195933976 : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d1238ffe4c847179bbf5f1a915aca07 : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b737273f6662404a9c5e7987fff9a030 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d9a327aa3fa86d6da6f4dad00707450382de072b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:39:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18619ea62ff4cc9afc16a1036241c90\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9248b5cb0c046cd84d7d740b10eccca : SUCCESS in 1h 21m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/100e32df13cd4b9d90ee04e5fed2bfa6 : FAILURE in 2h 15m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f65dfe201064a279e911c5427d0c25e : FAILURE in 1h 03m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e3ce7bf7d30405389c02088f674668a : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9c0f2d62bf34a748a60baf564ff1a1c : SUCCESS in 14m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a1e7dd96e7f4a728926192f599bc757 : FAILURE in 1h 49m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ee1aad759ed77be44a802ed8565a7919b74c700","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 19:09:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc29799a2c51207849e818f9e8c61408b79afd7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-04 13:00:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71fed26152816e43b263aff6167a171ef26e210b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:26:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b8656f6dea452e8d85b2aba58c09b3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1bad35f9dd9456880fdf66d527d5ced : SUCCESS in 55m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72a54944c4a94138ba10ad850c0a2d43 : FAILURE in 1h 12m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc7802ab4c4f49748eb648e6f21cb462 : FAILURE in 46m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/047487d0009e4e96bac58fc4ae6215d6 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/964f0bde621a4077ae4d43a62f3667d5 : SUCCESS in 4m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c0a49906e1a4e39bc1e22f621ca7d04 : FAILURE in 1h 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b2cd8411131275d6e610e3d4de99bd113247749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:35:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/751db0d5af304886b39a05ca480e5631\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4da7543a775d4fa9966e1a632e208e98 : SUCCESS in 30m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/91d5b0dcefaf491092bb4d80cc56bd23 : SUCCESS in 31m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/50ae294598964de29e581f999b3de75b : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bf9dd4bcf01b438f8256c1957606b547 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fab57810117946859618a756fd1a5ef8 : FAILURE in 1h 22m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee246b17cb634a1387f26488a2298cf3 : SUCCESS in 18m 58s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/57f91625646940c7a6ffdb47b3093d98 : FAILURE in 15m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ab08ef7577a4e3ea324e572eecef6c2 : SUCCESS in 57m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5e0ed115cb23491fab58ac7dc7efe3e0 : SUCCESS in 1h 04m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0df265bbff7c4caebe7d5e90874d4ee4 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43139bd437b445fdaec2d4d7f8fa0a7d : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e45fdf335041119c9bf1031b864993 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de883390e0a4f3db0b627b01f67311b : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3dae12dafff47deb0b3828ec6ac3582 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fdce4f43dc3c4aefb54c4ba8c99c153d : SUCCESS in 18m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52a0cecd41d94a6c813d8563dad0d5a2 : SUCCESS in 22m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af23eb7218b2499d82d0cef7ef6f762e : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41f6a2311d0c4129a00115e2430f86c5 : SUCCESS in 48m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/841348beb7ca44f593fe1ab3be3bfd25 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1c71f032c17c6eb122da5ac2d06d1847f7f00c54","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-07 09:02:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5559ca20ce4772f5df0f43fdf783fb397d68884","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:22:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0b6ed7cc084d569010c0a88bb84616\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd97b38c5e5c4beaa682049edb48fd7e : SUCCESS in 1h 17m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5ee35c6955347bb9b5e88e966f2a67d : FAILURE in 46m 18s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c909a14bc634944945b07c82fd3ca5e : FAILURE in 42m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea785e4aa0794c4d832f83b3f3ed5edb : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fc78da8da474f97af19bb4b38f5e210 : SUCCESS in 6m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1cc3416c3caa4d5ea65b55a50fae1285 : FAILURE in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38831621a7d93c3913daba1f6acaa3e5cf482741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:30:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d24d96c61fc47658e6809120b9e9535\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/182594e5cd4f4581af1926fa3a38bd6d : SUCCESS in 28m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f722be807fe7407197035600192fdcbb : SUCCESS in 32m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/13af0e9ed4eb4471ac818634d5600f34 : SUCCESS in 1h 15m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cb95047a1c93487093a63fe9326a6447 : SUCCESS in 1h 11m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7ab249630a4d129396cce2f545ba7f : SUCCESS in 1h 23m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/510e3177bf3a47d18c108a1fb1f20949 : SUCCESS in 21m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aa102382d36e45e1b3bd4d275f3494ef : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf03a8bc665d43a6aa9b121cd60deda7 : SUCCESS in 58m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7e9740d5eb5345ebab46f45c9caa8fd2 : SUCCESS in 1h 08m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826fdcdf5446407ab8ceb9677cc41035 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4afe1f01f2b145b2babd6b19eddfab99 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb91e3eeba7a4f028edd1331ffdf653e : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f594ce4296fa4770bc5e483411223f6f : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a2140ada5f94af4becf7cbc43747b15 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01d7168f05264dab9a92d27536fee72e : SUCCESS in 16m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be0e90eda9d2411a8e7e749bf6f59482 : SUCCESS in 19m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aaa3724efae24be2909970cbce37b8e8 : SUCCESS in 50m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c8bf4954e3c4be3b964116b3fa2ac00 : SUCCESS in 52m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1bd39c02cdbb412d8a3027b6e53561ca : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2637a207887d8daffb90326d61b03347e2972053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 10:31:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d72d0c1911272a8adb51a8f868b2e643eb62580","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 12:00:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0cb6ecbc234c87a74122223fb321db\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a26c3858a53b489b904d34963d5424f8 : SUCCESS in 26m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/564018118e71416eb52ee8644148fcfd : SUCCESS in 1h 01m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/41f78915a89d4787b973b381377e0459 : SUCCESS in 1h 27m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4936e3f991754ea0b07fbc76bed3426f : SUCCESS in 18m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54c77d8899d74f1b8a8bfce7e33e18df : SUCCESS in 1h 04m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3eff86925ea948c1bbdab4a2df16de37 : SUCCESS in 1h 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f82304a65c48bbb1c4f987b5ac3612 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/004727ccecca4017a37156936a6101f5 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fbb5a9164d84262917450f34bd334f3 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f996f9f670c4cc29dbc2f1a139d9561 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e24869fb1c4ade92f050de505c8ed9 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2f962d6e62c4c54ad4ad791bd0cac79 : SUCCESS in 23m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5778e29b65e3473781c58803755e9cfe : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a70e5362a094cac859b3066ff91a660 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":2},{"id":"06eabbc3c37fd1235c0e9327dd982dfad2908989","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 12:00:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f5b4dd59c74cf197d0cd374de3fbff0b51977a63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 12:01:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd2d9d6fa704501a1e4fc5b6a0551fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43fef77f39747e0aad1cf3937e29954 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/469e4c09ce604adeafeed935054d2a9f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcfdd7f53ec6d8d5396a29e27d4e4715f3abdb1e","revisions":{"1f53ec4401bdb5026a7599e1f8aabf2b9bc454c2":{"kind":"REWORK","_number":1,"created":"2022-09-29 14:21:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/859801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/859801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/859801/1"}}},"commit":{"parents":[{"commit":"49376fa851ffd49cbfe5b24b0e5c594dcd8ddc3c","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49376fa851ffd49cbfe5b24b0e5c594dcd8ddc3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-17 05:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 14:21:31.000000000","tz":0},"subject":"curl: Add retries","message":"curl: Add retries\n\nChange default 0 retries to 5, that should help for transient\nerrors (timeouts, HTTP 408, 429, 500, 502, 503 or 504).\n\nChange-Id: Iad3c7270579ffb38299269481ac64d5582abaa22\n(cherry picked from commit fd14318a8bcb5d5d7c3696af11059cc4137ae9ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f53ec4401bdb5026a7599e1f8aabf2b9bc454c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f53ec4401bdb5026a7599e1f8aabf2b9bc454c2"}]},"branch":"refs/heads/stable/yoga"},"bcfdd7f53ec6d8d5396a29e27d4e4715f3abdb1e":{"kind":"NO_CHANGE","_number":2,"created":"2022-09-29 14:21:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/01/859801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/859801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/859801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/859801/2"}}},"commit":{"parents":[{"commit":"49376fa851ffd49cbfe5b24b0e5c594dcd8ddc3c","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49376fa851ffd49cbfe5b24b0e5c594dcd8ddc3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-17 05:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 14:21:45.000000000","tz":0},"subject":"curl: Add retries","message":"curl: Add retries\n\nChange default 0 retries to 5, that should help for transient\nerrors (timeouts, HTTP 408, 429, 500, 502, 503 or 504).\n\nChange-Id: Iad3c7270579ffb38299269481ac64d5582abaa22\n(cherry picked from commit fd14318a8bcb5d5d7c3696af11059cc4137ae9ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcfdd7f53ec6d8d5396a29e27d4e4715f3abdb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcfdd7f53ec6d8d5396a29e27d4e4715f3abdb1e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
