)]}'
{"id":"openstack%2Fkolla~859802","triplet_id":"openstack%2Fkolla~stable%2Fxena~Iad3c7270579ffb38299269481ac64d5582abaa22","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-29 16:56:42.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-29 16:56:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad3c7270579ffb38299269481ac64d5582abaa22","subject":"curl: Add retries","status":"MERGED","created":"2022-09-29 14:21:58.000000000","updated":"2022-09-29 16:57:58.000000000","submitted":"2022-09-29 16:56:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859802","cherry_pick_of_change":853398,"cherry_pick_of_patch_set":3,"meta_rev_id":"71c440f4414a7a9b51f4e2c34bac17d9c13fb045","_number":859802,"virtual_id_number":859802,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 16:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-09-29 15:49:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-09-29 14:23:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-09-29 15:49:02.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 14:23:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-29 15:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 15:49:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"4dba2e8b24d10e574f11311825dc06dae4cb221d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 14:21:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"11775b54093b562056e655c753b2a733e6ad348f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 14:22:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c6e4212d5b0a64bf85da8e8948a69da19ddf3286","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 14:23:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee107aad0ab65d975d3ef30547d4d1442e2da60c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:36:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b321001355d14765a6f4f4d9d27c6ef1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1839c76e8a8849cf96151d0dc2b83771 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8f31380f3f524bfca63da1d0fc7f9668 : SUCCESS in 25m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4f7845b73c2644d2a68f4c6431c13c66 : SUCCESS in 57m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/223be9fcedf84917b810354170a64d9d : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f8d4fb9a98654eeda318dfefad565540 : SUCCESS in 14m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f55b1cee639e4fff908e683a60b045ae : SUCCESS in 18m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4caf779ffff042a3a9f884f27da77e27 : SUCCESS in 1h 00m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3468219660498782ccfce3eb9a4c01 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d9b88ec2ffb41e2b3245b050c017319 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e773984221ff4d528f40e6d80b91f220 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/710458dc6bf446b591e8ce88c4324dbb : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1509d84116be418dbd97f81ed5b6b7e7 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd8411df8d34d63aab0b8dc4a6dd62b : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fbc835916324f099f0d107743a57965 : SUCCESS in 20m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fad72103f1b34795b9ed228762174d3f : SUCCESS in 19m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ba58c4aa0974b6c9a0c0a129f63ed0d : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf39d8f90de243aa9a518b31baf1f4df : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c067d7037e7342cf99a531241c2e8fac : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b9c3249528dab959460e21d85a43bc218cdbf92a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 15:49:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22124ce5f839b622d6255b950404af155fe52562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:49:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"712086d70fce08a59cec4687990c68b0a32d6b5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:49:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a42504f6f0c64418b82b8421c17b1f77\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc00baec13dc451c9a919190dd3e5b09 : SUCCESS in 2h 21m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e41dd794c7744efb6c43422b46b5d15 : SUCCESS in 1h 12m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71f1cfeb9a04498ab1eeecbc14e333c6 : SUCCESS in 2h 25m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a40cec4bb6544bdfac1b54fb3ff9ab2f : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44cce2bf46804f08a8638595c0b08c46 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/16799f0157f14096a7dc06a99b603a1b : FAILURE in 1h 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64be64088e4a5d56433f27f3e0a14382fbeaa671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:56:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb21c3c1ddc4581b4ffd0c2c0313f77\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2ca8bc3dbabb4a628bc74d62fdc344f9 : SUCCESS in 26m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7c119f69a51645e68532ffd79fc6fa3f : SUCCESS in 1h 06m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1cc15797eb4c308c9fccde94acbd1c : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c811427b8f12468d987fcdf24a832bfc : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4d405758ea488d9783816416ede345 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6390f3059b7d4ad29fc212a4b3d18fd5 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ef3fa5ac924451c903bef0088da4531 : SUCCESS in 4m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bad4bd6923de43cc9cf9e625ddae2e50 : SUCCESS in 21m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5aaca8c57fec4b7dabc3aab771b8429f : SUCCESS in 53m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/acbb9b25e8e94d46bd90d9eadf8c29ca : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":2},{"id":"51b27b3441a590193f218d8a8eb13ac848390863","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:56:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"71c440f4414a7a9b51f4e2c34bac17d9c13fb045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:57:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6bd4452f67e4f219a0a199d216b574e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd928a1580704d728ba319c6a5ed77df : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d1fabb7eb6641388901c91bf928855e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42700cf5a68b39d2de6cedbd6651d7858214d99d","revisions":{"8afb979e2cf685c4284de7d26697d30a5089f175":{"kind":"REWORK","_number":1,"created":"2022-09-29 14:21:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/859802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/859802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/859802/1"}}},"commit":{"parents":[{"commit":"1011fc60c333d9d2e6276ebdd20e952092ef49d8","subject":"Fix CVE-2022-38060","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1011fc60c333d9d2e6276ebdd20e952092ef49d8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-17 05:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 14:21:58.000000000","tz":0},"subject":"curl: Add retries","message":"curl: Add retries\n\nChange default 0 retries to 5, that should help for transient\nerrors (timeouts, HTTP 408, 429, 500, 502, 503 or 504).\n\nChange-Id: Iad3c7270579ffb38299269481ac64d5582abaa22\n(cherry picked from commit fd14318a8bcb5d5d7c3696af11059cc4137ae9ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8afb979e2cf685c4284de7d26697d30a5089f175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8afb979e2cf685c4284de7d26697d30a5089f175"}]},"branch":"refs/heads/stable/xena"},"42700cf5a68b39d2de6cedbd6651d7858214d99d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-29 14:22:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/859802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/859802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/859802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/859802/2"}}},"commit":{"parents":[{"commit":"44aba54a4a81102fcac55ed9fb4bf92a0ce9df36","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44aba54a4a81102fcac55ed9fb4bf92a0ce9df36"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-17 05:09:17.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-29 14:22:58.000000000","tz":0},"subject":"curl: Add retries","message":"curl: Add retries\n\nChange default 0 retries to 5, that should help for transient\nerrors (timeouts, HTTP 408, 429, 500, 502, 503 or 504).\n\nChange-Id: Iad3c7270579ffb38299269481ac64d5582abaa22\n(cherry picked from commit fd14318a8bcb5d5d7c3696af11059cc4137ae9ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42700cf5a68b39d2de6cedbd6651d7858214d99d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42700cf5a68b39d2de6cedbd6651d7858214d99d"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
