)]}'
{"id":"openstack%2Fkolla~859803","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Iad3c7270579ffb38299269481ac64d5582abaa22","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-29 17:10:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad3c7270579ffb38299269481ac64d5582abaa22","subject":"curl: Add retries","status":"MERGED","created":"2022-09-29 14:23:42.000000000","updated":"2022-09-29 17:11:37.000000000","submitted":"2022-09-29 17:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859803","cherry_pick_of_change":853398,"cherry_pick_of_patch_set":3,"meta_rev_id":"2b4c62b813ab3279f7a0f70ce77a6c9f2706d2b7","_number":859803,"virtual_id_number":859803,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 17:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-09-29 15:49:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-09-29 14:25:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-09-29 15:49:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 14:25:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-29 15:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 15:49:11.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"71a894389716b949513a5ffb0bb6b76676acb757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 14:23:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b2eb970991af58efb708f8214be41ef9d13f6ae7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-29 14:25:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d172a55ff879f01e0daa221dba4be3131294219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:49:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cdef8254afe43ff8e44329856cfbb1b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eb8437bdaedc40388f527f252e2dc866 : SUCCESS in 30m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1dc2037d1105426db894edbd343aebc2 : SUCCESS in 41m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d7dd24f1ac3c4ffc977d7ab3a7861af9 : SUCCESS in 1h 08m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1ab80ba442a4a88af083778c7ea4439 : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ae712186031748b1869ecf4ca58796d7 : SUCCESS in 1h 20m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8e9c00f92edf426ebdd7ad29c6ac6d76 : SUCCESS in 18m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c4d01ec3463146cb9661d922882754f1 : SUCCESS in 16m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/999f0fd281454736996f3620bfbbf2de : SUCCESS in 54m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7ccb5e2ef044779f8fcb5ca264cd78 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/858d165f85884b29be0522de8639058e : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78eb0348651b4f4ebd33bd7bb1a3dd15 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2183296f0feb49048046d3bc27970269 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a80803fabcf4ed49a4c8a85f1717b75 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d81196bc3b67470496f7e1d881ef436a : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cac426a75d640a1a755995a22bac0b9 : SUCCESS in 20m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6aeddc8e4704487f97190e2a6c99fb53 : SUCCESS in 16m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82af2fd198e64949bdc28043402fc72a : SUCCESS in 55m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dab0dfae76664795bb169c3cd9712ff5 : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f21ac66be6d45b397a3002f834d1097 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3bc92c26344b0b0f310fe0f3da09cf68cd33b8cb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-29 15:49:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4493fb4dfc4deaa282f91a209c175b2acd50645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:49:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e519cd0bfe7bf712fcde757cde2d3c02b555adfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:48:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d67d0971b2b463e8ed2f2d3b0790101\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/72c9c65604ba455c96e1126b7562f7d0 : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/010104d636a4452698a638d625307c38 : SUCCESS in 1h 41m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f923ab9ac875429fa8a5b2cf7c8b8478 : SUCCESS in 1h 50m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/92654228ddf54697bbde65be57c57cb5 : FAILURE in 2h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"370489dd24bee2b8b1efd41a92cdba75f4ab18d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:10:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056346fc40ce4e63ac77ae83014ee944\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/edbdbc57807d4ded8864f84b91af0712 : SUCCESS in 30m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/913b35b328ca4719a5183d7de616bb15 : SUCCESS in 1h 06m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ee70a871fd294fbbba6c57f49a0f1c97 : SUCCESS in 1h 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff9fc29eed2423c9dde8c560731dd6c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b7773511de40bca52214af4a29d18c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d4cb2be961b467fb33f41afe49b0455 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a527d072cb540b59eb04a314f1a3100 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6751e4a63ac94a348412c72250126bc2 : SUCCESS in 4m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4938db5ae7941e48cdac9c4707ff3e6 : SUCCESS in 21m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75ee4ec12e0541e99df47d66cebec8b9 : SUCCESS in 47m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a936ab30207c494ba18bd26c62095996 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a2449a9294f0d4b6560ea1bd694c7e3307750fcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:10:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b4c62b813ab3279f7a0f70ce77a6c9f2706d2b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:11:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126fea29c1604ceb89e1c65636f5f291\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529b71118c2045938df0eba1f3119ec7 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b753c0c3ed447f38f361fc928b47a18 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6ee3a29179fcfcff2043f791b91dc8016656e8a","revisions":{"a6ee3a29179fcfcff2043f791b91dc8016656e8a":{"kind":"REWORK","_number":1,"created":"2022-09-29 14:23:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/859803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/859803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/859803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/859803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/859803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/859803/1"}}},"commit":{"parents":[{"commit":"1aa2ddcf492a9af95ff649b25841e92b0a78a61a","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1aa2ddcf492a9af95ff649b25841e92b0a78a61a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-17 05:09:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 14:23:42.000000000","tz":0},"subject":"curl: Add retries","message":"curl: Add retries\n\nChange default 0 retries to 5, that should help for transient\nerrors (timeouts, HTTP 408, 429, 500, 502, 503 or 504).\n\nChange-Id: Iad3c7270579ffb38299269481ac64d5582abaa22\n(cherry picked from commit fd14318a8bcb5d5d7c3696af11059cc4137ae9ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6ee3a29179fcfcff2043f791b91dc8016656e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6ee3a29179fcfcff2043f791b91dc8016656e8a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
