)]}'
{"id":"openstack%2Fkolla~860081","triplet_id":"openstack%2Fkolla~master~If077d60e9efd29fab80f4f5188a4f54015734a92","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-03 14:46:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If077d60e9efd29fab80f4f5188a4f54015734a92","subject":"proxysql: Use almalinux for centos/rocky9","status":"MERGED","created":"2022-10-03 10:53:47.000000000","updated":"2022-10-03 14:47:37.000000000","submitted":"2022-10-03 14:46:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860081","meta_rev_id":"3bf48dd91384fd16ebf109a8a80da91e6efc0054","_number":860081,"virtual_id_number":860081,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-03 14:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-10-03 13:24:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-03 12:01:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-10-03 13:24:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-03 10:55:02.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-03 12:01:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-03 12:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-03 12:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4922b5f4c1ebf45521ef2a13d7eab31f649b386f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-03 10:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fc2b6b238716a38425411de47275f4ee3d09caa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 10:55:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"787275b98d4335d01aa7d1bd340ce7a55ce80d7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-03 11:30:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96ecdc9bb700a0c0449230dc476e5c9250f7f650","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-03 12:01:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3ccf6025393d7089a9d5f2babde605a2457f9fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 12:42:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9595043c70c4acaaba93ee3d22af3c6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/45a26688654441fcbbd19e1caa1fe788 : SUCCESS in 1h 08m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cf194647f414927bdc299ae6c57244a : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40f1087f495e4ca98595b48e02731cb5 : SUCCESS in 3m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1813b82582d34173b2a5f799fa63ba63 : SUCCESS in 45m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6e56c33a95de4fd9bbced14e7f71acbe : SUCCESS in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c8e845e446a08e2849c9d93dee4559394a87f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 12:47:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/666c9a9c08dd477aaa63ecdf97ebdf53\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4c6e0691f3574c1697544576e62f3831 : SUCCESS in 20m 49s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/715b9eaa4e6b46c1b2ab5478887d7c9e : SUCCESS in 1h 14m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec23157671f44883809f5268e10e07be : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c08ee0530a48485c94aeedced4456479 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d72d79476fc4469086e3ee77744fb311 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d98bfd970d1403baa42c4bfb94f6610 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1285f3950ee74f16a61d38b8e417dfeb : SUCCESS in 5m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71cca051b89a4a84b6f01345ef300a59 : SUCCESS in 21m 22s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6e4e13aa834147c58d9fd708efb77a79 : SUCCESS in 1h 04m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/49bf839e3d184b45af7aa8f2e48351f2 : SUCCESS in 16m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eee62262f7d045dcad2617036d7336e1 : SUCCESS in 44m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d5d466c8a9d493fb83ca2bb7ca61921 : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":2},{"id":"cf58dd908b6ca684162552fc33fb91658a13e3f1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-03 13:24:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91a033915972fc5dac6bfa4829eaddf6efc04ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 13:24:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3991ef7153b85b8793a203d2dbad270da39a83b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:46:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce80cf9a61074888acd6023a25d8ebaa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c77d645177d045bcb9cbd67c252ebcf5 : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/31b8dc261e954f66be446edcb3c97cd8 : SUCCESS in 1h 21m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e215818355d94a05ae04ccefc51237c0 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aad3a50dee554575af122b364b552219 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8858c99b699f4b9b9a799ddb60ed88eb : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/449521e5a91c447a8dc3a604067fd772 : SUCCESS in 6m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d716dfa83e864a34bff7ce11cf45f08d : SUCCESS in 23m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ae623fc6e649446dabe9c712f1476043 : SUCCESS in 54m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ba818850ce44badb24ff223093ec205 : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1fbb17c69124e16ae9b7292cffacb62 : SUCCESS in 1h 01m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/40927ca2f13c477b94371b36ed281873 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"906ad2750cc9235d54a92b974e0ea5755c3c1199","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:46:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3bf48dd91384fd16ebf109a8a80da91e6efc0054","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0d0c8732794481a9e97ce34f461e55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f6a40e27724c9fa77ec27263e6f97b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f0849eca8df604c03ea8f589f5a331aff43c2ae7","revisions":{"2bb82de5a6033444f5b500a36861fa85d82b66fd":{"kind":"REWORK","_number":1,"created":"2022-10-03 10:53:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/860081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/860081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/860081/1"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-03 10:52:37.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-03 10:53:46.000000000","tz":120},"subject":"proxysql: Use almalinux for centos/rocky9","message":"proxysql: Use almalinux for centos/rocky9\n\nAs per https://github.com/sysown/proxysql/issues/3989\n\nChange-Id: If077d60e9efd29fab80f4f5188a4f54015734a92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bb82de5a6033444f5b500a36861fa85d82b66fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bb82de5a6033444f5b500a36861fa85d82b66fd"}]},"branch":"refs/heads/master"},"f0849eca8df604c03ea8f589f5a331aff43c2ae7":{"kind":"REWORK","_number":2,"created":"2022-10-03 11:30:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/860081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/860081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/860081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/860081/2"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-03 10:52:37.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-03 11:30:43.000000000","tz":120},"subject":"proxysql: Use almalinux for centos/rocky9","message":"proxysql: Use almalinux for centos/rocky9\n\nAs per https://github.com/sysown/proxysql/issues/3989\n\nChange-Id: If077d60e9efd29fab80f4f5188a4f54015734a92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0849eca8df604c03ea8f589f5a331aff43c2ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0849eca8df604c03ea8f589f5a331aff43c2ae7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
