)]}'
{"id":"openstack%2Fkolla~860222","triplet_id":"openstack%2Fkolla~master~I13e3cf5a02dadcedd1ebbd54d56f7f633e6e3d68","project":"openstack/kolla","branch":"master","topic":"ansible-bump","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-05 14:24:02.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-05 14:24:02.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-05 11:34:10.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}},"hashtags":[],"change_id":"I13e3cf5a02dadcedd1ebbd54d56f7f633e6e3d68","subject":"kolla-toolbox: bump ansible-core to 2.13","status":"MERGED","created":"2022-10-04 09:04:30.000000000","updated":"2022-10-05 14:25:17.000000000","submitted":"2022-10-05 14:24:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860222-ansible-bump","meta_rev_id":"469c43a1809eaf55d3eaeeb4f5afe800d66982f5","_number":860222,"virtual_id_number":860222,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 14:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-10-05 11:34:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-05 13:10:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-10-05 13:10:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 09:44:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-04 10:04:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 10:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 09:53:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-05 09:53:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"bed6af196bab1a460d2d33aabb8f073cd39d0f32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-04 09:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdbd6a27c1176686845a79509a1bb98385a42f26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-04 09:44:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e2f3ec5fe3a99c1c149056e1f713f794c68e27e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:04:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aad82f523655411dabfdc5ae2b9c057e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e2efee4cd96e4940aafed3b2a5b26895 : SUCCESS in 52m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d847918ef62144ada03439187841c4ba : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a50b2204cdc44995b39eadb55c0e16aa : SUCCESS in 4m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/42d339c7944d495cbf232d44cd48263b : SUCCESS in 58m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/de0f4bf96e78471589ba49b20d79c407 : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e36d22644f54b94bf09e4082ef931cd54bd8513b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:25:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27379a7aec4a4524835e2d9ff7d3c075\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6b09fe42053d42cf842dff9833f13fbc : SUCCESS in 30m 33s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/61d862d9394141a78e76a76cb5b0cecf : SUCCESS in 1h 16m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780e7d27b34e4cb4b6592e16b858bbd9 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259c50430dbd434bb6a6aa86b27a9256 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c87029fcad794b979ca9e89e81b68eb8 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed70bdf9eaaa4c2c85d89caf0848ae59 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda9952ca8a8463eb007b1d4380f5ca3 : SUCCESS in 6m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2be527de986d4578b5dba808b624591a : SUCCESS in 30m 26s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/060727a5d357400081f3ad9b00dcb2f4 : SUCCESS in 1h 11m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b0dcba2e9ee1412db4e816b88ab188f8 : SUCCESS in 25m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/801a2af9f22e4a7d8e7b58823ed347a5 : SUCCESS in 56m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ed8fa690c644d5cbd9b8944e84068c4 : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d0925b5a490dd0b841b2c37fde8c2f39be113c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-05 10:01:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa3275f0f27ef6db697773c9f94222c94786cf95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 10:44:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80cd8e1247b848b0b213bb208e7d99f8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/976b4b2442094d69bd77fc530bf57781 : SUCCESS in 30m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d9543abaa4f4d02a634e69ba2164f5a : FAILURE in 26m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/74b80331adc24a37b03d77cd8777f3ab : FAILURE in 35m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05748d868f3842db954fbcae9736e0e4 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5006c6746cf94523a0837ff0c6a32ac6 : SUCCESS in 3m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e188117bdd1f4231b77d14311141b683 : SUCCESS in 41m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc690885f513455f8bd26d932dbe1f97 : SUCCESS in 27m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"161f52eff025794fd256a046a442d4cc3cd2ec8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:15:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3350576903be4c71ae2cddf95b64c636\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/31181a122e424adfab455d5ced626728 : SUCCESS in 20m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/405bd1d9660e4c9abb0ddc09c5571edd : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf848dc96e3740ca948b943fb3ebfacd : SUCCESS in 16m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ea0dab1f45574791a15d27b3dcbcc929 : SUCCESS in 49m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e267783684bc4dc9884a48645d3dba08 : SUCCESS in 1h 09m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1972940548a4f958fcc094e7d8fa9c5 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32e638c0b02348ee83b70fc3dbbc66fe : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/725c1cd0618f4b8098c751bc8e5a7cd0 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10ff95fc61c9446fb6559f40c202964a : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f426954af4496995a95c64d60154d3 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd56fc260eb54b8b9f7c0ded8bb71b2b : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/42f0c978e7eb4af09314b70e8bb8d6b0 : SUCCESS in 20m 57s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1d5aed046677491f9380713a2e6d7113 : SUCCESS in 1h 01m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d7661d088716475fa81a35b7321d8cd1 : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a38e03ca3de459fbe805289a846036b : SUCCESS in 57m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70ed6cea512a49a1b667b0b7238fbf29 : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":2},{"id":"decc40b65727a31491577eaddaeb13616d5d6627","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-05 11:34:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"003a01837e292ade488ccbb92f0275c77614a51b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-05 11:34:47.000000000","message":"Topic set to ansible-bump","accounts_in_message":[],"_revision_number":2},{"id":"9e9873846e3bf57836fab517fda10259b853f6af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-05 13:10:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1503b0bea318f566371413f7d9ccdead6ab19b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 13:11:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9db12fb7fc7fbdf7f83199b193f5a9ac8822d128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:24:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15f987624f34f849ae05b5cf04019eb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/427b0a95c01b4b3db718cb155a89a5b0 : SUCCESS in 25m 25s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0eca60464d234a178282c53436b4905b : SUCCESS in 55m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7eea849256c64e22ad07858b97690dfc : SUCCESS in 19m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/04cefcbfebe6455b8156aabea3ee7eb0 : SUCCESS in 1h 06m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/18cf874f320c48ee88f9279065d4ab5f : SUCCESS in 55m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87f8717940e436caf49e2f968ec80cc : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8173ed62097549da9f2c33022dfaa301 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c64202abe9a4f8882eede35540b570f : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb808f99ad154179b84c18a164f277cb : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abba1633a03f4ef0baecbc7f5d3cd20b : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db94e048bc6e4185a66142503e1459c4 : SUCCESS in 28m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/11ef0f12ec144c3a8cfd70108d108f92 : SUCCESS in 1h 11m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c1d12024b84e4e9faefd8af57618ceb1 : SUCCESS in 23m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f8ba51c797b42a295900c1821db5d3c : SUCCESS in 47m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca772232594d40f29ef9fedd8d868768 : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e5e05ee76aeff6b552b2e5db4d1d832b9ec73415","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:24:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"469c43a1809eaf55d3eaeeb4f5afe800d66982f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39d96e8f0fef4c20b9b5c0a75edcf595\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3487926b0b374d48999fc90fe324c9cb : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5698849c7d994006892ea83d6167de7e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f07d8f7f679103d91c0864c38cea4bab03c6ac9","revisions":{"18f7b9a398567239a342f538882d72d2e0842b7a":{"kind":"REWORK","_number":1,"created":"2022-10-04 09:04:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/860222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/860222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/860222/1"}}},"commit":{"parents":[{"commit":"a9b0b63fedb5140ea6381c509c4dabde9772d5c2","subject":"Merge \"proxysql: Use almalinux for centos/rocky9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b0b63fedb5140ea6381c509c4dabde9772d5c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-04 09:04:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-04 09:04:29.000000000","tz":120},"subject":"kolla-toolbox: bump ansible-core to 2.12","message":"kolla-toolbox: bump ansible-core to 2.12\n\nChange-Id: I13e3cf5a02dadcedd1ebbd54d56f7f633e6e3d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18f7b9a398567239a342f538882d72d2e0842b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18f7b9a398567239a342f538882d72d2e0842b7a"}]},"branch":"refs/heads/master"},"3f07d8f7f679103d91c0864c38cea4bab03c6ac9":{"kind":"REWORK","_number":2,"created":"2022-10-05 10:01:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/860222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/860222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/860222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/860222/2"}}},"commit":{"parents":[{"commit":"a9b0b63fedb5140ea6381c509c4dabde9772d5c2","subject":"Merge \"proxysql: Use almalinux for centos/rocky9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b0b63fedb5140ea6381c509c4dabde9772d5c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-04 09:04:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-05 10:01:01.000000000","tz":120},"subject":"kolla-toolbox: bump ansible-core to 2.13","message":"kolla-toolbox: bump ansible-core to 2.13\n\nChange-Id: I13e3cf5a02dadcedd1ebbd54d56f7f633e6e3d68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f07d8f7f679103d91c0864c38cea4bab03c6ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f07d8f7f679103d91c0864c38cea4bab03c6ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
