)]}'
{"id":"openstack%2Fkolla~860225","triplet_id":"openstack%2Fkolla~master~Ibee94fe5a6217048eef01aa24ebed6ee05aae8de","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-05 11:26:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibee94fe5a6217048eef01aa24ebed6ee05aae8de","subject":"debian: swtpm is in backports","status":"MERGED","created":"2022-10-04 09:35:31.000000000","updated":"2022-10-05 11:27:38.000000000","submitted":"2022-10-05 11:26:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860225","meta_rev_id":"c3936c703f43a97bbbc986bd2713a606b8c2da45","_number":860225,"virtual_id_number":860225,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 11:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-05 10:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-04 10:07:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-05 10:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 09:41:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-04 11:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 10:07:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"fecb659950af51530944c9935d731eb2123f90e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-04 09:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdc4fa9fca47f77b37ac0328ab5743990b7961fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-04 09:41:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdb2dde748f1721234f12a1696c62c4f1ccc05d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-04 10:07:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5a13482a78a92b5f6dc650d46691e492ae6fec26","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-04 10:07:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9233e3e1ea8e14f20a3ca771866543b8914e92d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:42:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43a8b81075b74fbeb037dcdda8200145\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7c69b042b5544630aa0127e4cab32050 : SUCCESS in 36m 56s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/96c8930c7781405dacdeebeba8d44402 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/04bbd1319b954d638df2313b416c2c73 : SUCCESS in 19m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0732e56798614820b4af8e7f6525cd44 : SUCCESS in 1h 01m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/85cff96c8a1b4fd4885208eaf31903b5 : SUCCESS in 1h 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f604b69deb46fcbc6e3d3b0909f07e : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63dcd0f296464a97852e2ac1d7acad9d : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd6ea6d9ba1a4b069c6a25900df8aa6b : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae94014e6cb04e568200f1a63c55b595 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6092c43664444269a85b33a9de1d2d4d : SUCCESS in 5m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a7a5f0421aa044f4aedfb667fb8153a5 : SUCCESS in 36m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2563082452bc47a08af59d1400d8770c : SUCCESS in 1h 01m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e5e7ac018c524333b7ee8a08d3d8a53b : SUCCESS in 30m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de7c6d724a2548a690f9466b8b6a99e4 : SUCCESS in 54m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ef9e1b7ce61d4af79f8244db00e73e80 : FAILURE in 1h 01m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f7664e6cf07e540ae5437b48f3f0f183e8e4302b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 12:16:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238ef88157724970941e2285a059304d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8127b929c02f417d9e51aa800a1ca00e : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c43e8a869f04f90b527abe8b88999d8 : SUCCESS in 1h 40m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6627112bab7347fe89b2f59846cd5b55 : SUCCESS in 2h 06m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb211cb1852c44c5a6953258024ed92a : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a0b7d315b954f4284fef826bad05853 : SUCCESS in 7m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ee536d0e456746d6811d5642e189ebf1 : SUCCESS in 1h 35m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0980833777f3476bb65caab06d273e47 : SUCCESS in 1h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"078c351a99789e134033330cf46b5587c9af52cd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-04 15:49:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c78154f7197059ef52e330b2e4a09974f19005f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 17:06:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a98fbae2e1b4d429032acbf25081a05\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4a1d1b78c5c849b28f8ca7f689ce40bb : SUCCESS in 27m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8cacf409c3fc4edba9d30498a79d7543 : SUCCESS in 55m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0f4ce59a0fa64d43bdb197176e73c421 : SUCCESS in 19m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb22ef5f7502454ea367d8922f71fc9e : SUCCESS in 1h 01m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4f36de28b8204d6a804cb6f6750676f5 : SUCCESS in 1h 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217e3a731c8846ef9047d2da65119996 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb9d4b981e2a405f8b31bed0ed7ec00c : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84bce5e8bb0d49c99ec9f8d87701cd5e : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d0e861719694ca989cc444c3ac2ef80 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1355291d897d454e8ea9797d11bc3e28 : SUCCESS in 6m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/072030fd40584693a1bc9d56db8bc965 : SUCCESS in 30m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/80e7525438dc4fc3833bd6dfb1a3a5f1 : SUCCESS in 1h 11m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20974358695c4868acbc40f099259973 : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a600f59580504ea48e269e981c2f3455 : SUCCESS in 57m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e771b55adc9425d9dc4ed21c18b663c : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":2},{"id":"babd2e60184bb00c5534f6fd4c0aee29e11fc52c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 17:36:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef37139faabc48b2a65cb1db5615df6b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e7c2df208d1647d3ab262e0400fe57ce : SUCCESS in 30m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4bb2ff85d714a6290f8110afb476929 : SUCCESS in 1h 44m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f8bc57afc7b4a76872de2105121d53e : FAILURE in 1h 15m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f1b9202c1f34cb89e4c73d8dedb0bff : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0063bcbe5f9f4346a88a7734ae739c7f : SUCCESS in 4m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/413127d6355a4054b69659039a130654 : SUCCESS in 1h 25m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b68d88f4bdbc4859923c714d9d71dbcb : SUCCESS in 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb67f3827b4ee0510091a08fee9431cbb19b2f05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-05 10:07:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91c98a96d55e1ba757970d8f234b2f8a79ce9895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 10:08:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2aec1bf91a7a0c63bc98ffde3fefaa19486d134a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:26:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f5065814eeb44f4946ac0ddf58b47ed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f317011961b14976baabebfff26a4146 : SUCCESS in 22m 56s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cb64c4d8754147cfaa8fe1cc8ade650a : SUCCESS in 1h 10m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e4cd88c4e7904657ba933104db4bddda : SUCCESS in 20m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c4d097cb5154f0ea4cc6d4794cc697d : SUCCESS in 1h 05m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1e75970438594ad78a31b4e15d50503a : SUCCESS in 1h 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f78c18bf414d8d998e42344db5ea78 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c2c632ea52b46e9a009978943ac85f6 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad61dd08c66d47dfab797901f4042b02 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdeaebc248fd4ad9a5f631e0090a09c1 : SUCCESS in 6m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4950ea165d504fbc95fba2cadd6ebe60 : SUCCESS in 26m 38s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/aa469051eb204cb3a75b53fc82dedeb3 : SUCCESS in 1h 01m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1836b1982abb4a0e9a965afbf0d3ceed : SUCCESS in 23m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e380686995cf463bb2d68e29853b1077 : SUCCESS in 55m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4621ce14a6cb42d1bf7c2a2f773695f9 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b2d75ddd3c2729667717a08763627c9dd6af9f9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:26:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c3936c703f43a97bbbc986bd2713a606b8c2da45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:27:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d306f41c05c48cea664a45edf1aefc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f2e656c1af48559adb4e8d0090a42e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"446893d65da37c4c631a6ec27c24a80a6a3de13d","revisions":{"c0b567d0c4f48a45eb5af9b49e0eb26041d44872":{"kind":"REWORK","_number":1,"created":"2022-10-04 09:35:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/860225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/860225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/860225/1"}}},"commit":{"parents":[{"commit":"a9b0b63fedb5140ea6381c509c4dabde9772d5c2","subject":"Merge \"proxysql: Use almalinux for centos/rocky9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9b0b63fedb5140ea6381c509c4dabde9772d5c2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-04 09:34:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-04 09:34:24.000000000","tz":120},"subject":"debian: swtpm is in backports","message":"debian: swtpm is in backports\n\nWe need to mark swtpm/libtpms in APT preferences to make it installed on\nAArch64.\n\nChange-Id: Ibee94fe5a6217048eef01aa24ebed6ee05aae8de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0b567d0c4f48a45eb5af9b49e0eb26041d44872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0b567d0c4f48a45eb5af9b49e0eb26041d44872"}]},"branch":"refs/heads/master"},"446893d65da37c4c631a6ec27c24a80a6a3de13d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-04 10:07:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/860225/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/860225/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/860225/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/860225/2"}}},"commit":{"parents":[{"commit":"9e80733fb0a00012a344fef47c4b33ee47936f98","subject":"Merge \"Revert \"[CI] Disable Debian jobs\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e80733fb0a00012a344fef47c4b33ee47936f98"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-04 09:34:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-04 10:07:04.000000000","tz":0},"subject":"debian: swtpm is in backports","message":"debian: swtpm is in backports\n\nWe need to mark swtpm/libtpms in APT preferences to make it installed on\nAArch64.\n\nChange-Id: Ibee94fe5a6217048eef01aa24ebed6ee05aae8de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/446893d65da37c4c631a6ec27c24a80a6a3de13d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/446893d65da37c4c631a6ec27c24a80a6a3de13d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
