)]}'
{"id":"openstack%2Fkolla~860544","triplet_id":"openstack%2Fkolla~master~I093620679016b37e1664c9fe4cf7559433e744b7","project":"openstack/kolla","branch":"master","topic":"refactor/engine-adapter","attention_set":{},"removed_from_attention_set":{"29636":{"account":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"last_update":"2022-10-06 15:46:47.000000000","reason":"Change was submitted"},"34501":{"account":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"last_update":"2022-10-06 15:46:47.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-06 13:53:02.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-06 13:04:02.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I093620679016b37e1664c9fe4cf7559433e744b7","subject":"Move build related tasks to separate file","status":"MERGED","created":"2022-10-06 10:21:13.000000000","updated":"2022-10-06 16:08:42.000000000","submitted":"2022-10-06 15:46:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860544-refactor/engine-adapter","meta_rev_id":"f404b9a2f515f44f9427004f58740f432a0c8ebf","_number":860544,"virtual_id_number":860544,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 15:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-06 13:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-06 13:04:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-06 13:49:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 10:23:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-06 12:34:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-06 12:36:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"state":"REVIEWER"},{"updated":"2022-10-06 12:36:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"state":"REVIEWER"},{"updated":"2022-10-06 12:39:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-06 13:33:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"006d99c795ac62dfd516f7354f1546de73e89a02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 10:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31087bb11761ade35bbaa7f0065b4b1e6e8c2004","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-06 10:23:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b388ed1e66e32a6306ae9c390c7693b2d92c974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 11:14:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c870cb94963ccc2ff89795b71764bdda890dabcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:34:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d443779ca849e4b8ee25c134dd1cf9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a197bb1061dd48eea19d5726fee56d61 : SUCCESS in 25m 08s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/1092c61b529844e4ada2787ebc86ddcb : SUCCESS in 55m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d5069f4085af46e5855f33f4541e4712 : SUCCESS in 22m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2867e81092104290961490a6e89e955a : SUCCESS in 1h 13m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a7b31165d19648afb6c805b0fb17717e : SUCCESS in 59m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c787f26b06b94117848b3f090d143033 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/907b912ec7434bb6833b023f72c7918f : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d22d70e3c264671880c2b8bd1d728ce : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54053e3782da4d9fb57e946f6e26a6b4 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc05bcb0570b44f3bc4a99a277281a80 : SUCCESS in 5m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8b1e81bf55434b0c9cce0a7c60807810 : SUCCESS in 23m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4634d051153141f1b4a3d985ac79e912 : SUCCESS in 1h 04m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d9bed4338ee0449db078ea61ae0cfd80 : SUCCESS in 21m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be57fa44b6d04befb6f8c2bd7d7407ae : SUCCESS in 50m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7b8e33165e84e3888c02c3271c1f3b2 : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b27e58c35e6d29aa24975df90b2af23d2f7fbf40","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 12:39:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3430ac969f44459b6a0a0ed0e40c2b2365db9d03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:01:54.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"929760de6c5e2493d3aa092dc21abb54a67adb90","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:02:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36a9c3b6087c168e134d259b78a646738f962f2a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:04:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"524242ebc1d39b7aedcc13935886672bd82cf4d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-06 13:33:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdda4c69ea5124b725c02017330500b62832478f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:49:15.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4de530d00aad7c349cbefab46f299dd3fb6f043e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 13:53:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9570295fb4d739c065b6097fea6cd0a9bcf3ecb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:33:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da19cbbbe6c14c1b92ea8dfb90c53698\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c09889aa5d0e495cbffc66fca13e3006 : SUCCESS in 25m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7e58c05042864439acd9f74e08d077fe : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c073b169e7d24c91b7b56ea5667f3405 : SUCCESS in 22m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a5bc37bf50d449388428655f6ecf6553 : SUCCESS in 58m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/99aebd443f1d45b39c0dc526d8cbf82d : SUCCESS in 1h 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6cc688aeb3141599f8c4843904f6d55 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e524a579d76f4a57b134ef66f4bc5c29 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb3f2744c0134f118f079edf9ac15e32 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c18b5511cc1b4e5998642f5e417f2a04 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849bdb53f2f6418fb25e51358747ddf2 : SUCCESS in 9m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/df1fab62804347469befcd910133b878 : SUCCESS in 20m 55s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ee3097cb709f4598b79f451a6739d983 : SUCCESS in 1h 05m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0d6f63c571634e48ac115009f9f08661 : SUCCESS in 21m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2bbc79cc02a4e6f8b1d22bdf87bfb13 : SUCCESS in 47m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6dae44f5d29f4c6c819c96bfd31af738 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":3},{"id":"2036e754ae1b2223f03f0836a33abba640a49062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:33:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"22b0adad5b60edf1519d03ebb20910a397113ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:46:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ab8b31ca18422bb2da456dbaf8dca7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4e73bca097b44b84acb2768accb44a38 : SUCCESS in 24m 05s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2c5e1c7dae664e86a1eebab6eed4bada : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cad5bb6d3b7d4126b87c64742991293b : SUCCESS in 19m 11s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80f6eb39d594476b8878fff1c6291b8f : SUCCESS in 44m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/88df33d00f4947d1a7dd1b8ca242350b : SUCCESS in 1h 00m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db70b90f0744171bfd5889737ab310d : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d543e177d1e47ca8846ecddca3c037b : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde2db979ee04bcbb8708745c3d294e1 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2cb9ac7f2540389125118e026955b5 : SUCCESS in 5m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/982ce2838c564c819c449c469b09c927 : SUCCESS in 29m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f22f8b8847da49919a36f86ec32a8692 : SUCCESS in 1h 08m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e076a8f970a48648173802e5f74b5f4 : SUCCESS in 23m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f999de35d8274222ba687a99e97c32fb : SUCCESS in 46m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47947b1f860442fb809641d6108b3bc8 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":3},{"id":"afb6d12a39c2daf286741bef83694d1759d62fd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:46:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5bc48584392fd29ac92fd52c424e8d4fe506de8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:48:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20935bbae6004b58ab72234c9ddc20bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42180301631341af896dfbda30251bbf : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f404b9a2f515f44f9427004f58740f432a0c8ebf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:08:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe946a5aec045f195ad7819c4e5cced\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad9df107d8444eeda497a2a2d3fcead6 : SUCCESS in 1h 58m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/143c3931fad047838aa594defb00df1f : SUCCESS in 1h 51m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b500d97937f44b79b217a8d4e363f0e : TIMED_OUT in 3h 02m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3ffc20a0a0e43b4857b949c247975c2 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cdc064ea2334c048df6e05a518a4de9 : SUCCESS in 7m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4f897cb5c34a4908aa66c44687f9af1b : SUCCESS in 1h 59m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5a556d648fb145dc8fc42b859ef3eba4 : SUCCESS in 1h 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7","revisions":{"a7c6d2ef2f5d01ddd6903cd0fa38ad5c36e6f3b4":{"kind":"REWORK","_number":1,"created":"2022-10-06 10:21:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/44/860544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/860544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/860544/1"}}},"commit":{"parents":[{"commit":"1c9c23f8f7361e4f66d7ab5ea3b6a7fe7d424f03","subject":"Move status and log related things to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c9c23f8f7361e4f66d7ab5ea3b6a7fe7d424f03"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:28:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:20:50.000000000","tz":120},"subject":"Move build related tasks to separate file","message":"Move build related tasks to separate file\n\nAs part of kolla/image/ refactoring we move build related task classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I093620679016b37e1664c9fe4cf7559433e744b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c6d2ef2f5d01ddd6903cd0fa38ad5c36e6f3b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c6d2ef2f5d01ddd6903cd0fa38ad5c36e6f3b4"}]},"branch":"refs/heads/master"},"f70aad3b4132400c1cffabbf07edb2260f3e1fbd":{"kind":"REWORK","_number":2,"created":"2022-10-06 11:14:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/44/860544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/860544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/860544/2"}}},"commit":{"parents":[{"commit":"aee19bd1c4b45807d7d246b3d06710cca93f137a","subject":"Move status and log related things to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aee19bd1c4b45807d7d246b3d06710cca93f137a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:28:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:14:17.000000000","tz":120},"subject":"Move build related tasks to separate file","message":"Move build related tasks to separate file\n\nAs part of kolla/image/ refactoring we move build related task classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I093620679016b37e1664c9fe4cf7559433e744b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f70aad3b4132400c1cffabbf07edb2260f3e1fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f70aad3b4132400c1cffabbf07edb2260f3e1fbd"}]},"branch":"refs/heads/master"},"a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7":{"kind":"REWORK","_number":3,"created":"2022-10-06 13:01:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/44/860544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/860544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/860544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/860544/3"}}},"commit":{"parents":[{"commit":"aee19bd1c4b45807d7d246b3d06710cca93f137a","subject":"Move status and log related things to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aee19bd1c4b45807d7d246b3d06710cca93f137a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:28:37.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 13:01:33.000000000","tz":0},"subject":"Move build related tasks to separate file","message":"Move build related tasks to separate file\n\nAs part of kolla/image/ refactoring we move build related task classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I093620679016b37e1664c9fe4cf7559433e744b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
