)]}'
{"id":"openstack%2Fkolla~860545","triplet_id":"openstack%2Fkolla~master~I55c3f6b34631048b2463b02e6ea611d7e4c1a912","project":"openstack/kolla","branch":"master","topic":"refactor/engine-adapter","attention_set":{},"removed_from_attention_set":{"29636":{"account":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"last_update":"2022-10-06 16:03:24.000000000","reason":"Change was submitted"},"34501":{"account":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"last_update":"2022-10-06 16:03:24.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-06 16:03:24.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-06 16:03:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I55c3f6b34631048b2463b02e6ea611d7e4c1a912","subject":"Move KollaWorker stuff to separate file","status":"MERGED","created":"2022-10-06 10:21:13.000000000","updated":"2022-10-06 16:04:32.000000000","submitted":"2022-10-06 16:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860545-refactor/engine-adapter","meta_rev_id":"4bc0bd7e3f5e07d5050b04687fe113b9aba32c85","_number":860545,"virtual_id_number":860545,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 16:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-06 14:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-06 13:05:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-06 14:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 10:23:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-06 11:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-06 12:37:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":34501,"name":"Konstantin Yarovoy","email":"konstantin.yarovoy@tietoevry.com","username":"yarovkon"},"state":"REVIEWER"},{"updated":"2022-10-06 12:37:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":29636,"name":"Oliver Horecny","email":"oliver.horecny@tietoevry.com","username":"o.horecny2"},"state":"REVIEWER"},{"updated":"2022-10-06 12:47:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7698a936e13a8b325eb31b1b03a1cafec47162e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 10:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1db0e8eb52e6842418b5d2f7c462aec7fe75ccc6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-06 10:23:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1d1eca71705a4d89d4ffa492d2c7c1f2cccb0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:11:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cd34340942c452d97c0537e7e30b891\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9df4f311e34f4be49a5892335861a6f2 : FAILURE in 4m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/60c3cedba39141ca89c01d0b8c1069e7 : FAILURE in 14m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7197af472a04e24b62da46b85a4d82a : FAILURE in 3m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6408d7e79304af49803666a16e3e2ce : FAILURE in 20m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3117144f3d864f409c8102003b1f126a : FAILURE in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cff600987914f939903930f728aa4ec : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5c52af6407440578cb729f28d62767f : FAILURE in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66a095862e814fdb9ccc8f7342e655c7 : FAILURE in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69d96a361b51470e987c8d688d3e6949 : FAILURE in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce60d14c68ed44d8a3c4494f4af5e0b5 : SUCCESS in 7m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e991915361224bcf884066587e5fdcd6 : FAILURE in 6m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a282289e152f4ef29605032884d55948 : FAILURE in 17m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/55326b816aa54b1e9da6a329f18b318f : FAILURE in 3m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbcfb228c41843e0aa1e190fb7a4ddb3 : FAILURE in 13m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8fda544481d84065823be136d8d3c9f6 : FAILURE in 15m 08s\n\nWarning:\n  Comments left for invalid file kolla/image/tasks.py","accounts_in_message":[],"_revision_number":1},{"id":"01cf7a9949e9bdb454ca1516c5f59660acd7da09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 11:14:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"162c68380f2ab1e32340213a87dccd70b450555b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:31:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b31bbc3435d445c183c8dbbf08455b54\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/35bded4c8d4148b2808b88ad1fc8cd9b : SUCCESS in 24m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b301495af8c34389999af857aa7191d4 : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca7ca5645c5246edb522aaf4dd96eb17 : SUCCESS in 23m 10s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9484ed14bff4cb3872f343b595cffda : SUCCESS in 1h 00m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5f114fe2480d4897a896169f4d768b67 : SUCCESS in 1h 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6446d52316843f09e5a37591959b988 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bd7150c02d4b5e8bc3c788cbfc84f2 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1c6c23d71da406891a486aed5c88195 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3295650853f14db4b4df2a4f6adedcfb : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae24a156624c4f86b411eedc8caff56a : SUCCESS in 5m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ad20d6d9630477ba37f1c3d52f11c12 : SUCCESS in 19m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3ccf37327854578b900c0a9d652fb32 : SUCCESS in 1h 07m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5c696951d4684545bd7a763f558000d6 : SUCCESS in 23m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36e38b53869a4996a1797b3d4c851c7e : SUCCESS in 51m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c21ef64bc184fae89a9b81b7d1b041f : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"83744d907f730c777368a2f35ae0e6d82fa8b71f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 12:47:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f848087e587ea148e931a86b34371d0797363452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:03:45.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/860544 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ab252cbff257de9bf978fcf1bd3c9d2c5b76448d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:05:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d6be92506fc77dfb3a41d491ce1930095efbf84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:27:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18015c9bce1049a9a804f551e878f5b6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/46fdd623e9894707b649862498e4420c : SUCCESS in 29m 23s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7b76ce55d8d447939834c0ce3b720bce : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3fcad803d69e498ea90e44544a243328 : SUCCESS in 24m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8a9ee31c66948eeb81f61b79aa80180 : SUCCESS in 1h 04m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6638b94c43c94e9b951a439d01d3bd14 : SUCCESS in 1h 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba06ed9e953344269a53a9c8e3c5ce43 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb78a1a47984b3e9122824a6b70bdc0 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50efdb468aad4a469157c2e4abee6542 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48368a95287c4ada84eb16238e55074e : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bc4e0874f846138926cd3c5ee1872b : SUCCESS in 6m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b3b5649ab98145bc914fc0b64f1fe585 : SUCCESS in 20m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6e1c69c1979c479594e5e54cdd0f0f32 : SUCCESS in 1h 05m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/41868023d79a407c89ae2d8b94c6af43 : SUCCESS in 18m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86d10cb3ded94251aeaecc468521314c : SUCCESS in 55m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1157ef450df941998c9445d0919e5654 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":3},{"id":"a2b98829cfef83fb3766031f4949fe1d91db1f3e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-06 14:43:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d9af6a857d6f2d864970dea355162f52010df25d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:43:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"edaef87948c6d2c7e9bf5684deb6384d1fef2d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:33:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512b9e212a6d4e729f611802a3293062\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/12ce1245c53d45c6a2758dda62344732 : SUCCESS in 1h 50m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c9b59bdfff01486a86f91b75744c6864 : SUCCESS in 1h 54m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9a243e3d07a45ffabffc30dc840fe66 : SUCCESS in 2h 15m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/975831c0fa374ae89be2b83f3df6d62d : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e7afb14f9c146279f728ef9b3e87f43 : SUCCESS in 7m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d267f9341ae14d90b2202154ff872e32 : SUCCESS in 1h 51m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/24ba81f3c817448ab9654abf2759ed1f : SUCCESS in 1h 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"577418b5d393b252c7d876633d04d465fbb88b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:03:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b127aba9744b1a95b457392f2ed4ef\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20e611742fbe4435b199250d5bfe1c5d : SUCCESS in 23m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/649c1630b75c411a94aac48fecbd152b : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4972c8db9efd40878316195833817241 : SUCCESS in 22m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa022ac91bd84828871eb494331f7d16 : SUCCESS in 52m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d78931e6d6774ab7b4a60fcab41ba663 : SUCCESS in 1h 03m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b9e33a57674be28e7b5ba1b1cfebc1 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb6caab678da48e8a7c54454d2b5a163 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ee5bea6b3a94290bcb9530f4a2861e6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1274921d1bdb4f3bb5061ad3b4cb3ad3 : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07f2bc694ecb42cb9eaceb2070f5b3ce : SUCCESS in 25m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ebb20b01081b4fdf962a186430ea2d31 : SUCCESS in 1h 00m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e4e9f0aa8d947c1ba88e2d7dac14f33 : SUCCESS in 18m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3798c77b05cc45a89e39dcb3632cd0c7 : SUCCESS in 48m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3ad74752c0245dda8f7ae8e6c6b9cd7 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0d19291c836ed36652ff1c76ebcf52fd7282ac51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:03:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4bc0bd7e3f5e07d5050b04687fe113b9aba32c85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:04:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75e1c7a1d854dd796bf077aed66f80f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ade8f4313942a2b02dd3061f98acf1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"145d096bf7d93e200cd567f8177f09fd8492de61","revisions":{"caf1be72dff56987bc30d801892a6cfe3976d00a":{"kind":"REWORK","_number":1,"created":"2022-10-06 10:21:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/45/860545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/860545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/860545/1"}}},"commit":{"parents":[{"commit":"a7c6d2ef2f5d01ddd6903cd0fa38ad5c36e6f3b4","subject":"Move build related tasks to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c6d2ef2f5d01ddd6903cd0fa38ad5c36e6f3b4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:33:27.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:20:50.000000000","tz":120},"subject":"Move KollaWorker stuff to separate file","message":"Move KollaWorker stuff to separate file\n\nAs part of kolla/image/ refactoring we move KollaWorker related classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I55c3f6b34631048b2463b02e6ea611d7e4c1a912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caf1be72dff56987bc30d801892a6cfe3976d00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/caf1be72dff56987bc30d801892a6cfe3976d00a"}]},"branch":"refs/heads/master"},"2af88e979123085cda59e3cc0432c0d6308bc3b3":{"kind":"REWORK","_number":2,"created":"2022-10-06 11:14:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/45/860545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/860545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/860545/2"}}},"commit":{"parents":[{"commit":"f70aad3b4132400c1cffabbf07edb2260f3e1fbd","subject":"Move build related tasks to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f70aad3b4132400c1cffabbf07edb2260f3e1fbd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:33:27.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:14:17.000000000","tz":120},"subject":"Move KollaWorker stuff to separate file","message":"Move KollaWorker stuff to separate file\n\nAs part of kolla/image/ refactoring we move KollaWorker related classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I55c3f6b34631048b2463b02e6ea611d7e4c1a912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af88e979123085cda59e3cc0432c0d6308bc3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2af88e979123085cda59e3cc0432c0d6308bc3b3"}]},"branch":"refs/heads/master"},"145d096bf7d93e200cd567f8177f09fd8492de61":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-06 13:05:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/860545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/860545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/860545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/860545/3"}}},"commit":{"parents":[{"commit":"a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7","subject":"Move build related tasks to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8ee3244bfd4f3ee388308da6ab7e9da60d4a6b7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 09:33:27.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-06 13:05:59.000000000","tz":0},"subject":"Move KollaWorker stuff to separate file","message":"Move KollaWorker stuff to separate file\n\nAs part of kolla/image/ refactoring we move KollaWorker related classes\ninto own file.\n\nBased on patch by Konstantin Yarovoy \u003ckonstantin.yarovoy@tietoevry.com\u003e\n\nChange-Id: I55c3f6b34631048b2463b02e6ea611d7e4c1a912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/145d096bf7d93e200cd567f8177f09fd8492de61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/145d096bf7d93e200cd567f8177f09fd8492de61"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
