)]}'
{"id":"openstack%2Fkolla~860546","triplet_id":"openstack%2Fkolla~master~Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668","project":"openstack/kolla","branch":"master","topic":"refactor/engine-adapter","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-06 18:48:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-06 16:10:15.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668","subject":"drop shebang from Python files","status":"MERGED","created":"2022-10-06 10:26:14.000000000","updated":"2022-10-06 18:49:59.000000000","submitted":"2022-10-06 18:48:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860546-refactor/engine-adapter","meta_rev_id":"a02d4aeec19e0364bc149c22e803ff039242337b","_number":860546,"virtual_id_number":860546,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 18:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-06 16:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-06 16:10:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-06 16:10:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 10:30:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-06 10:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-06 14:43:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f9bbd5614d153cb8d89cdeb965567a96ad742178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 10:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85b74d0a58558c0afccf457cc63f5db9b0734596","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 10:30:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3da2c9f3dfbcb964271479906332bb54d943b4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:53:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6e8c0533e704029a511ef8ed547875b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0a61bb88e95340fbba453fdecde3e4ed : FAILURE in 6m 52s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a2cd8f292b724e4391039e4627332b33 : FAILURE in 18m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8999d9fb2a3449c4bdf3b72a9c99fdeb : FAILURE in 7m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/557a6c946ae44912b212ca2b05e37511 : FAILURE in 15m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c962e3da87934192b0ea95858650b89d : FAILURE in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32149ffc54dc41169e908d066f018b92 : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eb7c27c33b145b9a898975abfe80b74 : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80701a460fa433395b8befc4536330a : FAILURE in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe2336c23f7b479faf5c35446f6b2de0 : FAILURE in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3552ae7f2fff4b58864caea320535ee6 : SUCCESS in 6m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9523b9453d464c469eef261576972519 : FAILURE in 3m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bf5eeff71c2141978391d9fa089f9c25 : FAILURE in 12m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b9676e811d164e4289dcca8e2c0b8abf : FAILURE in 2m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/81aaf41a203b46faa13f432f8e39c154 : FAILURE in 9m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f911b5c13ac5485b9c0e018160e09183 : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d8d6c9ed6df94f83719920253e12f727af1402a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 11:14:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"db64e4eb5e9165f796f34c0f0ade1c2a50023160","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 12:08:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ac54cde27346b216edc53e11ed9e4fcb1729338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:31:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e42364789c0b458a84eb38126ae67179\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/db62d0c30141423b81bc89361b244f97 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a1c88f68598346ccb8968344ef234053 : SUCCESS in 58m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b7acfebb3fb141988c499298446da640 : SUCCESS in 19m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0a110a00be64dfaa90494c90a4ae32b : SUCCESS in 59m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d1524a8ee9724587a5c8f42c60c64e20 : SUCCESS in 1h 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b53fca89c444b4b60c011c3ef04dca : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a282813de304b8c8e9af117330b0f48 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742a94555f6546eeb11fb83ed5ad62fd : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47562d3ed8014d448aca32e304e0da11 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac187cc6a2e14793a0ad8199614cdbe4 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/063a7cca8a2942cc8b8bba1db381f2b3 : SUCCESS in 25m 31s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/afc5597ce62a4c369bb349f293298206 : SUCCESS in 1h 01m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ac85c8fd1cb346bebdd253c63a5510f0 : SUCCESS in 19m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6d471b03d7d4e80bb646dcfef1c9d36 : SUCCESS in 51m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a744f1a5e9704ffd8acd846abb2cbedd : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b7e1c81b9e6a0060e3af6f24c69eb84c8168e84a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:03:46.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/828716, https://review.opendev.org/860545 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"ff23fe94c8bf49c13282dd2d8d361b80935d248e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:07:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6a3c0245fa8a301d500b724cf4858a3a5763b4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:27:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6841d0ac8e194600a7c4df1755a678e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bc89fab9f5164f60ac342148b16b6cd9 : SUCCESS in 23m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f68f630a4aaf40758f7a6ed0b9115ae4 : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bb5bb4f04e314e38857e8455abe44e40 : SUCCESS in 16m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/06aad617a9254207a8d745bd9ef35907 : SUCCESS in 50m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c4c91ef3105a4f068713a6a328605da6 : SUCCESS in 55m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2857f7f352b745348855f35aa8887d98 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3895fef3944458aabdd0c5a20e6aead : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d39cc7084f1949fcb8ab1c4c7cca40e6 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa593e06bf214c49bebdf0391aa1674f : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6408835ef4d941c382bafebd38e94b48 : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/766a8f3ad9cc4091a938fb40e81db7dc : SUCCESS in 22m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/63ccfed1c27b46aa8801be00b53ac4da : SUCCESS in 1h 03m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c7583f1cb7c5444ebdff7d23d1d594d9 : SUCCESS in 20m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c27c98a4dd9648ea8dace5d4d60681c7 : SUCCESS in 52m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ae8e56da74d4dccb6e353ade33fcf37 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":3},{"id":"68d25ca45a35e18ea515333ec991943a93c6c78c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-06 14:43:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3642f17befed33a6dc221a47eb057abca875ca37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:52:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e20592f3e8834486b53a99b9666eb2eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/df52561b7be549eda479eb63b95c20bc : SUCCESS in 1h 24m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/417a3ded649c44a69dce3e5754114314 : SUCCESS in 33m 16s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c3e7edfc3a3403681cd45e1124b22b2 : SUCCESS in 2h 03m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa09e0038d0c4387a352819939e523ea : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/518e147c38654323b919ce1e89ab31ea : SUCCESS in 7m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60cfd710448e4b8f8aba8ba9eed40424 : SUCCESS in 1h 16m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/284db350f7a345f8a075327185c2e4cc : SUCCESS in 1h 16m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65884f7fbf7e39fe2ba1057e849bb944beba065e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 16:06:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6947ed46e6ec7a64562e38c15cac92e78b63beaf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 16:10:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"09f562cb053cb33292574785f438d1a90999d183","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 16:36:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"22c4d8f90bc4388a3237b5005c4552ddbd85f89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 17:28:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c341405d697447b9b5e5243a7a98f83\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/39100be138b9401ba6d4d5d829e9b381 : SUCCESS in 29m 15s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/cd3538a5485b423889b09dbf95e2d247 : SUCCESS in 1h 13m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d2093659ac24e9fb2e3072839704e8f : SUCCESS in 26m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2ff814a0bf244e98beb4e5504ec8e4c9 : SUCCESS in 1h 03m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/eecfbc61909142cf829f80bbe597eaab : SUCCESS in 1h 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a78d99282e4054b733462167ec9fab : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c6cf7411abc427385fc67f05be52769 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9079466a54374402b0ab99ecf5555baa : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/122f66832b10467d86ea45f044803312 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33591ab225f4be5bbb1d0b724116de3 : SUCCESS in 9m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bc7f154043c947aca701cec338b52619 : SUCCESS in 31m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/798183008a5446469bb9d75b5d85acd4 : SUCCESS in 1h 14m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc0f631b9fa0405790ef45b144e24bd4 : SUCCESS in 19m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/db88e74ba5b14dae84a8862519bfc902 : SUCCESS in 54m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0bd9240ff9304358ab1c9ef104ef702d : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":4},{"id":"73d015ba04756e7abb90cee7ba3f943254177801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 17:30:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6d7a180bd9858b45eb66241cd9083b30c54371a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 17:53:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc3a904deb5d4f2ca642d49a07f68ccd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4614ec0c5a734aea9e5190b4e4c4cfcc : SUCCESS in 1h 34m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a7d2b2ad23e485fb90d7121b30c2bd7 : SUCCESS in 52m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec2bbc59f9204bef824000dceb7e6cd5 : SUCCESS in 1h 44m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eed4b6182f944ddbc25ed986768cb69 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4551128b2fc24f379f8512eca684b7ad : SUCCESS in 4m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b899ba17fcb24b7d980acb5a63568ba3 : SUCCESS in 1h 19m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/592edcfc0d674040910cded0341de0cc : SUCCESS in 1h 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0a758cd8e7129023d078068683bb7bd877f68664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:48:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ba07fa9b5e41268b4107b8c1a02a98\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/89db18580cf244539e8604b73895a7aa : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/374d493ca9be4707b92e4cda4eb3b487 : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e2c42887f684117a0cae0869a306cae : SUCCESS in 25m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4e9ab026b3914eb5895d90e46428ddb6 : SUCCESS in 47m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8c5241d1b322477fb0ea1c1562b7e86b : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9547da952149d585e36172dd13001b : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1907ff52e9b4ea18c075d5249877c3b : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7612cb97733d459abc676e297dbaaaa9 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153ef1db9dee4ad4b45d03e2d0cce3ae : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/631d0c594503499ba3ff56d4cc63f83b : SUCCESS in 19m 34s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bf0c4131f2f3457da3b1b16154f741e4 : SUCCESS in 1h 00m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a689b18ab8d84da08cbfdd40cc3c620b : SUCCESS in 19m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a00b96d6fa804fb4a4718049c411fc70 : SUCCESS in 43m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed0d0dca434d473cbc0664ac0b79208c : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":4},{"id":"3bfda3459d87fe225c4f6700fa33138a51ef253b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:48:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a02d4aeec19e0364bc149c22e803ff039242337b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 18:49:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d175c8ff359f4367aac103898bbae421\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b77e8a2f2dd49f79430517148004596 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cd08d9b54383252b90d9184c8ce63ed8c66c2418","revisions":{"939aeacdcb622319247fad7585f2a52959199622":{"kind":"REWORK","_number":1,"created":"2022-10-06 10:26:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/860546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/860546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/860546/1"}}},"commit":{"parents":[{"commit":"9a3a1b90e7287c6da5049cc3783fe5e6a61f8e34","subject":"Add a way to add other container engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3a1b90e7287c6da5049cc3783fe5e6a61f8e34"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:25:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:25:01.000000000","tz":120},"subject":"drop shebang from Python files","message":"drop shebang from Python files\n\nOnly kolla/cmd/build.py is an executable script used by user (either as\n\u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly).\n\nChange-Id: Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/939aeacdcb622319247fad7585f2a52959199622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/939aeacdcb622319247fad7585f2a52959199622"}]},"branch":"refs/heads/master"},"b1dd86ce882b09e7199a27cb1df084f5af49a4a0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-06 11:14:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/860546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/860546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/860546/2"}}},"commit":{"parents":[{"commit":"8678d80f019856862769b68f049adf8cde8202a9","subject":"Add a way to add other container engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8678d80f019856862769b68f049adf8cde8202a9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:25:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:14:17.000000000","tz":120},"subject":"drop shebang from Python files","message":"drop shebang from Python files\n\nOnly kolla/cmd/build.py is an executable script used by user (either as\n\u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly).\n\nChange-Id: Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1dd86ce882b09e7199a27cb1df084f5af49a4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1dd86ce882b09e7199a27cb1df084f5af49a4a0"}]},"branch":"refs/heads/master"},"5d1eac058a33247e2ff7482ccdf96bb39aed580a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-06 13:07:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/860546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/860546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/860546/3"}}},"commit":{"parents":[{"commit":"e2ee4c3ff4ed8caf207395b16114775e6211a960","subject":"Add a way to add other container engines","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2ee4c3ff4ed8caf207395b16114775e6211a960"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:25:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-06 13:07:01.000000000","tz":0},"subject":"drop shebang from Python files","message":"drop shebang from Python files\n\nOnly kolla/cmd/build.py is an executable script used by user (either as\n\u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly).\n\nChange-Id: Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d1eac058a33247e2ff7482ccdf96bb39aed580a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d1eac058a33247e2ff7482ccdf96bb39aed580a"}]},"branch":"refs/heads/master","description":"Rebase"},"cd08d9b54383252b90d9184c8ce63ed8c66c2418":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-06 16:06:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/860546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/860546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/860546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/860546/4"}}},"commit":{"parents":[{"commit":"145d096bf7d93e200cd567f8177f09fd8492de61","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/145d096bf7d93e200cd567f8177f09fd8492de61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 10:25:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 16:06:16.000000000","tz":0},"subject":"drop shebang from Python files","message":"drop shebang from Python files\n\nOnly kolla/cmd/build.py is an executable script used by user (either as\n\u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly).\n\nChange-Id: Ia1ba2699f425749f9d4f2cdba27caeb97bc3d668\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd08d9b54383252b90d9184c8ce63ed8c66c2418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd08d9b54383252b90d9184c8ce63ed8c66c2418"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
