)]}'
{"id":"openstack%2Fkolla~860547","triplet_id":"openstack%2Fkolla~master~Iac6ea537491e5c00ce6fb4b4e8f07a9124e9fa8b","project":"openstack/kolla","branch":"master","topic":"refactor/engine-adapter","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-07 05:53:37.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-06 16:30:16.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"Iac6ea537491e5c00ce6fb4b4e8f07a9124e9fa8b","subject":"do not play with PROJECT_ROOT in build.py","status":"MERGED","created":"2022-10-06 11:14:31.000000000","updated":"2022-10-07 05:54:36.000000000","submitted":"2022-10-07 05:53:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860547-refactor/engine-adapter","meta_rev_id":"54153341e25963cb24a3f76f2e9bd750b0631d6c","_number":860547,"virtual_id_number":860547,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-07 05:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-07 04:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-06 16:30:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-07 04:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 12:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-06 12:45:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-06 14:43:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-07 04:30:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5f698badca44700a6518cb90b7ae1d0cba1c6219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 11:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c36bf64080358226241bf8866a31f03e1e16374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ad0b8318cf41339ec8199ab1472eba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0618153863e046538a4a84681043a9b0 : SUCCESS in 21m 06s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2fb81113e18d482e8ade028e7e020954 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dab235c2901c44d98b32eb20b63aeb27 : SUCCESS in 24m 11s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61b4c3f4bd3c496580beb7df419583e6 : SUCCESS in 1h 10m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0d8410dc22f341429c9a7405b2ddadc1 : SUCCESS in 1h 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dff0fbe627c4c21b032eba735b6430e : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42714615f7fd403aa9c2c3f94de354c9 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84d45776e926464eab9b725b5ff0c858 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/997b2958359c436e8d6e0db89e336855 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28f2d57b747422fb4cfd3d672fabc1c : SUCCESS in 5m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e70a90ea833545eea74307a8a287e8cd : SUCCESS in 21m 50s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0272a305b97f418e87f977ec3732f45d : SUCCESS in 1h 05m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ebc8abf7dbd45368a900a261dff7908 : SUCCESS in 23m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c736648a6f1452d96f718a470d59dad : SUCCESS in 52m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/070e0b86b54e45e9aa188cc65ad4e6ad : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"347caf7e8bd8235393244397a8699cb981e27aeb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 12:45:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf8143a1bccea379e6acd3ddeba835511c27a1d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:03:47.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/860546, https://review.opendev.org/828716, https://review.opendev.org/860545 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"6f7b1a3eaf6c7fa4fe2e494e4d918f37556d2772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 13:07:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"92744e42e14a821e17c1456a85fb362af4577252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 14:24:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4f732f445048c3b6dd8342a7d63aba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1d1fd88364f4430480c29f8f857b958e : SUCCESS in 26m 09s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6b29e4a77b954ac1ab10407658b944e4 : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77cdfd3aea0a4743a02d3c4a1a47816a : SUCCESS in 23m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/941029126305438dbd7720778cdc3697 : SUCCESS in 1h 00m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/94ac6ce6c87543208c4d5497c40629a9 : SUCCESS in 1h 08m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9694b170b6564afe95383f12bea863e6 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae25382bc8a04d7ca60fa29822b0b0c6 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c53aa9e5cc1470297d48e37872ba9c0 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6f8164fee834731a9568fea12abb0c1 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2d278adc854395964c518a78050808 : SUCCESS in 4m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5fc3ef056fc549fb843acb544c17c6b0 : SUCCESS in 22m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3aa48b079ca746658f6ab541c521fddf : SUCCESS in 1h 00m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d2f6584c59545c784b12b4a82056962 : SUCCESS in 22m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fbaaa817c864cc3a81b85dae8f89dab : SUCCESS in 46m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af10c9cfcdf64c4db561faa8a0e36c58 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f62a96ed7565be0f34a16409a1537fe2e2447265","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-06 14:43:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"048e4945b0091b471845d92128a6a40d67976c36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:24:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49d3390d1fe4e09a1e6d7b9745f0897\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e824f16c7e5541fc8fd280fd6740fce8 : SUCCESS in 1h 33m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b43849fab63b4da8aee7eeefb588fc1e : SUCCESS in 1h 15m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66e2d6c462d74b068e5a80db0354a1a4 : SUCCESS in 2h 23m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfef4b8a42ea489fb94b59cce2dc22da : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f646262eae614166b91d71085b4b5d8d : SUCCESS in 7m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf7853659af64831a1b79df4f4f022a7 : SUCCESS in 1h 30m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/52c3f7c16e06422d9a04a698a597c818 : SUCCESS in 1h 08m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e53d6461e7f4fb3cdd2570018e92f5e8b32a659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-06 16:27:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"77145ab6378428b5bc917e9254b841aad74ce769","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-06 16:30:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7bf09bd125cef4b6f50e32d7749f064d97d949fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 17:51:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae88294b156418792c02158262de435\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3b226ba06c5944728028ade81f100ea3 : SUCCESS in 25m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/75af2771d9324f649684f9b71e810225 : POST_FAILURE in 21m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca14389100014f929a238833c7216177 : SUCCESS in 22m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac3cec642daf42cc85a46f1ca59f9e2c : SUCCESS in 1h 10m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/84a71e2e5aa34807b434a6cc031c7aee : SUCCESS in 1h 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a40c219ce8964ec99534d0245bb8a298 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/002ea9f583aa41afa16149826b217b31 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3722039daf7747fe9c196117f5fede5e : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8d396a26a6240cd833ea940a25a038e : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/813a7eab9cb2477ebf7496b27b5fcecb : SUCCESS in 6m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2222fb7689c742e09ed3b61042006c56 : SUCCESS in 27m 41s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f3bc49846d014a86af62df64c3e2ce4f : SUCCESS in 1h 16m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/683ca993cb3b45f59e015de1cad1daca : SUCCESS in 36m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b468700084b4227b7777822effa5795 : SUCCESS in 54m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f579fe14683a4662b32d580f6f3846c8 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":3},{"id":"fd2fd1e816a7618ad1833af25f2a6d54097b0fae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 20:21:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb672a1040c4bd092b909d884e4a2b5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/35e9ffc91dbd4c86907a31882e0e6344 : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e24406aac6724dd5aa99a3137c8487d6 : SUCCESS in 1h 26m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f5d54bad2184ae39e2d1909f7ce940b : TIMED_OUT in 3h 30m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4d93a1704a84addbfe3db66acca70de : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc02bc35f5444264bd0dc756f18e2050 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ecee441d45b94d328f748c54a2a8e497 : SUCCESS in 1h 26m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5866fb292d484d7a85bf1023f9e5e576 : SUCCESS in 1h 09m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"da33c5a738bcec3ee420e38f87639b1b3f44fa93","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-07 04:30:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"00eaa98902338ac235f77d885e68bacd533f5d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 04:30:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3ad958aa70435e77cf23739bb6af64e5676327b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 05:53:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cde4eb0326b42b99b7a6c134d9d9193\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/265bbc3a839d46e8942eb2b4335204dd : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/bb8f283cbe0a4168981f78ab27ec5a9a : SUCCESS in 1h 10m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9fb330b7eb64ba887e81527dc0bb7b4 : SUCCESS in 18m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/72c44e1b4df344dcbee6558e20824bc6 : SUCCESS in 1h 01m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5fe06b3c629c4bad9ab8c163105378ec : SUCCESS in 1h 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c6e0d2be74642c2a6dbdd23e3674d39 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b40ad7ef0c047958b0f777584e203ba : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a53bc1e04f44d996acfe3ad095bc8a : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13a967fa2c3447cbe6f747d2a0c2d63 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/51e6d69ff85c4923aa32acbe32f04272 : SUCCESS in 29m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/d5891ecc5c194f63ad24a981f9ecf030 : SUCCESS in 1h 12m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a2348eb328484c8c82669c06359b2cdb : SUCCESS in 20m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36f1c86a83d547dea4a1aa534546bce3 : SUCCESS in 53m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00306c5ec9cb4b4dbb3db9e15dac0b4b : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":3},{"id":"96cbd1b47e4396070635308450198248f7ef2ef9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 05:53:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"54153341e25963cb24a3f76f2e9bd750b0631d6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 05:54:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb596a0de5740208d17faddf5155295\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15abb3be84674974bfa836eb3e49bff3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e2c061ee63a701284bc2f6b1470d9d33a58e724c","revisions":{"4c221ab8aa8ba44ec91d0bd6dd2e5ceeafefe3c6":{"kind":"REWORK","_number":1,"created":"2022-10-06 11:14:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/860547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/860547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/860547/1"}}},"commit":{"parents":[{"commit":"b1dd86ce882b09e7199a27cb1df084f5af49a4a0","subject":"drop shebang from Python files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1dd86ce882b09e7199a27cb1df084f5af49a4a0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:10:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:14:17.000000000","tz":120},"subject":"do not play with PROJECT_ROOT in build.py","message":"do not play with PROJECT_ROOT in build.py\n\nPROJECT_ROOT mangling is already done in kolla/cmd/build.py (which is\ncalled as \u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly) so it is not\nneeded here.\n\nNow we can sort imports and drop not needed ones.\n\nChange-Id: Iac6ea537491e5c00ce6fb4b4e8f07a9124e9fa8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c221ab8aa8ba44ec91d0bd6dd2e5ceeafefe3c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c221ab8aa8ba44ec91d0bd6dd2e5ceeafefe3c6"}]},"branch":"refs/heads/master"},"bb7873d6d5e00dbd204ccde639cf0b9b02c0c2bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-06 13:07:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/860547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/860547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/860547/2"}}},"commit":{"parents":[{"commit":"5d1eac058a33247e2ff7482ccdf96bb39aed580a","subject":"drop shebang from Python files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d1eac058a33247e2ff7482ccdf96bb39aed580a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:10:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-06 13:07:05.000000000","tz":0},"subject":"do not play with PROJECT_ROOT in build.py","message":"do not play with PROJECT_ROOT in build.py\n\nPROJECT_ROOT mangling is already done in kolla/cmd/build.py (which is\ncalled as \u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly) so it is not\nneeded here.\n\nNow we can sort imports and drop not needed ones.\n\nChange-Id: Iac6ea537491e5c00ce6fb4b4e8f07a9124e9fa8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb7873d6d5e00dbd204ccde639cf0b9b02c0c2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb7873d6d5e00dbd204ccde639cf0b9b02c0c2bb"}]},"branch":"refs/heads/master","description":"Rebase"},"e2c061ee63a701284bc2f6b1470d9d33a58e724c":{"kind":"REWORK","_number":3,"created":"2022-10-06 16:27:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/860547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/860547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/860547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/860547/3"}}},"commit":{"parents":[{"commit":"145d096bf7d93e200cd567f8177f09fd8492de61","subject":"Move KollaWorker stuff to separate file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/145d096bf7d93e200cd567f8177f09fd8492de61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 11:10:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-06 16:21:16.000000000","tz":120},"subject":"do not play with PROJECT_ROOT in build.py","message":"do not play with PROJECT_ROOT in build.py\n\nPROJECT_ROOT mangling is already done in kolla/cmd/build.py (which is\ncalled as \u0027kolla-build\u0027 or \u0027tools/build.py\u0027 or directly) so it is not\nneeded here.\n\nNow we can sort imports and drop not needed ones.\n\nChange-Id: Iac6ea537491e5c00ce6fb4b4e8f07a9124e9fa8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c061ee63a701284bc2f6b1470d9d33a58e724c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c061ee63a701284bc2f6b1470d9d33a58e724c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
