)]}'
{"id":"openstack%2Fkolla~860758","triplet_id":"openstack%2Fkolla~master~I3651576ee354364d639c187ff750491667ecab56","project":"openstack/kolla","branch":"master","topic":"swift-sudo-issue","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-09 18:12:16.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-09 19:37:24.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-10-09 19:37:24.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-09 19:37:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3651576ee354364d639c187ff750491667ecab56","subject":"Fix Swift deployment issue","status":"MERGED","created":"2022-10-07 22:44:12.000000000","updated":"2022-10-09 19:38:28.000000000","submitted":"2022-10-09 19:37:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860758-swift-sudo-issue","meta_rev_id":"6a0964d61377545d8ce7610415ec93ab3dac8765","_number":860758,"virtual_id_number":860758,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-09 19:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-09 18:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-10-09 13:44:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-09 18:12:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-10-09 13:44:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-10-09 13:44:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 22:51:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-07 23:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-08 11:33:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-08 11:33:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"66d2c8aaa118c49a6b9204c4e42e1ce9937876fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-07 22:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d152c8007b2106d0048a1ac6980d771bd925a727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 23:58:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8e2eddf4ffd4fde91c4539a86b6b4dc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7654bbf7a53c4ca1996e14588528cb9d : SUCCESS in 20m 00s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b777e5031a2e4cb1a255e02346c3436e : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/65b0419c68d643d69e1918b9798927ca : SUCCESS in 15m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2fd2a58a2e654845b9b26df38555d480 : SUCCESS in 1h 00m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/91b4f95923ff477abc1d053c1c5c92a4 : SUCCESS in 57m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd990c0cbab4b2ea5669bdd2e2ed56f : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3138fe176f4c139cfa85b380fa3ad1 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/951fb3964830416ca3c5e4b52e0b7f30 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e3b32b548e642f29fafa35bd9b451db : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e872e5026234fdb8591f95e2f81137a : SUCCESS in 5m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/12188518e186453f9c2843e56ffe956b : SUCCESS in 29m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/493c0213016042969db3e6090eaa7164 : SUCCESS in 54m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4d07c1dd530640c4a1d1f2801a526eef : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9df0fd28c7fe4ddda1968604dea15591 : SUCCESS in 36m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7af8b7acef084eef818749fcdc3ae1a1 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c05d624320f6307a1b2fc55bcd26ab3ab2b1bda8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 04:51:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7125b63af7c649d8982c4d3f860117a6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b8a02c12832d409c966f011273c8ddcf : SUCCESS in 28m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5625d5f2b436412ea88c5a93343a5d8e : SUCCESS in 27m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/195a1b66243b44f3ae54cd4e92c5c660 : SUCCESS in 1h 14m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e9d0fee4bfe440a97ee40f9d3be77e4 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/230d311950914c4d994143ca638dd8b2 : SUCCESS in 3m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6b6488faa4844f27a161811a8a7649f1 : SUCCESS in 33m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/95c6020b1a1a464b84913701294eb9c6 : TIMED_OUT in 6h 00m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7dece4be7f6d0fbf3663529fdc17cdea9fb24311","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-08 11:33:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92dec476f78fbf56d9b08245dab909da0207626c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-08 11:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e70d588bd457461eebfc2ce65fc0caa4d6798cd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-08 12:01:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8ccb8bd50578e23ab286937b545e5f7209a650e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 13:24:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47cc74b8f2b4021822f019a34564c50\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20c4607b82fc4fa6883304fbd7156648 : SUCCESS in 23m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d8ee9424f3c0485c8c8b18fb1930e0d3 : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e714337eb06d4da7ad82b672ad5e8232 : SUCCESS in 29m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a0f2f8955e54278883c3f9db4301b6b : SUCCESS in 1h 05m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7608da0da51e483d83afdbcf323721a3 : SUCCESS in 1h 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e951dc65ae401aa4c6f5326f8ddc65 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68755775cf6e4de5bed677eb39509e05 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db67facc1e1646bb8c7e59d5283d2cf0 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8240574bee784235949b516f1fdfd452 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d883feb4673d44c886a1ef038b1089f0 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5577769a691a44aa917945fafde5aa7b : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bfc98eebb9854479bb8d30c43038b1c5 : SUCCESS in 27m 53s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/fabaae291f77490dba6e39807fa68508 : SUCCESS in 1h 04m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70ff6b90926044a9afc9a31348004c87 : SUCCESS in 28m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/451baa47b7a544b6867666ee5ce79639 : SUCCESS in 47m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9cc99255abb24c59bbfba5f43cce1cbf : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"132626d8f9e632c3cd9fb07f1873770d93115d21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 18:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afc0f995de24c429263c69a3eba6c4a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6f154d420544464d9f636062f9010532 : SUCCESS in 30m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c706070cf0c14a808def301fa41c644e : SUCCESS in 30m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af1f969993b24245b58e81e2096653a3 : SUCCESS in 2h 02m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a178e356fc384f79a0f2dc2a3b1eac22 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f8cfc3e80f04af5a03c8d82bb93f2c5 : SUCCESS in 4m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0eb7797d5eab4807b100f0675c7a66a0 : TIMED_OUT in 6h 01m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/291a3b051f54409e92638370b24f16ec : SUCCESS in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1206dc8bbf078e99c560745ab5f1fc2d3764cf1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-08 20:40:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1df5517efd00250bb600ba66b8fa703476ab4cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 22:01:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b19874805e4346a5e052d0ccec9413\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e48f2e447ed74c80bda64869ec19cf3e : SUCCESS in 22m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6f94b00700344859bc0f0498bae67288 : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c7e956a8cd9246958cadb8a1f6f030c5 : SUCCESS in 22m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/781c0a2fd65e4717b903073b9abb2047 : SUCCESS in 1h 02m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dd776a139eab44c3aa8b03615cec826d : SUCCESS in 1h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda38af39a344c87b91d8e6da0973645 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e901c1ba865c4cb4b32b37f1e5af5dc6 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c39a542da29d47a8a53f4df2f2203951 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e89fb4e3419d4cdea2463cef823f9350 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2d6bc3f1b44cddaeda2526f57b9995 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db75c252e3c5428f96da45eb95fcb817 : SUCCESS in 3m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ff0835aab5654dd4b95487eda63d0273 : SUCCESS in 23m 12s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/4e30c5a9176b4b18be08d0ca440bb8c5 : SUCCESS in 1h 03m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/672f4d586afd49e2a27416cd453c2234 : SUCCESS in 23m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b19d2d60dd144d0a90d4b7dc63ea2109 : SUCCESS in 49m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8e115a3d3d34555bc113ee538436b5f : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e8dbc87c4b8b1ed16fb0ba98fafbb6a2d2383da4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-08 22:25:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a8683a3653489191b3cc0636f371a1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fe8f6c4f515745c79a03b2517cb72f98 : SUCCESS in 56m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4200c7ac1f2e42a19327a807466bd154 : SUCCESS in 1h 03m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdf320ee64564162892ff2e2566d683e : SUCCESS in 1h 42m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ff237042b2b4b3494b7f386add93423 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfdb880e0e98493183220e17c6d2357d : SUCCESS in 5m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9026d9ab891d4a82bb8340f1a8c62a78 : SUCCESS in 55m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ab2ba097ea0d45dba8f59c4d9f333f8e : SUCCESS in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5d719d4e66e418e9734027d2e85ea72eb8fb94b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-09 13:43:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"cd4d62e0c93db68e7a6a4411274ffff102bd958a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-09 13:44:32.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2623b19726536abce82aa384660f0d22735e7297","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 15:15:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d360ffb47604921a931e0c3aedfc6c3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/039158d9b3ca46efb48a8399c6d3bd70 : SUCCESS in 36m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/031d84bc2e5b411892a006bcd0903444 : SUCCESS in 34m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/978e05b1d78046be913af458076479d0 : SUCCESS in 1h 20m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8161e848e9534f02b552344c2306f144 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a2f4032a691409d8e446c9ef3eb13c0 : SUCCESS in 5m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf65b1b52cf548588232f2a824910d23 : SUCCESS in 33m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9b598f7a6682409aa22af9d71fee9452 : SUCCESS in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6066947818ff9cadbddc10e836b9aa1fcab5467c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 15:48:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24428d30357946429d0ef86859b9d46a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e606fdc3f2a4fc8b3d5cb38721f596f : SUCCESS in 20m 18s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/71dcc0713bd940e99b7ab608913b495a : SUCCESS in 51m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89d5bf38672b421c9856426e1838f46a : SUCCESS in 16m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c5fe980b35b4dfda01f2e7bf7759378 : SUCCESS in 44m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/83c7902b54be4869a45b23e2689fe19b : SUCCESS in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277d1e93bb7d473d89ff0fadd8ddc29b : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d849413d9784b378010115f900a66b7 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5dfad733d3c49d8ac2553cc85d043cc : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b240e45e90c478fbca2d3740ffcaa3f : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a866464a724c3a807a835f4eae4092 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdb4b0c58b9c4fb9a249351d17b0bc39 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5302e98e541b48a1bd5f7cfaa274befc : SUCCESS in 20m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9ba67a1a531d4a34ba71e3a08777ae60 : SUCCESS in 1h 29m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc4b95432b904f2492079e28de40cf44 : SUCCESS in 14m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf2574f3811040169bf37046b1e70756 : SUCCESS in 37m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d20e9e6e696d4fc09baa8d5cd5e4a35f : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":4},{"id":"deafec4c232971dcd948f6f68a77d3e693a5d96a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 18:12:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66244e0d6b330f313b5ef759925a7be3f9b76aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 18:12:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ced019cea2f05a893648fc00066479871a2c929e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 19:37:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06601d21cb9b4542b071c1d0873cdaba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c8416916139844a79025c0af2c59dd0d : SUCCESS in 25m 56s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/1e256137c3d24c729522f132d00f371f : SUCCESS in 1h 09m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/96588e39beb245cebdd7ef4d63ae8a21 : SUCCESS in 23m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50cc981e0bf345b2813f50b6fc03605c : SUCCESS in 1h 02m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4a319da8e76249ceac05547a60a74eb0 : SUCCESS in 1h 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39403b0f30b349afa407220f73ac39a5 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db20b0364483461fa054f58f33f0e353 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3559e6a063b64a49a984001f892b3708 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6520fad84a9a46eab135f0e6819bfca3 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4bb2c5e60df40f3a89a02514478c8ed : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/18f84b587752415980bcd5cf017950bb : SUCCESS in 31m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/8c2bd041fa11480294bf9145ca1d73c5 : SUCCESS in 1h 14m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3159f6ba69f04cdd8cbc1f97f9cec870 : SUCCESS in 19m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e764e3e925e1467cb37e78c2aa38da4c : SUCCESS in 49m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c054e1d1e8c4137a2b277c8ee9ea250 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":4},{"id":"441dbf1fe9483a38795a0fdf4abcb4695fe4595c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 19:37:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6a0964d61377545d8ce7610415ec93ab3dac8765","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 19:38:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43187723b35f437ba09bf95425ad535c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ab00f83130459a89e141d90502e300 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bdf48c7f484a8e9ac7f9290f357cfc : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b8a352647d57d35463e551a0d360c19fb2ad34c6","revisions":{"673f276f6b9404c3a286011e92cb86df02f8518e":{"kind":"REWORK","_number":1,"created":"2022-10-07 22:44:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/860758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/860758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/860758/1"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"subject":"Fix Swift deployment issue","message":"Fix Swift deployment issue\n\nSwift deployment is broken since CVE-2022-38060 fixed sudoers file in\nthe I66476a2b396e2cbe41e68ac51f57aae1806b2ed8. The kolla-toolbox\ncontainer have their own virtualenv path differs from all other\ncontainers. This change adds the correct sudoers secure_path\nconfiguration needed only for kolla-toolbox conainer.\n\nRelated-Bug: #1985784\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: I3651576ee354364d639c187ff750491667ecab56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/673f276f6b9404c3a286011e92cb86df02f8518e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/673f276f6b9404c3a286011e92cb86df02f8518e"}]},"branch":"refs/heads/master"},"815a1d9ef8bf9d5ae21080a95d60bade2008de0b":{"kind":"REWORK","_number":2,"created":"2022-10-08 11:59:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/860758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/860758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/860758/2"}}},"commit":{"parents":[{"commit":"28424514f392f50312af51ec20db3ea83d8cbf15","subject":"Add rocky9 periodic jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28424514f392f50312af51ec20db3ea83d8cbf15"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-08 11:59:22.000000000","tz":180},"subject":"Fix Swift deployment issue","message":"Fix Swift deployment issue\n\nSwift deployment is broken since CVE-2022-38060 fixed sudoers file in\nthe I66476a2b396e2cbe41e68ac51f57aae1806b2ed8. The kolla-toolbox\ncontainer have their own virtualenv path differs from all other\ncontainers. This change adds the correct sudoers secure_path\nconfiguration needed only for kolla-toolbox conainer.\n\nRelated-Bug: #1985784\nChange-Id: I3651576ee354364d639c187ff750491667ecab56\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/815a1d9ef8bf9d5ae21080a95d60bade2008de0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/815a1d9ef8bf9d5ae21080a95d60bade2008de0b"}]},"branch":"refs/heads/master"},"01aecf455ad8ad5f4cf49fdf5ee297b078d67ca8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-08 20:40:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/58/860758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/860758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/860758/3"}}},"commit":{"parents":[{"commit":"7c41c917242b1eb651942a3048e132f0c1d6ede4","subject":"Drop executable perms from non-executables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c41c917242b1eb651942a3048e132f0c1d6ede4"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-08 20:40:01.000000000","tz":0},"subject":"Fix Swift deployment issue","message":"Fix Swift deployment issue\n\nSwift deployment is broken since CVE-2022-38060 fixed sudoers file in\nthe I66476a2b396e2cbe41e68ac51f57aae1806b2ed8. The kolla-toolbox\ncontainer have their own virtualenv path differs from all other\ncontainers. This change adds the correct sudoers secure_path\nconfiguration needed only for kolla-toolbox conainer.\n\nRelated-Bug: #1985784\nChange-Id: I3651576ee354364d639c187ff750491667ecab56\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01aecf455ad8ad5f4cf49fdf5ee297b078d67ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01aecf455ad8ad5f4cf49fdf5ee297b078d67ca8"}]},"branch":"refs/heads/master","description":"Rebase"},"b8a352647d57d35463e551a0d360c19fb2ad34c6":{"kind":"REWORK","_number":4,"created":"2022-10-09 13:43:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/58/860758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/860758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/860758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/860758/4"}}},"commit":{"parents":[{"commit":"7c41c917242b1eb651942a3048e132f0c1d6ede4","subject":"Drop executable perms from non-executables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c41c917242b1eb651942a3048e132f0c1d6ede4"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-09 13:43:42.000000000","tz":0},"subject":"Fix Swift deployment issue","message":"Fix Swift deployment issue\n\nSwift deployment is broken since CVE-2022-38060 fixed sudoers file in\nthe I66476a2b396e2cbe41e68ac51f57aae1806b2ed8. The kolla-toolbox\ncontainer have their own virtualenv path differs from all other\ncontainers. This change adds the correct sudoers secure_path\nconfiguration needed only for kolla-toolbox conainer.\n\nRelated-Bug: #1985784\nChange-Id: I3651576ee354364d639c187ff750491667ecab56\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a352647d57d35463e551a0d360c19fb2ad34c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a352647d57d35463e551a0d360c19fb2ad34c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[]}
