)]}'
{"id":"openstack%2Fkolla~860801","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I3651576ee354364d639c187ff750491667ecab56","project":"openstack/kolla","branch":"stable/wallaby","topic":"swift-sudo-issue","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-10 16:01:46.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-10-10 16:01:46.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-10 06:47:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I3651576ee354364d639c187ff750491667ecab56","subject":"Fix Swift deployment issue","status":"MERGED","created":"2022-10-09 19:39:55.000000000","updated":"2022-10-10 16:02:51.000000000","submitted":"2022-10-10 16:01:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860801-swift-sudo-issue","meta_rev_id":"a36ac1eb6b2a0786f36eb355a8e96a5985f088e7","_number":860801,"virtual_id_number":860801,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-10 16:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-10-10 14:54:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-10 06:47:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-10-10 14:54:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-09 20:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-09 21:24:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-09 21:24:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-10 14:54:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"d15f6452e9d7511ed836085ec681c2eda952a860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-09 19:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56aa45b867eb5378909b43e796f19321ab3a4fbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 20:42:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c47f82dc7444b2b232c022576b0992\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a18fec0412e43b7af7c21d4cc76721c : SUCCESS in 28m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8f3f578f0cdf4eec9101d72b046f9f2c : SUCCESS in 24m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c4a04999c47f41a8a2a9eb934eb81f3b : SUCCESS in 52m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1ea60d69d8b408f9f60803e7b08adde : SUCCESS in 48m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/08d320ec140048269f960bbec3b6fd59 : SUCCESS in 1h 00m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6af3680c5470413db37d8815feaadc0e : SUCCESS in 19m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8cd978919de2400f89f967f514f619ba : SUCCESS in 16m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a59dbea57a284e528c234adf504228a0 : SUCCESS in 58m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810d42bd2bdd497ea76416a790072d94 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/326e9fca6f2149648fcb1f2357ea546a : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97935aa514d14634b79bae1c570700eb : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa8fb1487aea45228e9f4afcd416a626 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85bd87825668492cab764ec380ee5006 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a46b95c200b473a8cdf301b3198832d : SUCCESS in 2m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/569da4f6a122481fa0860302855567de : SUCCESS in 20m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a51f25283e2c4a698e4a222f5d23d388 : SUCCESS in 18m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a223f31d7fd34c548b504e9ba0fa8eb9 : SUCCESS in 54m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44e2d7b8a9054db499813e50d2b45f3d : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50ef36222dd143a6aa661c47ae31e3af : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"67982c26042fe64903f0a59a8fc8e935cacff2c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 21:21:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da415db806ce475788138588b77d0505\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2762606dedae4199ae27549f49f8d5f2 : SUCCESS in 44m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1e638f725b64eb89f437e5869d757a6 : SUCCESS in 1h 21m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea06ba6d45d34879bd1fb34a04945cfd : SUCCESS in 1h 18m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a46d93bef294291938dfea5d1ab5d1b : FAILURE in 1h 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94a6cd44cba2d49c10ae2bbca4354429db506c3f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 06:47:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03a48e42e94814600bc8cac91b6406d6787168ad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-10 14:54:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba4b9cd8c2eb9312b9672ddd254d32e692e1a56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 14:55:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1305420fded8ba8a07e0316ad82d12eddddafb00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:01:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12e7934784564bdbacee60355156ec87\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9a0db334a7a545f993290725dbc4e71f : SUCCESS in 30m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dad95c7c04d34868925a3240fb179d2d : SUCCESS in 1h 05m 36s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/18ce16fe2a0f427dbbbcf7fcad681700 : SUCCESS in 1h 01m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83842901cdff4809a34768b17b8108ac : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b54b8379aeb94b58a81d8b014abd47e8 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cafa773914d1485d82d2d82ffa66d07b : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e9e4015a8146ecbf00979b014cf5c8 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5a356b01cde43b5b71fb41313591ea3 : SUCCESS in 3m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82fda089c9ef4612ae9faae115109a3d : SUCCESS in 16m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/355444d3de384cb293ba4faaa5786e4d : SUCCESS in 52m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/041c93f5988d48b59076749e54c78b30 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2dcdfecd65999435e46f06569541a3fc4f0c45c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:01:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a36ac1eb6b2a0786f36eb355a8e96a5985f088e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86a107ef3ea446f2a7092f5a8c777a33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939c8c3f08a84cb6b7c7946803ab5359 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc0b4e9c9dec43dd887812e9367ea257 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56","revisions":{"3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56":{"kind":"REWORK","_number":1,"created":"2022-10-09 19:39:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/01/860801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/860801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/860801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/860801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/860801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/860801/1"}}},"commit":{"parents":[{"commit":"a6ee3a29179fcfcff2043f791b91dc8016656e8a","subject":"curl: Add retries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6ee3a29179fcfcff2043f791b91dc8016656e8a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-07 22:29:02.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-09 19:39:32.000000000","tz":180},"subject":"Fix Swift deployment issue","message":"Fix Swift deployment issue\n\nSwift deployment is broken since CVE-2022-38060 fixed sudoers file in\nthe I66476a2b396e2cbe41e68ac51f57aae1806b2ed8. The kolla-toolbox\ncontainer have their own virtualenv path differs from all other\ncontainers. This change adds the correct sudoers secure_path\nconfiguration needed only for kolla-toolbox conainer.\n\nRelated-Bug: #1985784\nChange-Id: I3651576ee354364d639c187ff750491667ecab56\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit b8a352647d57d35463e551a0d360c19fb2ad34c6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
