)]}'
{"id":"openstack%2Fkolla~861092","triplet_id":"openstack%2Fkolla~master~Id1a2742843f4b17f414be5199cc5a67591b73abb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-18 12:21:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1a2742843f4b17f414be5199cc5a67591b73abb","subject":"rabbitmq: fix version to 3.10.8 for all distros/archs","status":"ABANDONED","created":"2022-10-12 15:35:39.000000000","updated":"2022-10-18 12:21:41.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b95a0016cf2c72c09a4bd9a02a0e5352607adca2","_number":861092,"virtual_id_number":861092,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-12 16:50:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 16:11:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-10-12 16:50:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 12:11:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"a6715b8bdee2364186eb1a949bbbe4a598773060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-12 15:35:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35ce984304ee01572ccc722670fa4504065e36b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-12 16:11:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fd4a9c9bbc294c19422bb7c70c10b664ff2715d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 16:50:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b00b52704d8b4feb8eec755d28062841\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7c9a1d7da3f94571aa3154b17b99aff9 : SUCCESS in 26m 13s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/27eb9ef738d844b588d255a67d389441 : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a5a6acd29f1e4aa5b721200241de6173 : SUCCESS in 16m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9aad3fe1529d457d975cc14d93e366aa : SUCCESS in 1h 01m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0e1b9a36f33442b3a86418a2a8ec298f : SUCCESS in 1h 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8711634e6ea495d9b3a3bade031bfe9 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6de6e2d27167411587bce84ce76a114f : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28134157cbe64cb781dddb262706d1c0 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa1e8115ac874ce48a5c729166096fcd : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd411c77ce341e99e9d9cfe92a91d58 : SUCCESS in 6m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/90656a2178dc46f58edaf549dbec06cd : SUCCESS in 20m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ebac57cd1be84fc0b14393859c6ce31d : SUCCESS in 1h 02m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/174a44704eab4b8380635a81d4f7c1a5 : SUCCESS in 20m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/148670469896402b88b86da02657df87 : SUCCESS in 37m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82e0598551dc4c37a6385bbd1f3f7499 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e975bd5339dc098cc31272a2dfeb8e2d3827a117","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 18:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49aeb9425b47421f8166eed90ba8ff06\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/65cfc4cfe7aa4de3bcd38d0aa51332cb : SUCCESS in 54m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/651e7bf195f94e19a0f1423612d5d393 : SUCCESS in 41m 16s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/83bb01de6de94cb58de3c8c2b2d9c923 : SUCCESS in 1h 46m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1e4451f7b724fcba108e9b714d4457b : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a9daad83c114a22b5ae746cd2c6cd7f : SUCCESS in 3m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/00bfa391ba9245cc9df5e2e4ad0ea43a : SUCCESS in 36m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3a1c3340cf6e4f93b1afaa29e46f4f8c : SUCCESS in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0d9a94ffa4756806d7410d1376e19dd10de5123","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 12:11:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b95a0016cf2c72c09a4bd9a02a0e5352607adca2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 12:21:41.000000000","message":"Abandoned\n\nwe have patches to migrate to elrang 25.x so this problem is gone","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6557f5171cd32e92c99cea41e3f120fbd36d6d3","revisions":{"a6557f5171cd32e92c99cea41e3f120fbd36d6d3":{"kind":"REWORK","_number":1,"created":"2022-10-12 15:35:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/92/861092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/861092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/861092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/861092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/861092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/861092/1"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-12 14:45:27.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:34:51.000000000","tz":120},"subject":"rabbitmq: fix version to 3.10.8 for all distros/archs","message":"rabbitmq: fix version to 3.10.8 for all distros/archs\n\nWe have Erlang 24 on AArch64 and Erlang 25 on x86-64.\n\nRabbitMQ project changes minimum supported version of Erlang in one\nbranch (3.10.[07] wanted 23.2 while 3.10.8 requires 24.2) so let use one\nfixed version to keep things from failing apart.\n\nRabbitMQ needs Erlang 24.2+, we have 25.1.1 on x86-64, 24.2.1/24.3.4 on\nAArch64.\n\nChange-Id: Id1a2742843f4b17f414be5199cc5a67591b73abb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6557f5171cd32e92c99cea41e3f120fbd36d6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6557f5171cd32e92c99cea41e3f120fbd36d6d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
