)]}'
{"id":"openstack%2Fkolla~861327","triplet_id":"openstack%2Fkolla~master~I425ae2fab78e80ff56f21fd1b24708906ec32342","project":"openstack/kolla","branch":"master","topic":"rmq-erlang-update","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-18 15:53:54.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-18 13:24:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I425ae2fab78e80ff56f21fd1b24708906ec32342","subject":"rabbitmq/aarch64: use COPR for Erlang","status":"MERGED","created":"2022-10-14 09:47:03.000000000","updated":"2022-10-18 15:55:00.000000000","submitted":"2022-10-18 15:53:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861327-rmq-erlang-update","meta_rev_id":"700d0ea8586189b59abde379b2cce907577f1772","_number":861327,"virtual_id_number":861327,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 15:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-18 14:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-18 13:24:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-18 14:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 10:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 11:07:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-10-18 12:10:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-18 13:24:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"563a60f3602ee56e956f07f902cb267a32b6ef05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-14 09:47:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88a0b4946eaa72377b000978aab902cf10439495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 10:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a2c998d70664f84916f3d0548581545\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0aa3f073909b4061950d258a2fe3950f : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/940d41ba6871454282a5e45dd1cab4d7 : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/edcf1735121f43558fe22be2339e210e : SUCCESS in 16m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1421c958380042ae908afc60ee8bbffd : SUCCESS in 46m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d076e38667d94a569cef165482f5b0db : SUCCESS in 58m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d21d98abcf41aeaf1fb057acea066c : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c8a225dcd2c4859af0affefe4190190 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7906ba4718045a9abc3ae89b3ff4346 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a619868c359c497c82cb8cc350f0ce7c : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d61de89e4fd44cd8665d45c78d90b8a : SUCCESS in 4m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b1b04b56cdf64f82a76d062a3e1c1d9a : SUCCESS in 18m 45s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/af1d1c1626b7440498181510416430c7 : SUCCESS in 1h 03m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6e982e32f4a4064b8fd59cdb9243547 : SUCCESS in 16m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e85153a226b4469fa2e6f342af93e841 : SUCCESS in 48m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64f9e22215b84ba4be7bc6afbc3b1938 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3f269f4664afca08a246509207b0cb532847b61e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-14 11:07:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d9341d1557426ff8f0caf8507abda79b951d982d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-14 11:17:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d2d9fd337dcfa61df7fa523eac759b4fd6e84dd6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-14 11:27:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e7f00dbc6a672545243f328cb0b2f7706f72dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-14 11:47:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b83a319ce522ce797e86a5adad8ae8a576056f58","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-14 11:48:58.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"25ec4dfa315eaf56a1a7786486dada645c0dcf4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-14 12:06:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0c84565bc9329d13d2ba2a1b93173af8b5b20bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 13:04:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c95cc83cab44585aca6f8e6f5bd9170\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/142c9c67934a4a1086b7a3f693311452 : SUCCESS in 20m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/be8e069ad2b84e67b3213bd331020f9a : SUCCESS in 58m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fbb032af865547e6b9b2fbfd400bdb9c : SUCCESS in 20m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/78239e5b088e468da484a7dcc3d915f6 : SUCCESS in 50m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2839b75c8306463a9c3223dfdeaa94de : SUCCESS in 1h 00m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a519c4cfad420093d105a3c02107a0 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71777ad7a67940b78c394ffbfde63cb2 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a49fc4240474150a569f29fa9a709d6 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cee7877ab66d4841a6ce358fe973a192 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cf9348b987441ebe75dbeec03d42ca : SUCCESS in 5m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bde386f1eb5847d5a7476e5e70002fa5 : SUCCESS in 23m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/be5b4b97a4bf44b1af2d8ab1b8c46716 : SUCCESS in 1h 08m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3806aa5d506f483ba483003a917c5503 : SUCCESS in 20m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ea6070f7e984b8bb1893c89dd76b4af : SUCCESS in 42m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cfa293e90d44df6a2a814348092d908 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":2},{"id":"6c80a230ef6471257c61a5c84dac54339a1cbbd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 13:15:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83b51069b2fa4072b82184ea725f6004\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/74db4c94cf0e4d9b9af613f489af82a4 : FAILURE in 32m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1eef9920e034d0dab9c130f0088db8e : SUCCESS in 30m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b82b4a68402944b78254bf186c0f4cfc : SUCCESS in 1h 19m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61dec607363b48c7a583a246c7c137c9 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5151f2ce3bcb4d879a1706eaf70cc1b8 : SUCCESS in 3m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a0c9f141eaba49eabc9838870dda7cfd : FAILURE in 31m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5a52744988ab49caa640484e4693891b : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6360017c71af4d7379c27c54b9c2cae7935bb58c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-17 15:09:56.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2efc6745934a46a504861e41cc3e025e1c55573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-17 15:25:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8eab73291fbd5b117f9cfff9fd699042e7bb456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 16:43:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff2c53fac4942fc8f640b618e3c6484\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/614c911aa7fc47f0b8126cab063acdf2 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5be8565a59d84245b10326b238a8bd58 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d9130b3c18074de4a32ad8cd8d94839b : SUCCESS in 16m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b2d59ceb5df4a93a21377814f1083b4 : SUCCESS in 53m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/92ed25bad4ca4073977b9a5a148c52e0 : SUCCESS in 59m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed131638e5e44047b65694b44a042313 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4f62e5ed7194eee809b8c02779f01bc : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd38b09955dd45b48145bd796b355c86 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2487719d61db4e6dbbf8ac2594fd1aa4 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334c93e44ec840f58e71939e786991e3 : SUCCESS in 5m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1c6e0224323345109cd516706c84e352 : SUCCESS in 22m 42s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b7e4e082e931408da65b7e9b48f2dd63 : SUCCESS in 56m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/14d2622e49fe492d9b2fe8407beb4a0f : SUCCESS in 15m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91247100c0404866b8fd1566f2fa77ae : SUCCESS in 44m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6262c5ddc1d347ab8b4d1d22a1a36c5e : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":3},{"id":"380c0138bd237ba6285f2c6cf0813c8e23688c55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:53:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee83d4953d1d4a6fa3d0f3a7f739976e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1094b0248e954991b3a2766bbda852e4 : FAILURE in 1h 32m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5edabb80370d437793d9903aa8bf8e42 : SUCCESS in 23m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7e61a2598ba4121aceccf5fec6f624d : SUCCESS in 2h 25m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95728f5a5c12462b8adbb888fe307874 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b20998d12674ddbade100e82b490a66 : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3a13d71f41ec41d3a36327a3b5ee746f : FAILURE in 1h 12m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a01044e1dced4119bdbb54fc27e3d1b3 : SUCCESS in 1h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"95869b8e0e2920cfeb2b6b545e679448de6faef0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 12:10:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1ef26de7e91656dd7693ef8e1a19c00498f79e61","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-18 13:10:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e1c5e5845fc3959d383a374369c4f6c2a1e4316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 13:23:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e1a165ece3505344871a9f6147a654ed2ee3c63c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 13:24:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12bf799298cba474604d580394316d71e646798d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 13:24:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34c52a93682d14cfa452c1152298b7b3d04b296e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-18 13:24:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e8d5b9da702b29238e4d915eccdc47b80feca166","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 14:00:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7ef1d04964d4cd845a6695437f86c1aff46ee3d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 14:14:08.000000000","message":"Topic set to rmq-erlang-update","accounts_in_message":[],"_revision_number":4},{"id":"5f2094e16458520f6b19e2d9436a855f09a3eeb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:38:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dcda3715d724209973f9426874253fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2084091f8d904f6a8c808885debc6934 : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f7016983c2874396a95bf7f215c57b57 : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c84259fe5c794b84b01d48c7eb8dc581 : SUCCESS in 19m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af343a319e85411a9520bccbfcc28c01 : SUCCESS in 55m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/32ddeaadbe4f45448dcf57513c9a257b : SUCCESS in 58m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7ee88036134803bb140a479bc07eb8 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/147631e8db1d4d538055421e167da47a : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b49453497d664748b273d3f8fe495459 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4293efa03f8b4fa4aa7281b3a0de6d01 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78d63abfc1a4299b976886c8d5e624f : SUCCESS in 5m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/603f13f2c4414d80a63aae47cb8a867e : SUCCESS in 26m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/15a4b6c36221422a83cc0979bde3b3e8 : SUCCESS in 1h 04m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/393f005bc7774fcdad8a3760df4f1a96 : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22d3c7095c9a4c9aaf5d56983d4c4874 : SUCCESS in 51m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93facb88dd4f4fc0a35480208e41659b : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f584fe111f1ef9e4083bccada611624212c4a59f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:38:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3c2102bf88ab6a772976d6053abda6f205f7cbe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:41:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d62a38f50a438bbdbeb1f167f3fac3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fbbf991184364204962c878eb572d9bc : FAILURE in 1h 44m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1c9c6997e4884c6ea0c3107de22e4899 : SUCCESS in 1h 49m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0cb70859a09e4cdd956a31028225e0c3 : SUCCESS in 2h 14m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f66ebbd318bf4744ba1a697760a872eb : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3096cc2f2764c1ba4098b1afbd66103 : SUCCESS in 6m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6b768d4e90ad40fd87358ce3eb895703 : FAILURE in 1h 42m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/419bb5407cdf4fca9664abc45f27be0f : SUCCESS in 1h 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14417ed1b5172db4f4242a01b7808d7d3fafe246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:53:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1446b6157d4936a30b7a873e0f5044\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/12c72b09c06c4d68975ede16bb9d42b2 : SUCCESS in 25m 22s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/723a8d4214224edd91f69c17da6be4ad : SUCCESS in 1h 14m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c2d664579284bb1a05d6c49ada78307 : SUCCESS in 18m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbef5bb29b9540ebbd18c4b7bf820c91 : SUCCESS in 59m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2d31d4ce232642b19493e268dbcaf69a : SUCCESS in 1h 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/762bce305be4497a81a6c4ab3b53e458 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7adfaf6df22c458aa3cf5bf46039b541 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04d83697b9f34435b81c31c563fcbd9c : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d2e9d2f0b94b989c3d6aaf9367722a : SUCCESS in 5m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e3e4907e74424af7926d59b808745c55 : SUCCESS in 31m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1a95703f42004f6782aaa0c63c23a30b : SUCCESS in 1h 00m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a0b3ebb63274ad584a94e226d7c0bf5 : SUCCESS in 25m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a3c340f0bd8414fbaf8ef5cdfb5c904 : SUCCESS in 51m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7962d80a40546b8a631f3788cc36754 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":4},{"id":"31187f17d5c8c089bc6a724adcc8cca48ce3287c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:53:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"700d0ea8586189b59abde379b2cce907577f1772","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:55:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea171d6cfb2949c2b6a23660394d9384\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0b5124397845488fe16a3ab8f3a4d5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e104ef7fdcf0faf48a8f70b73a67103258004924","revisions":{"dfe2c81ef041abc7071d27d3c41e5f5c13c02603":{"kind":"REWORK","_number":1,"created":"2022-10-14 09:47:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/861327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/861327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/861327/1"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-14 09:44:29.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-14 09:44:29.000000000","tz":120},"subject":"rabbitmq: use Erlang from COPR on CentOS Stream/Rocky Linux","message":"rabbitmq: use Erlang from COPR on CentOS Stream/Rocky Linux\n\nRabbitMQ team maintains RPM package in Github repo [1]. We then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\nChange-Id: I425ae2fab78e80ff56f21fd1b24708906ec32342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfe2c81ef041abc7071d27d3c41e5f5c13c02603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfe2c81ef041abc7071d27d3c41e5f5c13c02603"}]},"branch":"refs/heads/master"},"9b31bafd9a9516b9844939fd3fd09773962ff651":{"kind":"REWORK","_number":2,"created":"2022-10-14 11:47:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/861327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/861327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/861327/2"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-14 09:44:29.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-14 11:47:22.000000000","tz":120},"subject":"rabbitmq: use Erlang from COPR on CentOS Stream/Rocky Linux","message":"rabbitmq: use Erlang from COPR on CentOS Stream/Rocky Linux\n\nRabbitMQ team maintains RPM package in Github repo [1]. We then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\nChange-Id: I425ae2fab78e80ff56f21fd1b24708906ec32342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b31bafd9a9516b9844939fd3fd09773962ff651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b31bafd9a9516b9844939fd3fd09773962ff651"}]},"branch":"refs/heads/master"},"19c5f4bc6a1c71a947b6f008d3b0ab2d3dabae76":{"kind":"REWORK","_number":3,"created":"2022-10-17 15:25:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/861327/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/861327/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/861327/3"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\nChange-Id: I425ae2fab78e80ff56f21fd1b24708906ec32342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c5f4bc6a1c71a947b6f008d3b0ab2d3dabae76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19c5f4bc6a1c71a947b6f008d3b0ab2d3dabae76"}]},"branch":"refs/heads/master"},"e104ef7fdcf0faf48a8f70b73a67103258004924":{"kind":"REWORK","_number":4,"created":"2022-10-18 13:23:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/861327/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/861327/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/861327/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/861327/4"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-18 13:23:19.000000000","tz":120},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\nChange-Id: I425ae2fab78e80ff56f21fd1b24708906ec32342\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e104ef7fdcf0faf48a8f70b73a67103258004924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e104ef7fdcf0faf48a8f70b73a67103258004924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
