)]}'
{"id":"openstack%2Fkolla~861355","triplet_id":"openstack%2Fkolla~master~I947a0b9155fea18a7781f694dbeafc49033fef8e","project":"openstack/kolla","branch":"master","topic":"bump-node-exporter","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-18 13:25:19.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-18 12:14:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I947a0b9155fea18a7781f694dbeafc49033fef8e","subject":"Bump node_exporter to 1.4.0","status":"MERGED","created":"2022-10-14 10:23:08.000000000","updated":"2022-10-18 13:26:27.000000000","submitted":"2022-10-18 13:25:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861355-bump-node-exporter","meta_rev_id":"9c8762c2f1a00b6721ecd0231fc9bf1c2a783a0d","_number":861355,"virtual_id_number":861355,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 13:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2022-10-14 12:40:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2022-10-18 11:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-18 11:56:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-18 12:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2022-10-18 11:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 10:25:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-10-14 12:01:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 12:40:24.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2022-10-18 11:56:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-18 11:56:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-18 11:59:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-10-18 12:13:19.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"be3d521c2447fdf09843976d717c3aafab269f78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-14 10:23:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b238d667563836aa388c830f126f4dab34d5c10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-14 10:25:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07f7016121c1295bc802cb3612af9bf30aaffc7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-14 10:35:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca6abd2fd513adec9cc6d4b4a456802a8722721e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 12:01:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3ef80dfaa44d709c269bc7f70865ca\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8b9552efa924fd88aef7dc3729a8c0c : SUCCESS in 26m 05s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7ba1be133b9f47b2b8cc41d5057b486c : SUCCESS in 1h 04m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25c4f31080fd4425be2fe712267cb556 : SUCCESS in 26m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0b6f778fee954b92941a6298e7401ebf : SUCCESS in 53m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c4a13f38a616441cbdc9ebc87372aa71 : SUCCESS in 1h 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2140c0cfe96e4502b6bfeeecdbb14470 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/427beed117a642d59068266b3aa2b010 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fa16b15932e49bd85b526258371a26b : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f1478dc945548dba52454e54e8c7516 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f90dd3796f46e3b3ff0e09b93ad270 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a66631cd79be424aacbf452c37600ac5 : SUCCESS in 3m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db14f379c6174439b24473783abdbeb0 : SUCCESS in 27m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/05c0557cb4f84d67b4122151445f0956 : SUCCESS in 1h 08m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d1c130530314e3292e8d6f752be0222 : SUCCESS in 17m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/428866e448d749b7a1219f1960f6095d : SUCCESS in 40m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/25e210781a2c40acb97006ae7dcba59e : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3b8d0c5f84f212c62ca3293d71b4e63ac06a517a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 12:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9abfddb205684e1e8bc493b5034bb694\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/29627500b0a44920b02995e57a66d1ca : SUCCESS in 39m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5a0e78f873954430a89b09f4afb9346e : SUCCESS in 38m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e7780de50b54e71b9e58fa0c23e037d : SUCCESS in 1h 33m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/308be4468c7640c380088e77df78094a : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bc31577ee4f496eb09ca4fbaf2087be : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/72b0bf952c63440cb01888de7f0956fb : SUCCESS in 36m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/730e8c4ff6054758917d8eb26ff13e6c : SUCCESS in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a90fe2ce373d2b62b79be36077102e7a87116ad","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-10-14 12:40:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da233418d01ff5ea9fb3ef8b66489e81c3fd0d19","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 11:56:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fc285bd736ee5a425bccc9656e83ed9702e30791","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 11:56:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c7d3dd0aa9d82f256f46f0779aca5a9de55cffdb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 11:56:41.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d839595256d37e4da6f179f843e0b99afa2297e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:57:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f44d6e01bc5bf2a299a56fd81e29c8eafe05ffd3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-18 11:59:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be081a859c13dc25dac672adf10846491e801974","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-18 12:13:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f822cece7f396d94be42834e9e72526864a2bee7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-18 12:13:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dfd66dde407d2b3fce39fe9005c1ff5beca0ab0f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-18 12:14:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d951736a6066abeb143237a1e420e76031f93a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:25:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e810d187811d4b2fbca8e45a3344ff97\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d59e67321144473b9103a5e3466e3b2b : SUCCESS in 22m 12s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4e5129d12363449f869af1756cf538c1 : SUCCESS in 1h 11m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83e42168a7284084ab97e0ededc6446b : SUCCESS in 19m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/00c90a5203b3400a8f72837c85ffa715 : SUCCESS in 1h 00m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3943367b53fe4f14b4c34a26fe91bd3a : SUCCESS in 1h 08m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9251cf89e921422f9099442701534b7c : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ee7fe0f1d87458b8832ecb01337c738 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6007bd9837441829a1d5fe821404424 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756eca4673104ca190515ee68adb2790 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39844c823fc04b1f9acebd6e2107dfa5 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3054748f524b4c2fa3a6cc98f956ab7b : SUCCESS in 20m 36s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0df518ec32c848dc82c031d34ad89e89 : SUCCESS in 52m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7852af58e1464688addf45eac759d757 : SUCCESS in 18m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/21ac557470334c5e818fde99ac6961ae : SUCCESS in 37m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ecdd2e5d02844b539e5925bbf9e0e275 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0c2107d4e168765726e3aac526fbf9d2aa5b185c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:25:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9c8762c2f1a00b6721ecd0231fc9bf1c2a783a0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:26:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5441e1d54ae84c48816df2ef81269f1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d74e5d5c34495786d1a9739d5f1185 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ae44ccaabf64f6c989bbd1a0511d9da : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7063c02b6036e0b6e1d678d9c7af168983e9cdfc","revisions":{"38e17f5ce5c83591753d6276d4cb7d57576ac152":{"kind":"REWORK","_number":1,"created":"2022-10-14 10:23:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/861355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/861355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/861355/1"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-14 10:22:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-14 10:22:55.000000000","tz":120},"subject":"Bump node_exporter to 1.4.0","message":"Bump node_exporter to 1.4.0\n\nChange-Id: I947a0b9155fea18a7781f694dbeafc49033fef8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e17f5ce5c83591753d6276d4cb7d57576ac152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38e17f5ce5c83591753d6276d4cb7d57576ac152"}]},"branch":"refs/heads/master"},"7063c02b6036e0b6e1d678d9c7af168983e9cdfc":{"kind":"REWORK","_number":2,"created":"2022-10-14 10:35:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/861355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/861355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/861355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/861355/2"}}},"commit":{"parents":[{"commit":"2daf4331a648cc2df6982c1a6ec47a705e038255","subject":"Fix writable rootwrap/privsep config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-14 10:35:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-14 10:35:19.000000000","tz":120},"subject":"Bump node_exporter to 1.4.0","message":"Bump node_exporter to 1.4.0\n\nChange-Id: I947a0b9155fea18a7781f694dbeafc49033fef8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7063c02b6036e0b6e1d678d9c7af168983e9cdfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7063c02b6036e0b6e1d678d9c7af168983e9cdfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
