)]}'
{"id":"openstack%2Fkolla~861762","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I6b1407cc94ffa1893dac204cf4836a7d0ae0d575","project":"openstack/kolla","branch":"stable/yoga","topic":"rmq-erlang-update","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-19 10:06:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b1407cc94ffa1893dac204cf4836a7d0ae0d575","subject":"rabbitmq: use Erlang from PPA on Debian/Ubuntu","status":"MERGED","created":"2022-10-18 14:13:23.000000000","updated":"2022-10-19 10:07:50.000000000","submitted":"2022-10-19 10:06:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861762-rmq-erlang-update","meta_rev_id":"96b98c088e91a63ca1925e94bc704240ddadd152","_number":861762,"virtual_id_number":861762,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 10:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-19 08:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-18 16:31:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-19 08:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 15:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 16:31:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-19 08:38:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1868619f2ed125f9752f292e70d8b750b5b9f7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 14:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"216807f70dc1254ecd2bab0e9280bf8497f5a002","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 14:13:58.000000000","message":"Topic set to rmq-erlang-update","accounts_in_message":[],"_revision_number":1},{"id":"6c5b0d24cc4d7b4eb454a1314bf24288de427b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-18 14:19:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"64d89acd95a18dbce61d04d9ea1a19ba5e40712c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:50:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b661b75ef84cb1b9ac15202bd50393\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/63c44c827c834e4092789dfb2e1b3292 : SUCCESS in 25m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/950af272b3114bbebfd66c079c0c4dc0 : SUCCESS in 24m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bc9b24d2192d4790a3d5da04ab24e983 : SUCCESS in 1h 26m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8184e1b7b5bc49329d965ef2ba91d0eb : SUCCESS in 1h 10m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/38a89519099340afa8ef3457326f9f00 : SUCCESS in 1h 25m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/233f75bbafa144eba4b022291b22dacc : SUCCESS in 22m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d7c54b453af40e78a6ae8636f64b4e9 : SUCCESS in 16m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/23ef0d87997143fba6f90a3c274fadd7 : SUCCESS in 1h 03m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/90f63fb13859477684877d4dae4db79f : SUCCESS in 1h 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8572b5e306f841f692881ddb80880765 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afb04a0a8e744945b802460f6280a0ba : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6827f462055542a2bbe7bea52d3b6a3e : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2e83d0b3124131989ba877b6eac438 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0644876bdc24aababd5a4066f289082 : SUCCESS in 18m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a836f55fb9ac49fc80b7223b395f8977 : SUCCESS in 20m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc96c3d46a6149fa9ca521f80d6611b7 : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/48c00149cd534ab084cf78c44faccbc0 : SUCCESS in 49m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d72682aadbcf4722bba50c7b685a1f82 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b860e56ceaf325ee3d1828d0dec47888a0a8a389","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-18 16:31:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8f024da9f18c33b748a09608741705a4d7dcfb8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 17:37:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b751737d426741b3b179e0f29de68453\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2500359ddca74e94bb8b7b9f05007931 : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f7495466eab4da1b20c961289509cd9 : SUCCESS in 1h 31m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42d6ae98d1534724a6cbfdc73ba805e3 : SUCCESS in 2h 03m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d433cea513084151a08cec73a7730f8c : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e10f58d85edd4905a4125ed15ab61a24 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1c81512ec234ee8bfa1361d2706bb39 : SUCCESS in 1h 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd34f1cdc27acaac8e1c7b09c05cd4d980cbceb0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-19 08:38:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8730cfb3f92b2937315c6862d4a168242ca9b578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:39:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28225a8ba5e16368a603bb9a0f71eb484e3b93de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:06:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1000eda2a4dd4725904b38085ee4ef71\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8af4035470e046928c4bb208ad9d1a52 : SUCCESS in 26m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5709ee4ebf2c45d3a4b2187ff9ebb842 : SUCCESS in 1h 04m 51s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b6b9bb5f7e324705a138876368eb0d4d : SUCCESS in 1h 21m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d9899aa00f9042e5affab33261d8ed2f : SUCCESS in 15m 13s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c97adc9b37b4d71b51ad4a3667060f0 : SUCCESS in 55m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b933d660566b458ca98ef1933aa46b70 : SUCCESS in 1h 03m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da98f1b356141b8940636c249ab9579 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/661aaab1c9bb46339e0c439f30e0dbef : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46036d290d4c446d9137439945355f47 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92404da334594f9582d5cbe837aa7543 : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7fc993a2efe47ff88e44f3355c1d2a7 : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bb9187177a8457fb6337ca6494c32d2 : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee0e08881be8450ab9f694c03b0b3ce6 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c3b9d523c31d10dc8d6e1c718a3e3fb2315b6ee5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:06:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"96b98c088e91a63ca1925e94bc704240ddadd152","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 10:07:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7b09de6d0f4bfa8188bf04769e4c2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8908ca1527d7487b9fd76f22a310fb4b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b5e7b6cb5cabbd452543723fbc025e0d157094b","revisions":{"a2674d0b6fc9fd0775e42b0f925c20575ccc3fef":{"kind":"REWORK","_number":1,"created":"2022-10-18 14:13:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/861762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/861762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/861762/1"}}},"commit":{"parents":[{"commit":"420244f7f0850c1048b41b9b916d13114994bdcb","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420244f7f0850c1048b41b9b916d13114994bdcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-18 14:13:03.000000000","tz":120},"subject":"rabbitmq: use Erlang from PPA on Debian/Ubuntu","message":"rabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nChange-Id: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2674d0b6fc9fd0775e42b0f925c20575ccc3fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2674d0b6fc9fd0775e42b0f925c20575ccc3fef"}]},"branch":"refs/heads/stable/yoga"},"9b5e7b6cb5cabbd452543723fbc025e0d157094b":{"kind":"REWORK","_number":2,"created":"2022-10-18 14:19:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/861762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/861762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/861762/2"}}},"commit":{"parents":[{"commit":"420244f7f0850c1048b41b9b916d13114994bdcb","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420244f7f0850c1048b41b9b916d13114994bdcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-18 14:19:03.000000000","tz":0},"subject":"rabbitmq: use Erlang from PPA on Debian/Ubuntu","message":"rabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nChange-Id: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5e7b6cb5cabbd452543723fbc025e0d157094b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5e7b6cb5cabbd452543723fbc025e0d157094b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
