)]}'
{"id":"openstack%2Fkolla~861851","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3","project":"openstack/kolla","branch":"stable/yoga","topic":"rmq-erlang-update","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-19 15:12:39.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-10-19 15:12:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3","subject":"rabbitmq/aarch64: use COPR for Erlang","status":"MERGED","created":"2022-10-19 08:21:26.000000000","updated":"2022-10-19 15:13:42.000000000","submitted":"2022-10-19 15:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861851-rmq-erlang-update","meta_rev_id":"df7a088a50dd5e186165eccf8f7e24ecee84359e","_number":861851,"virtual_id_number":861851,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 15:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-19 12:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-19 12:38:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-19 12:44:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 09:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-19 09:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 12:17:38.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2022-10-19 12:20:21.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-10-19 12:32:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"775aa042d0df6642df8ce53104cd06412e071711","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 08:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3554b56ca23c75e1be2ecd9286a96b091635d593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 08:21:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"afcb1c45a5393672077cdee13ca4a25cf625708a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 09:17:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d39bb6c4338441dae6a9edb8cd58d43\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce56467d14b944db92ebdd0212507af6 : FAILURE in 11m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d435edb47ab74625b31845561b2d3b60 : FAILURE in 36m 18s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8890ca7945440c1ace961b8c8041162 : FAILURE in 33m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cafd7bac7a274c658c1b8d0355296ead : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/892bd3d04ade4ddcaf085885b5d6ccd1 : SUCCESS in 5m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b1e16e744f148eb9054d6b0bd033231 : FAILURE in 53m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7c23e769878ec93805e7cb8dced2c185d8db533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 09:38:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc22cb82a3644c9a8b4dd2a26779fdf6\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/14e272b8184a48da8868ea0de6ebe49f : FAILURE in 9m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d5debfc445de4059b90e0a4e4d403b6a : FAILURE in 10m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f24b94e612ba405e8c9ecbebf31d04ce : FAILURE in 15m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ccb49f9e20fb44d5b01645e831a78228 : FAILURE in 18m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/35afd44a90bf4bee90d3911d7fd17011 : FAILURE in 15m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ea86eb1a530846cc867fd0e53a519413 : SUCCESS in 21m 29s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/de9a4d6d20fe4e35a4b339e89f4902ef : SUCCESS in 15m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/700abf785e0241faadb3a52176a3a078 : SUCCESS in 46m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/445042b225c842e2b5118882cc046138 : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f83f459d4ea42269a0ed916c9fefa2a : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f9201f0da6b41c793d166e28b0d8328 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b153cb8a7d74a0785b85989d4b825f5 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83a14ef168d4017acd396316bea560b : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8cc5dc34857a4741a67e3917cb3c575a : SUCCESS in 22m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01c035329ea548d39edee30036ad2eea : SUCCESS in 23m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85c5e5c888f3444285042b460b0d0319 : SUCCESS in 47m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97a4b13af8274a4ca2d2f174fc074532 : SUCCESS in 51m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af6fff20fa42444bb3a37360e65be566 : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f273c40e1a7afb9d9e6dda36e30cd81bdb5449b7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 09:46:58.000000000","message":"Topic set to rmq-erlang-update","accounts_in_message":[],"_revision_number":2},{"id":"50d33e813783d495491ded5781ee2cd87cdbd2dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 09:59:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"47794e016abacd20eeb66ffb209391ac0fdf608b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:00:57.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f61494d090324e43b958b64ec39d5d4c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7eff596602ee491dbd3d484e6e06584c : SUCCESS in 53m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5adcf795f1524b3589508a280078635e : FAILURE in 43m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b8c92f6069547a0bd60d8828a07da30 : FAILURE in 44m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f2b6741f9dc40758269e31ec105c384 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c8d4786e2b64eb99b5e446ee73ed877 : SUCCESS in 3m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1d3d2024b8341588211dd547034ee8e : FAILURE in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f011a6ebe13e5acfd34538fbddfd3951363b04ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 11:20:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8027af69a8faa9a3c303d0f1f8f80d7e08a7084e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 12:17:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c78f86155461b7988131987c35555f4eace5b774","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 12:20:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"34e2f9777be6a7c989775027d1f09c3473abc3bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-19 12:32:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8d79fa497a0ac4dd6d61977be6c6ed553f250c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 12:37:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d204f4c92da3521949a1cc824534f429e199452b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 12:38:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb0c39e884d93ac17aa9adc9d1edc33645148039","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 12:38:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7ff36d0f17150987bed88cf90a7ea9cccbb4eed1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-19 12:44:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c29a0e19a0572334f54f4874855af5983600512e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 13:54:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/484ab25d98de441d8fa5d89b7987c644\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/66acabb1d8d443e89d0fa322b96fd378 : SUCCESS in 35m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/adf1aff8c5cb4265a415eb5fc2425026 : SUCCESS in 30m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/69f38af18de3429caeba88bdbbced827 : SUCCESS in 1h 07m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c89d161d82414e6c8d76e20ccb696f56 : SUCCESS in 1h 00m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6a8793ca80144801942e8cc4c69e2e6b : SUCCESS in 1h 15m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed2188140f0047a19bf1af3b715cb875 : SUCCESS in 19m 29s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1058681d3d9c4bca80a920db698d600e : SUCCESS in 15m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a081dd236a9d4765b32249d8cd54050b : SUCCESS in 54m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b9dea73dafb2471d8c2b9261e2948bc9 : SUCCESS in 1h 09m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cddf55d595d94966a976939f55d217ab : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f35645a8e35b4869bc77004f7d232eff : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3ec875244e46fe8871e20ee4ef59a3 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f9997ae5de4234ba9fbd0b953f1117 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/70376ba8011d466c81fb074a0f071b43 : SUCCESS in 24m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e36c7aabb2a4e46ad84f3221db2d6b1 : SUCCESS in 22m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b071f586c2d94b7382c1f292d2e2ccf5 : SUCCESS in 56m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71b3a85f53ae4264a3a013aa3251a626 : SUCCESS in 50m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50889ef6c5af433a9cf6450a1900bbab : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":5},{"id":"5fd66baf683e2435ef946ede3a1b24b5b524ccd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 13:54:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"58cd307be6a54761cc7bf622fcac1d8dc5aa2332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 14:40:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf1c0b9583f842b4bb9ec734837865d2\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c20797dcc2a54eb6822d6480d3954c2f : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/89c4e915ab864bfdb3eb528ff49ede0d : SUCCESS in 1h 19m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5933d5b3b3d24e7dbe94a00e4998c803 : SUCCESS in 2h 00m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5b7c74e6aef430c8c4ae87dbeb02484 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cb9be213b5e4338b6689106b5b8eb7f : SUCCESS in 6m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/142aab1d62cf440e8e2704493268feff : SUCCESS in 1h 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"15c7885e661b3eaf74b7050f1b969b583341674c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:12:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c675007d5849e0aea03a227af68e0c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2e174f0e3ba54bcc825279919fe328ff : SUCCESS in 33m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f74d34cb7ff04ad6a957d0d7ecf599a9 : SUCCESS in 1h 13m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68edf947f23e4d52b6de475b066c99d3 : SUCCESS in 1h 07m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7ddd58e4a49423a96a08a053f75d1e2 : SUCCESS in 25m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf65cb1e0afe45b3819448cba54eb5cf : SUCCESS in 47m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0301d34a7e3448d8adb3ba979a78e468 : SUCCESS in 1h 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805151b792ab4e279b6eb728bb4ca696 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5089fa48bf4245b46627c06c9d654a : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dddd75f383ae45a1bb49f78af51d3367 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a91f1f623a64df68eb50e583f05fe12 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df150d5c79a04dfd9d669e322dc67c98 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f66dc99564e4c11974423d4f3ea29c4 : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2f62e2246144d2fa4f1481e9e5611ca : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":5},{"id":"498a4fdfe2496d34114ffbba67406ced8f831621","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:12:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"df7a088a50dd5e186165eccf8f7e24ecee84359e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:13:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69818439a3e34852a27299cdf3c62bcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59d2bcbcd084265b313f7a1fc3f299f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b7c886ae08997745d63b8ba0912652bc1b533ee3","revisions":{"3c73867fbaf7a5b0304fba28cb01e443376caa7f":{"kind":"REWORK","_number":1,"created":"2022-10-19 08:21:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/861851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/861851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/861851/1"}}},"commit":{"parents":[{"commit":"420244f7f0850c1048b41b9b916d13114994bdcb","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/420244f7f0850c1048b41b9b916d13114994bdcb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-19 08:21:18.000000000","tz":120},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nChange-Id: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c73867fbaf7a5b0304fba28cb01e443376caa7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c73867fbaf7a5b0304fba28cb01e443376caa7f"}]},"branch":"refs/heads/stable/yoga"},"9e55446416d7add44cc80a487ba4fc84bde333e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-19 08:21:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/861851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/861851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/861851/2"}}},"commit":{"parents":[{"commit":"05a316f9fb4d361b363eb8b3ffe58386d738237b","subject":"rabbitmq: bump Erlang version to 25.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05a316f9fb4d361b363eb8b3ffe58386d738237b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 08:21:36.000000000","tz":0},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nChange-Id: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e55446416d7add44cc80a487ba4fc84bde333e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e55446416d7add44cc80a487ba4fc84bde333e1"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"7a7919c894818606dc224c323fd82b02c8ed7587":{"kind":"REWORK","_number":3,"created":"2022-10-19 09:59:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/861851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/861851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/861851/3"}}},"commit":{"parents":[{"commit":"05a316f9fb4d361b363eb8b3ffe58386d738237b","subject":"rabbitmq: bump Erlang version to 25.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05a316f9fb4d361b363eb8b3ffe58386d738237b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 09:59:59.000000000","tz":0},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nChange-Id: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a7919c894818606dc224c323fd82b02c8ed7587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a7919c894818606dc224c323fd82b02c8ed7587"}]},"branch":"refs/heads/stable/yoga"},"686a10d5e97c8d0ef0de146793a417b7023ec5bf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-19 11:20:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/861851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/861851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/861851/4"}}},"commit":{"parents":[{"commit":"aa92e60676d20628878de642675d6f4ae7355491","subject":"Merge \"rabbitmq: bump Erlang version to 25.x\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa92e60676d20628878de642675d6f4ae7355491"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 11:20:47.000000000","tz":0},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nChange-Id: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/686a10d5e97c8d0ef0de146793a417b7023ec5bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/686a10d5e97c8d0ef0de146793a417b7023ec5bf"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"b7c886ae08997745d63b8ba0912652bc1b533ee3":{"kind":"REWORK","_number":5,"created":"2022-10-19 12:37:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/51/861851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/861851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/861851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/861851/5"}}},"commit":{"parents":[{"commit":"aa92e60676d20628878de642675d6f4ae7355491","subject":"Merge \"rabbitmq: bump Erlang version to 25.x\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa92e60676d20628878de642675d6f4ae7355491"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-17 15:24:06.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-10-19 12:37:07.000000000","tz":120},"subject":"rabbitmq/aarch64: use COPR for Erlang","message":"rabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nChange-Id: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7c886ae08997745d63b8ba0912652bc1b533ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7c886ae08997745d63b8ba0912652bc1b533ee3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
