)]}'
{"id":"openstack%2Fkolla~861862","triplet_id":"openstack%2Fkolla~master~Ia83bd72a837609c6cf15ae9a88bd28bc589cbf9f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-27 14:07:36.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-10-27 14:07:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia83bd72a837609c6cf15ae9a88bd28bc589cbf9f","subject":"proxysql: use Jammy repo for Ubuntu","status":"MERGED","created":"2022-10-19 10:12:53.000000000","updated":"2022-10-27 14:08:41.000000000","submitted":"2022-10-27 14:07:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861862","meta_rev_id":"0d0bbd20bc88f3025109287699eeaa072daaecf6","_number":861862,"virtual_id_number":861862,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 14:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-27 12:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-10-27 11:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-27 11:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-27 11:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-27 12:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 11:31:22.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-10-19 11:36:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-26 10:18:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-26 11:03:40.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-10-27 07:39:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-10-27 12:40:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"fc8a2d0491c31e6ca6b9eeb2f5c7a35d72262828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-19 10:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6fb5c5f671c33cd4f96edee59d67bc185321683","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-19 11:31:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f8b575ef6799035d0bc40e1f6cfff1c649687bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:36:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a458b23f76d14232908d317a8a406a58\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ec16b4a0a3514b2ab79def5a80de9707 : SUCCESS in 27m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ae4a9bd8a6dc4bdbb716be87711acf23 : SUCCESS in 1h 13m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9f288267e37f44ffb025341581e7f42d : SUCCESS in 23m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d665b0e53ea34cf5985b0cbc43531777 : SUCCESS in 1h 03m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6fbae007c36041f189e6aa334f041157 : SUCCESS in 1h 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c32e5052671041149b270bada436f6f9 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/031efed7ff504c66bfcb8a080082f588 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ddeeb509dfe4d4a8a0905abf6f68f6b : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d05fc803a5104d6d8bc6a5e1cd620e28 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b86715fb2b431b9e9b85a54c440c6b : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5877e82aeeb54eb3b24a4b4db43277b0 : SUCCESS in 31m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/df828ad3b5a84d1c83a9ed6e802e37e9 : SUCCESS in 1h 05m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0c6f42863bc842d59eefc55f05dd0546 : SUCCESS in 22m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff55d51bf7314c0ba28cea6d1fe097d8 : SUCCESS in 50m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2744acaace294e238ccd24bb5db292cd : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c7da4a13171d023ef203ad21807194d80a081cba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:18:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b26f471a5b9741cab05789b3aebbc9e0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/698b870a337f47cc9459a92abbc2450e : FAILURE in 1h 10m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/edd0eaf957fd455eafa1fb3896d35e73 : SUCCESS in 1h 20m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4adda630b8cb4b1fa57e6b70c0d4da30 : SUCCESS in 2h 03m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9516ac2f0f044f4a9225c427fe262f42 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5cd9cfdf725b45539e3b0e0439835c8c : SUCCESS in 4m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8959445bc3994320bde24a9530e6526e : FAILURE in 1h 04m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f9e6126d3a644de5864e6631ee8fb904 : SUCCESS in 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bc53e776c58989ded314f0a74fdd5f82bbe962d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-26 10:18:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"953ba0f92b33aa72d01724932784b6688f3fcf7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 10:19:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46d0930e5985dd8b913c22708b87059969c8f743","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-10-26 11:03:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7afaab00bfc416c69a35473d62e60666b19604ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 11:52:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62d89e4136974d3e8782dc8db22b2f0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f766c1bd638640779504994c75557f50 : FAILURE in 19m 39s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6f1242c5080f42faa03396d9903a5f7d : FAILURE in 22m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83e5ea56664249af87ca5b3fb3b88f95 : SUCCESS in 21m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/804cd3b2d976478fb8073048da6b3187 : SUCCESS in 1h 06m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3944011f98b64f55b22190108fc30a23 : SUCCESS in 1h 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca34b436381b4b119d26c1d545c33a5c : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f21bde6510841ebb38a57e6ab40e768 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00b5e6f53db14870835b62056f279bca : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ff128db8274956a6564eaae13e7e26 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ce314dd5b65a4bc7a7beda445c4300c2 : FAILURE in 19m 18s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/40821ef145914cc799402e7a4158126e : FAILURE in 19m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/65ecf5bcaf124ad7860cd620981b72f0 : SUCCESS in 22m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a15635c7477c49c6a1ccd5c7048dd2a3 : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92516ee973624e3189d0ef4ead7b5186 : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"de6cfcc55a8068d393c8bcfdf64920041a0e7967","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-26 12:07:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a796825676e1ece7402b7cd82571ad3000324f23","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-26 12:30:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc0708d08b6ea4bd3bdc309306776212e6a19be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 13:48:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4085243c4da94df6b98ad5fbc16fe23b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/904ae8c9ebb6453c91a7a2b4e101fe92 : FAILURE in 19m 58s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6a951b0f7ac24441aeb7973f78355d49 : FAILURE in 22m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4ac75ab3e02440428ffc4115e5bc73ce : SUCCESS in 23m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/22d73b329968474785ea12d990b75a8f : SUCCESS in 1h 05m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d35878db42984362a84bb1356b891295 : SUCCESS in 1h 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80eefbccddde49ea97e29546b92ce2e1 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b182baf631c94fd5aa4cfd9ebf813d5b : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c928bb37218e457aa32c51d666183130 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28ee157efd55419fae4fb465f7dccebe : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55652f2745545caaa909ed4f4d23194 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8a419ea1795242d7be41f5d476558cc9 : FAILURE in 19m 14s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9ed75535954b40008dfea166dfba097b : FAILURE in 21m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a06c49416b97430a9fb8ca6d9a90cc39 : SUCCESS in 21m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1b087233f844d47a2b1174598d7d401 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8113a4750caa41e0a543e92bde2032e9 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b634e91de001361e776d16ce883a0748fb8fc875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 13:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4a36170c5e45f3a8b2f36a870edbc2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0524325f3a0544ed8ea04886b1e02832 : FAILURE in 1h 10m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b5e18c51c3944b0a83281ffe83797bf : SUCCESS in 1h 23m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4aafd9d95aa492c94f1cc79758a9cb5 : FAILURE in 1h 26m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/178f9ed54dea467d85d8afbb7d4a944f : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f1494b3381443e838d66ea5e0bd840 : SUCCESS in 7m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/21777f414a6c4221989e32ae360a1381 : FAILURE in 1h 04m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2c074e0942814481a4a67039a5a08fa4 : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee7ac2445d22dbab9be547159ea22c067920f05a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-26 14:38:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae8fb69be068ce9387cd467d0b196a997eb5e238","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-27 07:39:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b77aee185762e95ae93f511c726aab9f919a1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 09:01:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eef59de2a93b438a9b185ba16870b2fb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4644bda652fc446fa995a8207ea002ef : FAILURE in 17m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/da082c44ccbd4ccbbd8876d3c97c708f : FAILURE in 20m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ebc15e6dd94d4ee684c17a873e452072 : SUCCESS in 19m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/268457b6ae3a48f79ef03cebfcf4ee21 : SUCCESS in 58m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/00def4e146e24cfeb25992efcaaa2456 : SUCCESS in 1h 01m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbbeda055ad490fa4150b2c030259e3 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3188b2ebba8418aae478956b72d99e2 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e67cb0af2964095b7eb5a7a10909f5f : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/032e492bb2a24f7bb3cad69a54e49a56 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3477854826404275aed2d520719ceb5f : SUCCESS in 4m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3dd2e5568f70447fa8537cf0c7d853cd : FAILURE in 24m 08s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/7af90872af684d0da30ebc15e8adace7 : FAILURE in 19m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b44cb1651cf4453c945533e22c1db74d : SUCCESS in 25m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63154f891fc446658ba01b3e45f5e163 : SUCCESS in 47m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ec569bb85964c7e999454bfececf08b : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9698c1fc316e20eeb038cef6dbcaf382265160e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:00:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9c0bd33496494fa2e27e15e5b45d1c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/36c2a99c3a9a474eb11054adc38745e8 : FAILURE in 57m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c4a800ca6b0b487081a31ecd912086c6 : SUCCESS in 57m 09s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4f4f3b12d6248778f0e44cfc917afc1 : SUCCESS in 2h 10m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/571b8f7c30dd40a8b7ac4875184089e5 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad537e43d95544e9b84b0683f9bfb847 : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3b356cf458384f538209d021dbd4fe0a : FAILURE in 56m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3e7c0f05742346aeb088e51cfadc6c95 : SUCCESS in 57m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6322562336949646372b1493c4f4e1b320f04472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-27 11:11:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7e09f9c47d53ebc5918794dca5c3bc6043d8ed15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:28:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ffc416cde0d4458b93ab718a20a6036\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d372c8a4238f47e88a54f3d56aecc7d3 : SUCCESS in 17m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/1de90dc8382140f38620105df81345c1 : SUCCESS in 57m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d103905eb8d44db8828e59cd22ab817f : SUCCESS in 14m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11116a2c5b8341ed8737327564aade8c : SUCCESS in 52m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4c22ce5cd7743d391ebd45fd82ecd2f : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6f6d5234d244beb8eab1ce4f6cf75f : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af980c81799349d9a0a918560fabb9d5 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab5f72528aa4f3e9642463cba9131ae : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cad010555b845eaaf85c5ebc7642f9c : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98e6bfaf67c42c585d68c8dc3f297ce : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/38a9d45f7f514ac8b84439e1407cd9ba : SUCCESS in 24m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a573fed848904a109f98773d85f9cb55 : SUCCESS in 1h 10m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/45144476cb98489fa4a3830a0e406d46 : SUCCESS in 25m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/89bf20d607e14f498e5bef6870330563 : SUCCESS in 47m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b5a9f87aa83460ca0aa1c5feacfb655 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4598162af0aa9e4b7c87fb140b436f41569e726b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-27 12:40:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43a14c938f68792e351d7a193a1b2bae78f151ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:41:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9c317ed8f90084d89a935c46b38376d11241e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634e3b80e12c440f9e604ce1802c178b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3fc0bd1a1c174b0baf0c9041b6b645d5 : FAILURE in 44m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66d127a8b6694bf79bd425dcd52e81b3 : SUCCESS in 1h 12m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ea85230bd3640c7a758b7559919cc9e : SUCCESS in 1h 34m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c61bf85e89cc4c8d81606e4d33d2f09b : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00b80413a4c943498532c20657ad55ed : SUCCESS in 3m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/49021e56f3dd47c8bdb4ef2b5f2fbc6d : FAILURE in 56m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/75924228e7f8492cb814ed2cfffa6d87 : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d80342d082b1adbe2f9c6a3101c2bf20f0d0a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 14:07:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7665a918114541be8d73be94dc0c02d0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cc76d39ff6c041cdbec8dd9f39ace2d9 : SUCCESS in 25m 16s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6d0c917ced8d4065ae80ac1b56cf9b9e : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2c4b3622266d4549b85759224777a735 : SUCCESS in 22m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/585105a9f5b04a7ba1f077419e89c4d6 : SUCCESS in 1h 02m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f0bb6d43d5b04090aa3303ffb83a8dbb : SUCCESS in 1h 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6e5e167fd04cada8aeef5b9d3000d4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/279b13d4ede6499c945529c04ac12fcf : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74620dc025f7465191382efcc1805c5c : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83a5e3e1a5847189eb7980eb5139268 : SUCCESS in 5m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/becf9e480c5d4b88a673fe861a62313c : SUCCESS in 24m 32s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1575ccbc9cb3484b968c081703268c99 : SUCCESS in 1h 08m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/be30c74a3bea4d229653360603bbbd82 : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6dcc01a1de84749906062013ac88c9b : SUCCESS in 50m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c6e6ebb40aca4be2a8fc266e89716362 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"23ab7dc91be3cd7a8df8af49af38beceb7fda042","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 14:07:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0d0bbd20bc88f3025109287699eeaa072daaecf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 14:08:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938a0613838e48618297142c734f874d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383345cfb71848509f7f3fef74517d99 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1da94b16f018440894f284f1011eae24ef568bbb","revisions":{"20c8bbc3fe30a03b32a8e7fe7cc8beb5d5a15a97":{"kind":"REWORK","_number":1,"created":"2022-10-19 10:12:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/861862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/861862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/861862/1"}}},"commit":{"parents":[{"commit":"4a90585634c91ac78c8443b98578e8721db786f4","subject":"Merge \"rabbitmq/aarch64: use COPR for Erlang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a90585634c91ac78c8443b98578e8721db786f4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 10:12:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 10:12:45.000000000","tz":120},"subject":"proxysql: use Jammy repo for Ubuntu","message":"proxysql: use Jammy repo for Ubuntu\n\nChange-Id: Ia83bd72a837609c6cf15ae9a88bd28bc589cbf9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20c8bbc3fe30a03b32a8e7fe7cc8beb5d5a15a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20c8bbc3fe30a03b32a8e7fe7cc8beb5d5a15a97"}]},"branch":"refs/heads/master"},"1da94b16f018440894f284f1011eae24ef568bbb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-27 11:11:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/861862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/861862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/861862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/861862/2"}}},"commit":{"parents":[{"commit":"0fa3fd885e03e887b11b2c960564a92c423a360e","subject":"ci: disable usage of EPEL mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fa3fd885e03e887b11b2c960564a92c423a360e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-19 10:12:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-27 11:11:15.000000000","tz":0},"subject":"proxysql: use Jammy repo for Ubuntu","message":"proxysql: use Jammy repo for Ubuntu\n\nChange-Id: Ia83bd72a837609c6cf15ae9a88bd28bc589cbf9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1da94b16f018440894f284f1011eae24ef568bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1da94b16f018440894f284f1011eae24ef568bbb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
