)]}'
{"id":"openstack%2Fkolla~861940","triplet_id":"openstack%2Fkolla~stable%2Fxena~I3e2b37f106ccc096a7aede7b07197219373610e9","project":"openstack/kolla","branch":"stable/xena","topic":"rmq-erlang-update","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-28 11:16:58.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-28 13:42:18.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-10-28 13:42:18.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-28 13:42:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e2b37f106ccc096a7aede7b07197219373610e9","subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","status":"MERGED","created":"2022-10-20 07:21:31.000000000","updated":"2022-10-28 13:43:16.000000000","submitted":"2022-10-28 13:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861940-rmq-erlang-update","meta_rev_id":"2a26117541946958129130ed5dbbe0a118567e85","_number":861940,"virtual_id_number":861940,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 13:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-28 11:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-10-28 12:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-10-28 12:21:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 08:50:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 09:10:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-21 09:10:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-21 11:33:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-10-21 11:33:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f91705efb198e85dc39a523bda8f486209cf6ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-20 07:21:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79a5f0dddfd1e56852fd82b3bf7dbb6981f42835","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-20 07:28:56.000000000","message":"Topic set to rmq-erlang-update","accounts_in_message":[],"_revision_number":1},{"id":"4ec82d0eb098deb35654efc2df604df3284e6f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 08:50:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a41db851734eee841315b0b2a28af0\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/059fc6afcdd34b26b8b05f66fb5fc7be : SUCCESS in 28m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a92ed9ce7c29458ba20ee598c8266dd4 : SUCCESS in 32m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7a0b421f70d442d69635ccf61c69f1df : SUCCESS in 1h 22m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2148aa11d03e4f3bbaefbb72580cda45 : SUCCESS in 1h 18m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/62133fd1f85d47bea826f6c8ab5a5f34 : SUCCESS in 17m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/61cdf10e367a47d4a700a2136970b8a3 : SUCCESS in 19m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2181eb6756d34529afb81b44787f4d53 : SUCCESS in 1h 04m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d890d99a4a1941f789e9894e4468de7c : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed456a306a6645fa813db9400ce8b02c : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1c19bbd79134c3bbc458f73659ef5b0 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7720983021144ca68b9c0e3ad437c6e0 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4594a3c57f2a4b348fb3063d2469626b : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42591b074d544c9d92b4dc32d1bdc297 : SUCCESS in 18m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4135b195ea3f4431970c59c05d0e97eb : SUCCESS in 18m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e3a7dcc42ee49fa8a54b8cb3cd634c1 : SUCCESS in 40m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1647126a666547dfa140289752cfe32a : SUCCESS in 39m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff21aff9adba46d0a6241f53344c88a8 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c4e668b57e7a8bb51cb70dbdcb6bd9b8e1c4a125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 09:30:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0cc05f0918549a9945d3e7e5a1e70f3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8bb9bd0d05a4fa289c6ee6541540ace : SUCCESS in 1h 42m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d9f931be0af4dd88f8eeaddab1ca4b3 : SUCCESS in 26m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8db6dc985cc14742939c5b1d8a472176 : SUCCESS in 2h 08m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b344b3b2e9e94922a632696d232e4cad : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67f12d15232c4264b035530a2c3f7782 : SUCCESS in 8m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/15a9765a884444a9967d30ac2a4345dd : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c9faa53db93ce43fb2209eca647cea973170317","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-21 09:10:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abc9b033a009f6ca2f04fb7b92eab00bf8eafdd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-21 11:17:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89fe7af9bb4888cefaaa814cfbe476da0af15cc6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-21 11:32:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"47c1108552ea2f2a81f702462ee2c3fd7d2b876f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 12:30:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a640c5a7cb49da865daa91082c3cc6\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0292be2e7fe442d296919af6d8baa315 : SUCCESS in 26m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/11bfa72990e245719dc81064f8a4c9f1 : SUCCESS in 28m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f63292e4040044338d23410a7177f890 : SUCCESS in 1h 02m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd6e5ab794aa4a268b7e4b3510df4726 : SUCCESS in 1h 11m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e6184816ff754dcbb1a66478780f64f8 : SUCCESS in 14m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f59fc655dc36446c9c9f1be30854e154 : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d27b49d843b545a58659d79d4f2c02fe : SUCCESS in 56m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f11e520d584e58bf458d41e7bb3d20 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7defe6bbc4f424da0e5cc76c3518141 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03acd39e98f44314ae77d46e8bbd5c4c : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddd876ae8a8e4f23a9d2b2632ed3c6fb : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e7c5e5e0284fa2a165c784b1c32c82 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f071c8568c56442aa5f8827950bf29a0 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03580b8fd83441dbbba532fa2718f6d7 : SUCCESS in 17m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d232d6df3b54ce48c60ba2cc5e8d371 : SUCCESS in 17m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8a835b726854054bfc326bd764acbd9 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e07634f21fe84811914b92f524478353 : SUCCESS in 48m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c8853ebe31144e89f37ca6a89c8bec0 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":2},{"id":"04ff5a7a56b3cf36bd60dc7ebd2bdce0d1017deb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 13:22:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7e06f9e3d643e1b42e044376ecd8a9\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1a4ccd5801745ecb998171b7d995a8d : SUCCESS in 38m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/23d91d413e744518b519484aa07d0427 : SUCCESS in 1h 26m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b560485d740419da843b20a24ed5627 : SUCCESS in 2h 02m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b31b300dd1d47619b39d501c3f9b53c : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8921785590af47b0a42c0920cefd6548 : SUCCESS in 7m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4b89475980349a58d141f217d056a07 : SUCCESS in 1h 45m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb4ae38b8f92a53e0e532148ab25453fa2b3591b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 09:34:09.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a9edb0ce0834b14818b94ee2863653178e2e9bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-28 10:24:11.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bf8b3e1b7977fadfd1d4fa386b859c673328f40a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 10:37:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"71666ad4a6c82407da29ed4c9c01e56035d82a84","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-28 11:16:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dcf0df002b97c67e83546cc1ea3a0614dd5fc0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:40:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea34dd9851e4088a45e1780ee34ba01\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c7a121e2da5c49fb9bd27bd049f90dd4 : SUCCESS in 32m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d02a174df9b04ca2bcdb3519127ac8ee : SUCCESS in 27m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c9785b8e4aa743ea92ea0d3c95dafe60 : SUCCESS in 1h 11m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/11833377f26b4d96b6dc3cbadecf18a6 : SUCCESS in 1h 06m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6587fe45b3d64c808ab9d801206cac7d : SUCCESS in 15m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c8cd7e5e268c40a884d92ca060ed9585 : SUCCESS in 16m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56af995a70ec4e9f89e2427d0f22258b : SUCCESS in 1h 01m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd198e0bfba42f3ba95d1e8dae8ecb6 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/546cd3ad2a694da3b6971a6c3b15e252 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e0b97ca622642d9a66d02c52a5131b3 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15ecc22f206948329c110018d693f2fb : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d320ba7d1f234414abb4e89274a8f983 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aed495c6491a4accba7041f3740dfd4d : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8cdf6e5a386e4b4490f6b76d26f819e5 : SUCCESS in 19m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9faee80c43cf4c908436751505685a3e : SUCCESS in 20m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ce3d3606a1643abb7187497572547c2 : SUCCESS in 48m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c07a8d030010484eb0c8e3f28f3e76c9 : SUCCESS in 51m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb57b7644d9c43a4b72e9e154dd017bc : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9534db75913e1ed3dd50fc951b0cc5c4612b8431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:18:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7110fce33097448d9e961bf337276f3f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e25cd5333384264a2f472d0e4aaf132 : SUCCESS in 1h 51m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7733058b12e041ec8afb66d9344d9d01 : SUCCESS in 1h 33m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b16edb695644415d98cd896ef294fa65 : FAILURE in 1h 10m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1764eb8cf94e4aefb63389f8c7bd1fa0 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/498e8566a38141f38574bf806b63b541 : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/99c1a069a0974ee4a9f8490e49b30419 : SUCCESS in 1h 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a33a47dafca88f89cfd1db8c4ad1d56e72e27b6d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 12:21:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"282420fabb796ee07ab73664e9a1cde5adfec3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:22:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70694ddc6409c2332c90d65b30028046990908ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 13:42:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07c864057aa4fdc9a394cea3d614e01\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/312c3751128d4fac9330ba64bf320b0c : SUCCESS in 30m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/13cc257e519748d6bd6ebcb5e19bc2cb : SUCCESS in 1h 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b7c204bad14f2089680ef1e3106890 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1675cc10cdfd492bbc3bd20a67c404b1 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a94f78597c4595a8f05f82d150e0b5 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aae9439324140a9a5faad43acdd2578 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05b183a7108a4e62bd907a9806d84bd1 : SUCCESS in 3m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ef5dba01bdb4b578679d319f428f847 : SUCCESS in 20m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/709d7a990b6042439fdbfb7678ca80d6 : SUCCESS in 58m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c672f0f39f124ac18e0ae1416b9873dc : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":3},{"id":"ff2a52b4593c5e6fb1d281a4b7255f257607a74e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 13:42:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2a26117541946958129130ed5dbbe0a118567e85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 13:43:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1d77912fad405b85ed734e6861bc07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86819044f1c0446bbc0b148e99fafe20 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe93bc5ed044650804102184ed223be : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f3edd4d6706c97da9c954edcac6ef976366a073e","revisions":{"1ef1d6e7eda09baba4668e0ab506681f960d39d8":{"kind":"REWORK","_number":1,"created":"2022-10-20 07:21:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/40/861940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/861940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/861940/1"}}},"commit":{"parents":[{"commit":"75a21a45d326c9f196b93113e55889fceb1fe92b","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a21a45d326c9f196b93113e55889fceb1fe92b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-20 07:21:18.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nXena provides RabbitMQ 3.8 series with Erlang 23. Both are not supported\nanymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\n----------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n----------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n----------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n----------------------------------------------------------------------------\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ef1d6e7eda09baba4668e0ab506681f960d39d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ef1d6e7eda09baba4668e0ab506681f960d39d8"}]},"branch":"refs/heads/stable/xena"},"cccb406bbc7b8c3fc8b0ed3671e173ca9c93dbec":{"kind":"REWORK","_number":2,"created":"2022-10-21 11:17:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/861940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/861940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/861940/2"}}},"commit":{"parents":[{"commit":"75a21a45d326c9f196b93113e55889fceb1fe92b","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a21a45d326c9f196b93113e55889fceb1fe92b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-21 11:17:43.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nXena provides RabbitMQ 3.8 series with Erlang 23. Both are not supported\nanymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\n----------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n----------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n----------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n----------------------------------------------------------------------------\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cccb406bbc7b8c3fc8b0ed3671e173ca9c93dbec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cccb406bbc7b8c3fc8b0ed3671e173ca9c93dbec"}]},"branch":"refs/heads/stable/xena"},"f3edd4d6706c97da9c954edcac6ef976366a073e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-28 10:24:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/40/861940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/861940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/861940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/861940/3"}}},"commit":{"parents":[{"commit":"75a21a45d326c9f196b93113e55889fceb1fe92b","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75a21a45d326c9f196b93113e55889fceb1fe92b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-28 10:22:55.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nXena provides RabbitMQ 3.8 series with Erlang 23. Both are not supported\nanymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nRabbitMQ 3.9 and Erlang 25 are final update in stable/xena branch.\n\nAccording to upstream [1] 3.8 -\u003e 3.9 migration is fine while 3.8 -\u003e 3.10\nshould be done through 3.9 so we can not move to 3.10 version here.\n\n1. https://www.rabbitmq.com/upgrade.html#rabbitmq-version-upgradability\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\n------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n------------------------------------------------------------------------\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3edd4d6706c97da9c954edcac6ef976366a073e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3edd4d6706c97da9c954edcac6ef976366a073e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
