)]}'
{"id":"openstack%2Fkolla~862346","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I3e2b37f106ccc096a7aede7b07197219373610e9","project":"openstack/kolla","branch":"stable/wallaby","topic":"rmq-erlang-update","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-31 10:53:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e2b37f106ccc096a7aede7b07197219373610e9","subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","status":"MERGED","created":"2022-10-21 13:11:16.000000000","updated":"2022-10-31 10:54:54.000000000","submitted":"2022-10-31 10:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862346-rmq-erlang-update","meta_rev_id":"f40a417315b4d8889c0c60b4bed2dd51f4a82476","_number":862346,"virtual_id_number":862346,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-31 10:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-28 10:23:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-31 09:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-31 08:37:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-31 09:31:37.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 14:15:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 09:33:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-10-28 10:23:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-31 08:37:14.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-10-31 09:31:37.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"f6c28a04225339c677eff006901ccbdebdf900e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-21 13:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c53a1792e6968eb14621007cbc4d8d19f5e28b46","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-21 13:11:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"273a2067ed34447f69379e6940a7e871662d07a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 14:15:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45d091de06d446faec4252af7b3f181\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7cc3f61460c84a07a4bd9fb77295998e : SUCCESS in 31m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bb9212f23d344aab90d153968724114d : SUCCESS in 26m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d77f3b167bcb4ea6bfc074597efaf81b : SUCCESS in 1h 02m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2b6f6be8054149cfaa2a65dddb936175 : SUCCESS in 59m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/536535c54c5b4cfe8637305385b2f47d : SUCCESS in 1h 02m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b4334c97a5ad48cb997e4445600d1767 : SUCCESS in 14m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6dc88e04764b4ce492255326dc20593d : SUCCESS in 17m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c81394a38e74103b4fa7873c71bc384 : SUCCESS in 55m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc13d4a03c64936a6bf0b98d666c722 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86053e2e540947ccba8d901be04e9f17 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/906ee2b9ae0843019f401f476925b553 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71f35db4a8ac46aa90caafe5fd66059f : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1580bcbf6924848b8de6584f869ae0d : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44ae5b6887d54b259fbebf14bef5b2bb : SUCCESS in 19m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d99ba229279b44ad8d7a9f478f6cbf2e : SUCCESS in 18m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28738da0178048cdb19ffb17b5245b5d : SUCCESS in 53m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/67a300ce6ba14153a69f71be00918668 : SUCCESS in 50m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba2f2252b34e4d4fb6bc6b0d2626247d : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c8027e633e7cd787be70ef921cbe39f4546d7771","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 14:34:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a514e14315c41d08f974560902f1cb8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c69578b38f8840f8b6170e1fc3392dba : SUCCESS in 41m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bbe5f882b694444f98620b79df5dd1a1 : SUCCESS in 31m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9576a8521c3e473db8360ac172775749 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9090085b26f4758997d31c995c7d1cd : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52aeb13f813dd1857dd2dec2272227ce262d9f67","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-21 15:35:44.000000000","message":"Topic set to rmq-erlang-update","accounts_in_message":[],"_revision_number":1},{"id":"ca35eeeb43506e61ae69fbea64ad13004ab52461","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-24 08:45:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2299de8248932266800430f3d2ac2ad40971fb09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 10:16:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5602fe061cad4adea94239312c28c7a5\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/106c99b484374bcda72b9a6ad7785b40 : SUCCESS in 40m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c89f3fb322849a195bb58479e459964 : SUCCESS in 29m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a00c7c524f434a01b686c39d708f583f : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d742e867a804aef8bfc6204c875df3d : SUCCESS in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac31d8889ea6fab13fee49f8362e8e41e346eb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 10:37:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f07286395bb242c2a09e3c74bdb7278b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5e79d3c1eb32486d8b9382429e2fcf2a : SUCCESS in 30m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ce3acd8350d04478bd349cfcc7a5f7fc : SUCCESS in 31m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b597c3206d3040459ffe034b67301d91 : SUCCESS in 1h 12m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/be599305217347ac99f7b9e08e90eacd : SUCCESS in 1h 02m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ae22719205474781b787147bd6701997 : SUCCESS in 1h 45m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/baa640c1dbbd4dabbc3e6868e4d46d5e : SUCCESS in 19m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dcb49f2b418c40868c52ab65cced28c5 : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9b93ea2f8cc4580b3e3dcbabb28c59d : SUCCESS in 1h 02m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d921b944de4c9ebb835f093721e027 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f928259b32af462f9561e7009b1b34f2 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac135b26d9e246b6bbeca45d647abfa5 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a27ad91ed274013b1502cfd2671e91b : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126f5aceae8f4bdb9fb431ddec0131d6 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de6479ac4c914d3881d8550278d7169e : SUCCESS in 19m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d05bd556eb14fc8affe7d68b3845e6b : SUCCESS in 24m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb852e366a2748dc94ea6a089a755fc7 : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7bbbd6309f7469ab6fc9a3e07ff8e6d : SUCCESS in 51m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f4090a114c0b4842b9433d4f18506da2 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b44ac57b0e534f09198cbedc070fa30c3e741ee6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 08:51:06.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"ca517aafb19d679ea528dd49ecaef86a76e692b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 09:33:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55d1edfb63a2055594e37da0550a5fb372bd1de6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-28 10:21:44.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"313d13602f0fb802de78dfbedd0fd3f6e020b168","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 10:23:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84aa7a08282f21f745c245d1404dc837b776f2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 11:40:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25ab80bb6e94c85ae2b328182d4b940\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7559e5f557704b9f8683677a05aa1743 : SUCCESS in 29m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ae85b1c5ffcf49fd93cf401195dab985 : SUCCESS in 30m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d5d01adefd34594a30d020ad3d1328b : SUCCESS in 1h 05m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9500f0e766524c02ae632f5084047d71 : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/32cda6791be841a7b55a5abbc1937141 : SUCCESS in 1h 17m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/740e7ee80e0f402ebffeb3fa7980d33f : SUCCESS in 17m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2a9e45f82a114acf9b4a9943a2df42e2 : SUCCESS in 17m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4caacb71ebda4c89a79cf9212789945d : SUCCESS in 1h 12m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41382693d92c492fa597720d15f16be2 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ada49a479244555985c221b842e8fb7 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bca6039c6e64ef8b196c11160153020 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afcab6854a614d31837d30da4f4e40c4 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d3761858ec424c98b7edc52aa0fc5c : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/582bbdd80a5f4eeb8e1a640cd4928342 : SUCCESS in 20m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/76bab264bc4146f3809153a56a84f9db : SUCCESS in 20m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/04f26afd43e4428b84999510fc571042 : SUCCESS in 56m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8ab92ce28a946a49d251fd287c5cac3 : SUCCESS in 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff9e121dd00e43418a449ace3d23528a : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":3},{"id":"99febd13e9f151d1fdd19e27db9fb83744332132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:50:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/362a87b382f94ddeaa918bad2ab7d774\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d2acd9760e641ec83ac58829253d929 : SUCCESS in 44m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0c5a981633b4884980f2af43a29d7a7 : SUCCESS in 34m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ea96d1829224bbd930b7903ec32a23f : SUCCESS in 2h 25m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b152249594d24016b9ef465b78df106f : SUCCESS in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"741f1077e7ddb20cf845d65d0a2844a11d846935","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-10-31 08:37:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8c422cdb5b8f8a962a344696b70400f353f48b82","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-31 09:31:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"944b1d9099d5c7123170f2317dfa085ec6e1524f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 09:31:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60bd09fa174276e607f37dc1148cdb24a3dafe10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 10:53:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e04c555ea6841bc81924fcee546dbeb\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e3b1f3d4ab045fb8a0cb785376ea34e : SUCCESS in 20m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8c8fba67b8fa4de1bf2998c73916d858 : SUCCESS in 59m 00s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dae6a2db9d9c40338c5ebb0f5bfb61fe : SUCCESS in 1h 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a4921b60b54d349c0e41e5f8954437 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6da57efa5a8431c9fb854158d774eba : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e14724dad1864b399c69cf63da0e8653 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb9b234bd9c4f1899cb4997f85244cd : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed2b1295783044a8843f4366c4bd93b3 : SUCCESS in 20m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4727627b67b4719ab7a672c12b8c707 : SUCCESS in 52m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/819b77cb6d744aaa91eb1ebd99be7593 : SUCCESS in 42m 29s","accounts_in_message":[],"_revision_number":3},{"id":"26e061378db426aa21e128382ff0a4102f3e6856","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 10:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f40a417315b4d8889c0c60b4bed2dd51f4a82476","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 10:54:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c7d1433e3434b26bf1ac0e347e32616\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5a46223c3043faa973054feef1e7fa : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d67bf58de385d24cd5ea2b40966e76d99bfd1938","revisions":{"42930b495a687b14d487a50fc8c26df7b649d19f":{"kind":"REWORK","_number":1,"created":"2022-10-21 13:11:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/862346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/862346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/862346/1"}}},"commit":{"parents":[{"commit":"3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-21 09:51:18.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nXena provides RabbitMQ 3.8 series with Erlang 23. Both are not supported\nanymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\n------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n------------------------------------------------------------------------\n\n(manual backport from stable/xena)\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42930b495a687b14d487a50fc8c26df7b649d19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42930b495a687b14d487a50fc8c26df7b649d19f"}]},"branch":"refs/heads/stable/wallaby"},"9a364493b5d508c40c4adb4642f66e085504cea5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-24 08:45:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/862346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/862346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/862346/2"}}},"commit":{"parents":[{"commit":"3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-24 08:45:57.000000000","tz":0},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nWallaby provides RabbitMQ 3.8 series with Erlang 23. Both are not\nsupported anymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\n------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n------------------------------------------------------------------------\n\n(manual backport from stable/xena)\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a364493b5d508c40c4adb4642f66e085504cea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a364493b5d508c40c4adb4642f66e085504cea5"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"d67bf58de385d24cd5ea2b40966e76d99bfd1938":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-28 10:21:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/46/862346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/862346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/862346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/862346/3"}}},"commit":{"parents":[{"commit":"3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0c7e31a3616f41bfac9ae8ae8c6ce453ae6b56"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-10-12 15:58:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-10-28 10:19:20.000000000","tz":120},"subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","message":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25\n\nWallaby provides RabbitMQ 3.8 series with Erlang 23. Both are not\nsupported anymore.\n\nThis patch brings RabbitMQ up to 3.9 series with Erlang 25.\n\nDue to fact that 3.8 does not support Erlang 25 we do all updates at\nonce. Therefore this commit is composed of 3 backports from stable/yoga\nbranch:\n\nRabbitMQ 3.9 and Erlang 25 are final update in stable/wallaby branch.\n\nAccording to upstream [1] 3.8 -\u003e 3.9 migration is fine while 3.8 -\u003e 3.10\nshould be done through 3.9 so we can not move to 3.10 version here.\n\n1. https://www.rabbitmq.com/upgrade.html#rabbitmq-version-upgradability\n\n------------------------------------------------------------------------\n\nrabbitmq: use Erlang from PPA on Debian/Ubuntu\n\nPPA from Team RabbitMQ [1] provides newer Erlang for Debian/Ubuntu.\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://launchpad.net/~rabbitmq/+archive/ubuntu/rabbitmq-erlang\n\n(manual cherry-pick from master commit\nb40506f6dd382e1b2443dc59b918b01e2db45850)\n\nId: I6b1407cc94ffa1893dac204cf4836a7d0ae0d575\n\n------------------------------------------------------------------------\n\nrabbitmq: bump Erlang version to 25.x\n\nWe decided to ship one Erlang version for all distributions and\narchitectures.\n\nRabbitMQ 3.9.22+ requires Erlang 24.2+ and 24.3.4.2+ is recommended due\nto CVE-2022-37026.\n\nId: I1eb1e1bd2bf74dbef214d458cbde13aa8c2a286f\n\n------------------------------------------------------------------------\n\nrabbitmq/aarch64: use COPR for Erlang\n\nRabbitMQ team maintains RPM package in Github repo [1]. Hrw then use it\nto build package using COPR service [2].\n\nThis way we have latest Erlang on both distributions and both\narchitectures.\n\n1. https://github.com/rabbitmq/erlang-rpm/\n2. https://copr.fedorainfracloud.org/coprs/hrw/erlang-for-rabbitmq/\n\n(manual cherry-pick from master commit\ne104ef7fdcf0faf48a8f70b73a67103258004924)\n\nId: I6a6a9c3d5ffff884fc61baeafcc46ad82afc3db3\n\n------------------------------------------------------------------------\n\n(manual backport from stable/xena)\n\nChange-Id: I3e2b37f106ccc096a7aede7b07197219373610e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67bf58de385d24cd5ea2b40966e76d99bfd1938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67bf58de385d24cd5ea2b40966e76d99bfd1938"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
