)]}'
{"id":"openstack%2Fkolla~862724","triplet_id":"openstack%2Fkolla~master~Ibf85ae4b5b374b4831b9ba33addd896c0507e8aa","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-27 19:23:14.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-27 19:23:14.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-10-27 19:23:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf85ae4b5b374b4831b9ba33addd896c0507e8aa","subject":"Revert \"ci: disable usage of EPEL mirror\"","status":"MERGED","created":"2022-10-27 16:41:44.000000000","updated":"2022-10-27 19:24:16.000000000","submitted":"2022-10-27 19:23:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":862788,"submission_id":"862724","meta_rev_id":"7536af8e669c130c06400ea09c74e23b00d33b64","_number":862724,"virtual_id_number":862724,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 19:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-27 18:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-27 18:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 16:41:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-27 16:41:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-10-27 16:41:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-27 16:41:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90914781172dcbc10a8b414719ce6e5d3a139581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-27 16:41:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f2bf006b03e43c37fb37c65d5656c69aefe846f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 17:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1fd2fc5c76240518588b93dcb12a672\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1df6ba1cc37743e2a91b59ffc53ea8b2 : SUCCESS in 21m 35s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/14aab1172df24778bc9e0ab1f297b22d : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6446931487c54373bf9a9802fefe1351 : SUCCESS in 16m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e2f66816a7bd47a2b57c2fe399213acf : SUCCESS in 56m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/496b51b6712c4417b4c529833d66e14d : SUCCESS in 52m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60d6b8de96646219a7edea33ea974c9 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1574ea7b0b6c4904aef8134b3a0ac7bf : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cabca16781d84c28a0323cb2cfccc457 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a2ea44358844d94ac6df5ddecf97ba8 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317fdb1d1b484d90b466dac3a69a781a : SUCCESS in 4m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4b02ed35f08e412596b480511df78629 : SUCCESS in 26m 44s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6d86cf4951e54789924b73d6cc87fa92 : SUCCESS in 59m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/215cae2d8c1a4f16868735906df99a14 : SUCCESS in 18m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3453881c505492c9c9e8315583ca941 : SUCCESS in 41m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62023785661944d4b916de751e3a837c : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b8416a0cc8e97a7828800b9a1a80ea53f717fdb0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-27 18:07:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4726a5bd77784584abda37e5c695cd241ccf6da2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 18:07:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33f8498e5ad57ed83bbe25c8b01f8b9c64f6fb99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 18:23:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a894386cc9b4460db4d79a65e736f25c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0ef6a7cf1593438baae715baed1badda : SUCCESS in 37m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/10b300e34c29418fa0b2643437971907 : SUCCESS in 35m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cbc2a8c3fc444258ff97492ebf0085d : SUCCESS in 1h 32m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e127349a14674538b1f54a55f8f5139f : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35a0c5fa819d44e4bcdacd65bbf3e3cd : SUCCESS in 3m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b3c801734e2947dca5d386d5efd7d3bc : SUCCESS in 35m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4294f084205c4384bb516f848260b1a5 : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e1daf58bc72c72794b2dce20353e035132f2987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 19:23:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d16d801453f48bca806d25cb76925c9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d78d46267a0142aea4d4bf10b087090e : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/30f687c320c54ca7bafc534ffc35c49d : SUCCESS in 57m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7cd8721fccb43579b22966bfa897178 : SUCCESS in 21m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a1532e023cb3430b9d6f630693d39760 : SUCCESS in 1h 07m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/56de0016f8ea46c0b6c143d9fbb76f66 : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bfd7762e8e244fdb9da02677ea76de4 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/022fbde4da2247be8869663a60313d1b : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/839cc5386d8147f6a683e0a329577056 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed591b6e356745e79ebacb61af020376 : SUCCESS in 5m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/717f19673460440682fdfff130a8f6b4 : SUCCESS in 30m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/67506b8635fd47b9a082961d987d36c6 : SUCCESS in 1h 14m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2b0a21f517fe4f82b0ad022035c272f6 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20ba5aef620149c58b22b9c32aafd61e : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2eb372be1214f04924ffe23d00462d8 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"99fc381d8be71568b040af276c2d91227d5099d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 19:23:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7536af8e669c130c06400ea09c74e23b00d33b64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 19:24:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f7fd4602f149d5bda96cd698101e07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad25ff88fba44097ae2eabe520474157 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32400a5821cbfc146cd4774fb7c77809414e2049","revisions":{"32400a5821cbfc146cd4774fb7c77809414e2049":{"kind":"REWORK","_number":1,"created":"2022-10-27 16:41:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/24/862724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/862724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/862724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/862724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/862724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/862724/1"}}},"commit":{"parents":[{"commit":"0fa3fd885e03e887b11b2c960564a92c423a360e","subject":"ci: disable usage of EPEL mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fa3fd885e03e887b11b2c960564a92c423a360e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-27 16:41:44.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-27 16:41:44.000000000","tz":0},"subject":"Revert \"ci: disable usage of EPEL mirror\"","message":"Revert \"ci: disable usage of EPEL mirror\"\n\nThis reverts commit 0fa3fd885e03e887b11b2c960564a92c423a360e.\n\nReason for revert: OpenDev mirror should be fixed now\n\nChange-Id: Ibf85ae4b5b374b4831b9ba33addd896c0507e8aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32400a5821cbfc146cd4774fb7c77809414e2049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32400a5821cbfc146cd4774fb7c77809414e2049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
