)]}'
{"id":"openstack%2Fkolla~862788","triplet_id":"openstack%2Fkolla~master~I4d881e6a1af07f7c6cf4547cac72e971bfadeefb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-27 11:47:10.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-27 12:07:09.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-27 10:22:06.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}},"hashtags":[],"change_id":"I4d881e6a1af07f7c6cf4547cac72e971bfadeefb","subject":"ci: disable usage of EPEL mirror","status":"MERGED","created":"2022-10-27 09:19:57.000000000","updated":"2022-10-27 16:41:44.000000000","submitted":"2022-10-27 12:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862788","meta_rev_id":"2806ebf245d0f2a21f206f68d486dd16993f239c","_number":862788,"virtual_id_number":862788,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 12:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-27 11:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-10-27 10:22:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-10-27 10:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-10-27 10:54:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 09:20:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-27 09:20:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-27 10:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-27 10:57:47.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"0cf7eb2392f3454d882712595eb83edc3c5e78e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-27 09:19:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84f971ed8935a1537a536026bef79d8d5da8e970","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-27 10:22:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68bda87561c9d9b6d3e97ce8a4b69e8e04d26b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:32:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e47c43274547a5bb5d631ded0cf772\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6ba4b5035ccb49aa9e87e9c8fb99d69f : SUCCESS in 23m 15s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/8dd424786ce54b228ad913de1efaa953 : SUCCESS in 53m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bd9d8fca478a4d6f872272238d9c0274 : SUCCESS in 19m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a06a85305be14a0db505e55ae0c0df2d : SUCCESS in 44m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/02975c30a0cf4db19fa6c85b80883ef8 : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09e95714b974ffebbea03cc549e7429 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07bc2a872251453799f3b928fad222db : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57e8a366042748de86b3db26491a6243 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3b4e4f3946744f9a2e0263e2271c642 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e82fdb5d8ee4d1b97459d785e894f66 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/62950759853846f4bb5d15540d81c360 : SUCCESS in 25m 52s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/18b7a62a00db4addb2dc9b9377f1b432 : SUCCESS in 1h 11m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35cb4d7c65184df8a51f4a26bb471fbb : SUCCESS in 24m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec0f339dbe6546e78fc45565c452dce9 : SUCCESS in 48m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85915ab3fa2c473fae369d31a9ce3524 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b1e59f430c650c7ca438f4d59e3ae0f9464319de","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-27 10:54:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83972a40d975e7920be7e31b478ad254ab8441f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:55:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64264fa41fedb95a614af6ee5143b68cee4251e8","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-10-27 10:57:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e512bfe14a8ed859405b095e9e1a3783535d120e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-27 11:47:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d0f393fe299edc0842ce268b601fb25c6fd0481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:07:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/468d3287a90e4ca4bf8f870850bb314a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cc5980dd203c436faefd8031f4232484 : SUCCESS in 20m 11s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/677c16c7295f40a1b9bfdca61c440c35 : SUCCESS in 1h 00m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f81566d8dd10417794d53f195602898d : SUCCESS in 19m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d810c3af08f24eee958b97aaf47a2e3d : SUCCESS in 51m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3c19da53efb04c2f908af62d35a235ae : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c477c6ae6d24302b7dedc2d38cb08cb : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c16eecba0f24d6cafb0154c9e34e7b6 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8080452df7c9466f8c79a8fbc4483d63 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bed250dab54b0a92339420c32de4f4 : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b3fcf728c0ae464bb2780d5d729717d2 : SUCCESS in 24m 33s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/dbd510fb7b3d4b9b9733666e97db89d0 : SUCCESS in 1h 09m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/34d23dd3009944dea4d37d22d2c70689 : SUCCESS in 18m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/473e6705446041d88cb4679c780e7b14 : SUCCESS in 39m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87fd3f42e7b84cc292d364eb02ca9620 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9bd9ca741f467c7388e42126103be8d078ce7b82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:07:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4540286a384f15a287a6c632f60df1c5094e1ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:08:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debd4e92f3824e728430c9ee396adf6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e952880dea5449a9b78d084d7b255bc3 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"c3133f3d986d22f43ff54637433e3d2740be7bd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 15:32:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfd9a6e843a64919838c196bd7d74149\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e34f4c7047de48e8be0f21157a6ca9b4 : FAILURE in 36m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ded52d2e3704e26b4428acb920d31f2 : SUCCESS in 34m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c04be69d36be4c5eb511a4cd17574974 : SUCCESS in 1h 25m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a252ae98e0144e8eb9af76dd8099bd37 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a15142de8658437eb3d0c5700b64dae6 : SUCCESS in 3m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/080bc5270d51493092619040f820f55e : TIMED_OUT in 6h 00m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c7c6756958cc40b4b5d4582f8bc56c76 : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2806ebf245d0f2a21f206f68d486dd16993f239c","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-27 16:41:44.000000000","message":"Created a revert of this change as Ibf85ae4b5b374b4831b9ba33addd896c0507e8aa","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fa3fd885e03e887b11b2c960564a92c423a360e","revisions":{"0fa3fd885e03e887b11b2c960564a92c423a360e":{"kind":"REWORK","_number":1,"created":"2022-10-27 09:19:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/862788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/862788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/862788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/862788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/862788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/862788/1"}}},"commit":{"parents":[{"commit":"4a90585634c91ac78c8443b98578e8721db786f4","subject":"Merge \"rabbitmq/aarch64: use COPR for Erlang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a90585634c91ac78c8443b98578e8721db786f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-27 09:11:51.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-27 09:19:55.000000000","tz":120},"subject":"ci: disable usage of EPEL mirror","message":"ci: disable usage of EPEL mirror\n\nIt\u0027s broken now - let\u0027s stop using it.\n\nChange-Id: I4d881e6a1af07f7c6cf4547cac72e971bfadeefb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fa3fd885e03e887b11b2c960564a92c423a360e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0fa3fd885e03e887b11b2c960564a92c423a360e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
