)]}'
{"id":"openstack%2Fkolla~862841","triplet_id":"openstack%2Fkolla~master~If7ca1fe249c63c68c5cfe8155df88f8edcf2c352","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"26721":{"account":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"last_update":"2023-09-26 10:36:24.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-26 10:36:24.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-02 07:39:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2023-09-26 10:36:24.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-09-26 10:36:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If7ca1fe249c63c68c5cfe8155df88f8edcf2c352","subject":"Revert \"Remove ec2-api image\"","status":"ABANDONED","created":"2022-10-27 22:29:13.000000000","updated":"2023-09-26 10:36:24.000000000","total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"b066a84dc44be1eda6bc50f7ffda709a34b95d75","_number":862841,"virtual_id_number":862841,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-27 23:51:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 23:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 11:33:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-10-30 23:52:55.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"CC"},{"updated":"2022-10-31 05:41:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-08-26 19:51:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"93f335d2db7ca74ec08598a24f2d5bd93fd33535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2022-10-27 22:29:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e58c89e798f5a273d6758fb4890ba3c9958479f","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2022-10-27 22:32:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb9be442e223b8b887cc71ebf3a3c7ba259bdf3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 23:51:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297797a1b7184eb2b91ef93a9b37b26f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/222b1d10fc2f4259b7392b0623636235 : SUCCESS in 27m 37s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/0fcaed078ff14305ae8faa04ef8ef3eb : SUCCESS in 1h 15m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7df963cdd104bd599e2a5c926c0ba27 : SUCCESS in 21m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7af213f96f3b42ea8678f825f8c7a518 : SUCCESS in 1h 19m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/810445b269874910ada252fd3f863d0b : SUCCESS in 1h 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca48b7833f274484a840aae5acc072a0 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0b65ed864b94753be0e6f248b73a012 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a98bb014e8849fe98508667c8bb2017 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05540b34b7f948cf9e98ee8f35f4d45a : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6895d577baf040fa9f0682d989b5d180 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/16a14139099241f388e332cb39b868ed : SUCCESS in 35m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/91d39d57345247d5b9dd2c19ad6791c5 : SUCCESS in 1h 17m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fb52a3f58b4843878d7cc1fb761962ed : SUCCESS in 21m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/98351aa0c02047909e8f4d484ce9b522 : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2a3a0086f3743f592f2dd45ce15af4a : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"639c7f94d07c6315255f87150b332f3095ead760","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 00:01:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0486ef07c241798b31b37df84a7db4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e2c7fb2ae22e4265b0528ff707dff66f : SUCCESS in 30m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f84bf936b89944b19bfc527b79e6f4db : SUCCESS in 28m 40s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a2a6a6f761884a478fcaf5f062689a3c : SUCCESS in 1h 24m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc8844559b4c4ebe85ff272c53226a0e : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f8369bf59c8420d8e89f1b814a9c4d8 : SUCCESS in 5m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/44faf450686443778dd69f8ded5307e8 : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/30a330e8c1eb4bc791c5223f89a81bb6 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ac2d1ca17311fcc2b2befc6eb60957a650321c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 11:33:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d680e20a63d03e05b3535f98420ad5087ff4e703","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-10-30 23:52:55.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a3a2268ad5b72bee4cb544353cb3b26d644f4be4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-31 05:41:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29581c40f944104493e1b97db774c5d8e09232f5","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2022-10-31 12:38:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b64857f7905f44f5ee136d9867e411bdf55c87fc","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-11-11 04:08:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"414f68a2b2ae70a666d7720307b4dc4c307dd23f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:39:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b066a84dc44be1eda6bc50f7ffda709a34b95d75","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-26 10:36:24.000000000","message":"Abandoned\n\nSeems no interest from the author, if you\u0027re still interested in reverting - please restore the patchset.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e8f4d7478e4a9cfc65da5b7959a8e5a083a58ae","revisions":{"3e8f4d7478e4a9cfc65da5b7959a8e5a083a58ae":{"kind":"REWORK","_number":1,"created":"2022-10-27 22:29:13.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/41/862841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/862841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/862841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/862841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/862841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/862841/1"}}},"commit":{"parents":[{"commit":"718eaef3e383b5161dcc58ec6f24a527b1856957","subject":"Merge \"Revert \"ci: disable usage of EPEL mirror\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/718eaef3e383b5161dcc58ec6f24a527b1856957"}]}],"author":{"name":"Shi Yan","email":"yanshi.403@gmail.com","date":"2022-10-27 22:25:30.000000000","tz":660},"committer":{"name":"Shi Yan","email":"yanshi.403@gmail.com","date":"2022-10-27 22:28:43.000000000","tz":660},"subject":"Revert \"Remove ec2-api image\"","message":"Revert \"Remove ec2-api image\"\n\nResurrect ec2-api image as it is still supported in openstack.\n\nThis reverts commit 69aa2ab56e50c2c07dbbd79f0710788a272d1874.\n\nChange-Id: If7ca1fe249c63c68c5cfe8155df88f8edcf2c352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e8f4d7478e4a9cfc65da5b7959a8e5a083a58ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e8f4d7478e4a9cfc65da5b7959a8e5a083a58ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
