)]}'
{"id":"openstack%2Fkolla~866228","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ib3a37eebb29d975fc51a117cecdff74baafd8941","project":"openstack/kolla","branch":"stable/yoga","topic":"bug/1969096","attention_set":{},"removed_from_attention_set":{"35536":{"account":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"last_update":"2022-12-01 10:54:21.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-12-01 10:54:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3a37eebb29d975fc51a117cecdff74baafd8941","subject":"Fix plugin builds with sources using type\u003dgit","status":"MERGED","created":"2022-11-30 22:54:40.000000000","updated":"2022-12-01 10:55:21.000000000","submitted":"2022-12-01 10:54:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866228-bug/1969096","meta_rev_id":"c4ac28d044a28a32e6d657c04e621917df5a3ba2","_number":866228,"virtual_id_number":866228,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 10:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-01 09:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-12-01 08:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-12-01 09:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 22:54:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"state":"REVIEWER"},{"updated":"2022-12-01 00:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 08:43:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-01 09:20:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"51fd2bf5b0bc6a0983a56eebf91825d9477dbad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-11-30 22:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e588f49c4444b3c776b557466e759070e13ae6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 00:25:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3ddc809b8e45b08d2f07c0de82ae4d\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b704886cd8174acb9bd4dfe6cfdd3f54 : SUCCESS in 35m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2f156d57264c4d8c807d4699cfb2ce40 : SUCCESS in 36m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4429e7e6194149c3af99841b1fbd87c8 : SUCCESS in 1h 18m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1e4fcccbe9884914bff48aa3ad1b24df : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/159023fae9454deea27971b179294d7c : SUCCESS in 1h 27m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5cc242e7fa0413cb16e85ffde597419 : SUCCESS in 24m 12s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e4b2813e0e2d48538172206b103ac414 : SUCCESS in 20m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e3a98fa727e49cf9e0d751ab98d7d90 : SUCCESS in 1h 10m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c8037f5395db429d8e91291563db8533 : SUCCESS in 57m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434319288954469c89a7086faf66705b : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df7fbe4d538e4723858a494d31e68f88 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e4c30bf9ec8419a83bbe7d8ebf9c5cc : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3266fc2a638a43ad916a936c2d29e876 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29393f7f3603410596ec35d13743fa09 : SUCCESS in 17m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/973dd1bf19cc484ab436048b3ae8799c : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1e86faafa814065b9e0fbda2121b7fe : SUCCESS in 53m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6815f7a6c06e4e73a993e9945172bb7e : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/20140140ecb84d4286b57d4123d65b60 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"270df8175800be9f0ed871b917e2e64e249f3861","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 04:58:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11322170b2940a486c925f37bcf4d4e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f770b8ea08d4d299ed5e112fd9ed91a : TIMED_OUT in 6h 00m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e84cbfa7788f43adb34164b152a41803 : SUCCESS in 2h 27m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8b6f8f25ef84a74a76794b9a54b9b0e : SUCCESS in 2h 50m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/011c721e1d964c97ad950c0ad2bf5800 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a1a335f871d4530bf49174d29d24329 : SUCCESS in 5m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1a5a213b58d492db47d65d15487ab58 : SUCCESS in 2h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2349b854090a95b1373a548d8cb5c1917752ac81","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 08:43:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46e478bf9415744f321ebe5a90acda15cb9c4a97","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-12-01 09:20:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc7433c44c199c700f34755975b14ae89f9e0ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:20:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d32bf3194ca46fab7937bc148736edc69f2ebec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:54:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daef08e5314646aa92931dd1bba28ce5\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b1df9dd9e7264abcb47471f3d6b2452e : SUCCESS in 28m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2235ce871f15459e86e6793185505988 : SUCCESS in 1h 19m 25s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/45ab43a6b3724f2e9cb4fe4d491f3261 : SUCCESS in 1h 27m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ce235d0f17c4d53abbd56fb44364b96 : SUCCESS in 18m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed250fc74266415c8090d2e6d5711a56 : SUCCESS in 1h 04m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/89fc7294c09a4c6d84deca9503dc5fdc : SUCCESS in 1h 07m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98585fe2c79541e1a6bdadc0c8e5d935 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d6fef9b6764a2b895076954590ee2b : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd35c2930dd647eea2d7816d7331f0e3 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7870f0c44900473daafdcf56d663b052 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af96e811ecf041f09592343c61e79344 : SUCCESS in 19m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f0c598852144db8b6076cff839d88a2 : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8cc6c46287c454ca249cf32655cd5b0 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d526bcb2c6d890037c61b3c733e7b2a09b33de1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:54:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4ac28d044a28a32e6d657c04e621917df5a3ba2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:55:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81efac4885df48229f5657378d32eea0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187772ea98224efc8fee4186707f1bee : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ba7df36182e709f154e8ec7a482464d054e99f6","revisions":{"6ba7df36182e709f154e8ec7a482464d054e99f6":{"kind":"REWORK","_number":1,"created":"2022-11-30 22:54:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/28/866228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/866228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/866228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/866228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/866228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/866228/1"}}},"commit":{"parents":[{"commit":"b7c886ae08997745d63b8ba0912652bc1b533ee3","subject":"rabbitmq/aarch64: use COPR for Erlang","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7c886ae08997745d63b8ba0912652bc1b533ee3"}]}],"author":{"name":"Jakub Neumann","email":"jneumann@cloudferro.com","date":"2022-11-24 09:29:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-11-30 22:53:07.000000000","tz":180},"subject":"Fix plugin builds with sources using type\u003dgit","message":"Fix plugin builds with sources using type\u003dgit\n\nA recent change to git [1] introduced a new behaviour to work around a\nCVE [2] that disallows any git operations in directories not owned by\nthe current user.\n\nA fix was introduced for general checkouts, but it was not applied\nto the plugins archive, resulting in PBR still not working as intended.\n\nFixed conflict added in I093620679016b37e1664c9fe4cf7559433e744b7.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n[2] https://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2022-24765.\n\nCloses-Bug: #1969096\nRelated-Bug: #1968877\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n\nSigned-off-by: Jakub Neumann \u003cjneumann@cloudferro.com\u003e\nChange-Id: Ib3a37eebb29d975fc51a117cecdff74baafd8941\n(cherry picked from commit 6be0068f376b0ae67bc81b50a97e042a88317d28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba7df36182e709f154e8ec7a482464d054e99f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ba7df36182e709f154e8ec7a482464d054e99f6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
