)]}'
{"id":"openstack%2Fkolla~866229","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ib3a37eebb29d975fc51a117cecdff74baafd8941","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1969096","attention_set":{},"removed_from_attention_set":{"35536":{"account":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"last_update":"2022-12-01 10:54:24.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-12-01 10:54:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3a37eebb29d975fc51a117cecdff74baafd8941","subject":"Fix plugin builds with sources using type\u003dgit","status":"MERGED","created":"2022-11-30 22:55:25.000000000","updated":"2022-12-01 10:55:24.000000000","submitted":"2022-12-01 10:54:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866229-bug/1969096","meta_rev_id":"80ee99547dffa9a5d92b78091ac1366ff1542161","_number":866229,"virtual_id_number":866229,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 10:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-01 08:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-01 09:21:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-01 09:21:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 22:55:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"state":"REVIEWER"},{"updated":"2022-12-01 00:10:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 08:43:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-01 09:21:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"eab647249d93341a48d0c665a94056bc91610ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-11-30 22:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52e31df7e581dfc890c18a478aa4ba1a115179db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 00:10:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83457d6971214fda9aaa0d3b3143e569\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4f3b5196840440c6a1f53c92e51a1a4e : SUCCESS in 27m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8b6517e582ef495eb91455adca389ff4 : SUCCESS in 27m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/95e58349afa545b499b61c418033090b : SUCCESS in 1h 04m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2c7a9570930e46ffbce5693a0624a1c4 : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/14bf8f794ea0468197c567160de99b8a : FAILURE in 14m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc896c798295401eaa371a3b7f817a8d : FAILURE in 15m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3f1f658284514547b6f02536b5adf699 : SUCCESS in 54m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9307db20ade34f1b8e319ba8e615b57c : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23266c33492a4fbcade91b9708ac6d5c : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33a5e4d412f74bae9519ceed578eca79 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/471e43605870490781ad4d44709787ac : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbe65c30c174e50a21e877f358ed950 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7e7a069e67440eba313140504e71f4c : FAILURE in 19m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/12515aef038e44d884247547d598c4b9 : SUCCESS in 14m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff25a7a2c0b0479aa64731117bf4ff95 : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1e0000ab56f4da1b43803141ff2f1fd : SUCCESS in 52m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae4e12847f234780aa983820c108f526 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1175bb1d938833273aacbf2f8036187c1410acad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 01:28:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61fadb61b224186a873dc5aba97995c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f8d38a540174d51b85c9f381e61997d : SUCCESS in 2h 31m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d9d472bd4b134efbb54c902bf8570fcf : SUCCESS in 2h 07m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/48c430d2b73245689386a756cb317562 : SUCCESS in 2h 18m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/735e63dd33844c76a376ec85444e284f : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f02166b6ba8f47718c9db42fb37216cf : SUCCESS in 6m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f05ba1687dc349aa922e9aa167b108df : SUCCESS in 1h 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"532b6f7633fb905e19e264c1232308568855a7d5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 08:43:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f92602b0e6f464a7a64dec8f21abde9b62262520","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-01 09:21:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a7f507083bfe77bd9f48cf255bcc6a54d50033d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:22:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06552008e1055bbf40935f2fa2c7b49d020fb20f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:54:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f28d3187c14ff39e777a7f1b822097\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/35bee557127f44e190f8dfaf7e276d25 : SUCCESS in 25m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5e53ffbb1cb4446a82d07fffd1e1c76c : SUCCESS in 1h 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb08df86e7ae48a996b3f92fdb2712f4 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/613a1be74c7142489720eaac3d9a0439 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6abe5708a87c40568c78bbe83d1d8f97 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa8ad3e0b5946b1a549a6baf07db754 : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db697d55310b4c9dbbfd0baee195ec2c : SUCCESS in 18m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/04526f3673d34d49b68e95fb09a308ae : SUCCESS in 53m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12e74c39aaa343cda55882e836dd0bed : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b86e584dc63c78eebb45d4eb8d8c4b803df5e405","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:54:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80ee99547dffa9a5d92b78091ac1366ff1542161","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:55:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccd78e612f9470a8ae9489d14e425a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e87badbdb5404681c5b01c8042e8b2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b222ce035a5be5ec037314fc838025ae4fe5d30","revisions":{"2b222ce035a5be5ec037314fc838025ae4fe5d30":{"kind":"REWORK","_number":1,"created":"2022-11-30 22:55:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/866229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/866229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/866229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/866229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/866229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/866229/1"}}},"commit":{"parents":[{"commit":"f3edd4d6706c97da9c954edcac6ef976366a073e","subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3edd4d6706c97da9c954edcac6ef976366a073e"}]}],"author":{"name":"Jakub Neumann","email":"jneumann@cloudferro.com","date":"2022-11-24 09:29:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-11-30 22:55:00.000000000","tz":180},"subject":"Fix plugin builds with sources using type\u003dgit","message":"Fix plugin builds with sources using type\u003dgit\n\nA recent change to git [1] introduced a new behaviour to work around a\nCVE [2] that disallows any git operations in directories not owned by\nthe current user.\n\nA fix was introduced for general checkouts, but it was not applied\nto the plugins archive, resulting in PBR still not working as intended.\n\nFixed conflict added in I093620679016b37e1664c9fe4cf7559433e744b7.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n[2] https://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2022-24765.\n\nCloses-Bug: #1969096\nRelated-Bug: #1968877\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n\nSigned-off-by: Jakub Neumann \u003cjneumann@cloudferro.com\u003e\nChange-Id: Ib3a37eebb29d975fc51a117cecdff74baafd8941\n(cherry picked from commit 6be0068f376b0ae67bc81b50a97e042a88317d28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b222ce035a5be5ec037314fc838025ae4fe5d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b222ce035a5be5ec037314fc838025ae4fe5d30"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
