)]}'
{"id":"openstack%2Fkolla~866230","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ib3a37eebb29d975fc51a117cecdff74baafd8941","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1969096","attention_set":{},"removed_from_attention_set":{"35536":{"account":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"last_update":"2022-12-01 11:01:36.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-12-01 11:01:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3a37eebb29d975fc51a117cecdff74baafd8941","subject":"Fix plugin builds with sources using type\u003dgit","status":"MERGED","created":"2022-11-30 22:56:08.000000000","updated":"2022-12-01 11:02:31.000000000","submitted":"2022-12-01 11:01:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866230-bug/1969096","meta_rev_id":"607a7a90d5d9f49a0cd63ac5d2d5aacadda00030","_number":866230,"virtual_id_number":866230,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 11:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-01 08:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-01 09:22:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-01 09:22:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-30 22:56:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35536,"name":"Jakub Neumann","display_name":"Jakub Neumann (Cloudferro)","email":"jakub.neumann@gmail.com","username":"jneumann"},"state":"REVIEWER"},{"updated":"2022-12-01 00:29:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 08:43:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-01 09:22:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"57fba7cfbf5b465ccddc586d5c8aa785c240cbc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-11-30 22:56:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e11664b6fc9ef039a7ec8581d63d58e145610f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 00:29:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ff566222e74c91866f13d370a2674d\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/54353ca9a5a84d81afd6a734579f84c8 : SUCCESS in 24m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/78976d854369488e824da62f3d47bfa1 : SUCCESS in 32m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2c2f7f54151a4c8e8ff09ee4b70ee5e1 : SUCCESS in 1h 11m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd1cfcf5c5e14ccca4a29e8e38647dc8 : SUCCESS in 53m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/321d8007e6ee4c779ad3a7d1a348286e : SUCCESS in 1h 29m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/08b6893b590e49dab56f9bb0880eb12e : SUCCESS in 16m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/df6ef9b1611a41db8a4d311ac3342e1e : SUCCESS in 21m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/35e7401adc7d473ebc1830081049637a : SUCCESS in 1h 10m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73825da70654b12be6a02c3344f652d : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/541c5ac7242848d9b98aaab73722dc7e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe203596910f4fc5b2b8cd23e61eb5b6 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a71d5a87decf483b9eb9d3c47ca0344b : POST_FAILURE in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e3a2cfdce140ceb62344b531303278 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/159358aeb69343a9b6c9137c71ecde1f : SUCCESS in 19m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11ca683cd3e94de4a8362cd28de11d54 : SUCCESS in 23m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/566636dace42428983be3b7c58bb5802 : SUCCESS in 54m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cde899f4e95a4b9c94a3a4621f6f7446 : SUCCESS in 44m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4bd43cc96e3c4b69afc2abab407c7d60 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"21a473085bc4618576085c81be8b715e0e665bed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 01:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee05defb360f4acaa1b68f45aadd7db8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5749b923af6146578295f058b57b9dbc : SUCCESS in 2h 22m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/710cad3dc3e14cedad968a645d4c9d36 : SUCCESS in 1h 57m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77a0e724a95c40119de7328a2ac10504 : SUCCESS in 2h 23m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b7a46064ee6d4873b1bf08cc8d383d8d : SUCCESS in 1h 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76cb9dc7e34f95857576fe3b6d6a09abc80e0f6d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 08:43:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da61fb7f025431def74e325ebb059b5c287f9be0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-01 09:22:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d107b3afc02549bf07117d3632e10d4e35e49e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 09:22:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"601f7d9b0d8c7e79ad376c6a1919bd85b823f9ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:01:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0e04e0a1bd4575a28956feb394db0a\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/940005980c46466d878720d77e23eab9 : SUCCESS in 26m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6da5e5614204412fb4307d626a66df28 : SUCCESS in 57m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/712e6a4c49974713ba095ca9d7109a5d : SUCCESS in 1h 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f2656716a34ea6b9397fb864e1a06b : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa851a82b4741ee972547b7d3cfa977 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9332269d46f47b2b4b44442c86a5bdc : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d727bfb3ea04686959ba1af4f1b2cef : SUCCESS in 6m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/12a3103dd6ee45eca22bcc8953ffc6f2 : SUCCESS in 14m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a2eaec92f614c0cae160ce3cf701135 : SUCCESS in 1h 04m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31c40f16b6be49938426219dfd309da2 : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d8a527fee615fdbc59aa2a54afa006418c938eaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:01:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"607a7a90d5d9f49a0cd63ac5d2d5aacadda00030","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:02:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a85a7b627c042179fe7b41ac0892975\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d33dd771ad444fa0155eabe90fa627 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8c92dfa1404a647587d948f6c73091005bfbc2a","revisions":{"e8c92dfa1404a647587d948f6c73091005bfbc2a":{"kind":"REWORK","_number":1,"created":"2022-11-30 22:56:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/30/866230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/866230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/866230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/866230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/866230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/866230/1"}}},"commit":{"parents":[{"commit":"d67bf58de385d24cd5ea2b40966e76d99bfd1938","subject":"rabbitmq: update to RabbitMQ 3.9 on Erlang 25","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67bf58de385d24cd5ea2b40966e76d99bfd1938"}]}],"author":{"name":"Jakub Neumann","email":"jneumann@cloudferro.com","date":"2022-11-24 09:29:47.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-11-30 22:55:43.000000000","tz":180},"subject":"Fix plugin builds with sources using type\u003dgit","message":"Fix plugin builds with sources using type\u003dgit\n\nA recent change to git [1] introduced a new behaviour to work around a\nCVE [2] that disallows any git operations in directories not owned by\nthe current user.\n\nA fix was introduced for general checkouts, but it was not applied\nto the plugins archive, resulting in PBR still not working as intended.\n\nFixed conflict added in I093620679016b37e1664c9fe4cf7559433e744b7.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n[2] https://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2022-24765.\n\nCloses-Bug: #1969096\nRelated-Bug: #1968877\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n\nSigned-off-by: Jakub Neumann \u003cjneumann@cloudferro.com\u003e\nChange-Id: Ib3a37eebb29d975fc51a117cecdff74baafd8941\n(cherry picked from commit 6be0068f376b0ae67bc81b50a97e042a88317d28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8c92dfa1404a647587d948f6c73091005bfbc2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8c92dfa1404a647587d948f6c73091005bfbc2a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
