)]}'
{"id":"openstack%2Fkolla~866304","triplet_id":"openstack%2Fkolla~master~Iff646c87a5c8d8d1831fea00ae18457f93cca2e5","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-12-02 14:24:23.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-02 14:24:23.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-12-02 13:09:46.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}},"hashtags":[],"change_id":"Iff646c87a5c8d8d1831fea00ae18457f93cca2e5","subject":"telegraf: mark buildable","status":"MERGED","created":"2022-12-01 08:42:13.000000000","updated":"2022-12-02 14:25:39.000000000","submitted":"2022-12-02 14:24:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866304","meta_rev_id":"891bfedae5655089888966e5e99f10bcdafea7d1","_number":866304,"virtual_id_number":866304,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-02 14:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-02 13:09:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-02 13:09:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-01 10:32:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-01 10:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-01 10:32:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-12-01 10:32:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"3b9e390235415500a51ae55481e8855b9438825c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 08:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cadb5f042a0e06888e0bc0c6c7609537436e468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 10:05:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c89c39341c445db012576ca477a097\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b96aa71543c2414aac43ac3e5ee26007 : SUCCESS in 28m 39s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/52487a343bfc4ff1b54110d000424067 : SUCCESS in 1h 17m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1d98f809665491f91649c70ee18202c : SUCCESS in 27m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e223d696b624764b4707800b5c43ecd : SUCCESS in 1h 10m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d662bda911254270ac2731bd4c766d1b : SUCCESS in 1h 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed04f17a7834b44a14e686b99c6120c : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9554b18f119e4f3999d2580f6b5d3380 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ae2f16a2c641c295894d50ecff48e9 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8232f4d262634891b5f16973d62b1b02 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd21c26d3e44d53ad6801e2beb3d50e : SUCCESS in 6m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7077dcc2eb8c42fd881f759163d9e3cb : SUCCESS in 24m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/b3727ab355db496f99c668648f8b629d : SUCCESS in 1h 17m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8224acde6fe43edb4571e17a7cb3ea3 : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcae628d369643ff82d5c54494c060a3 : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73654b8847824c6cbd416b01925f9a54 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4d2402f8a3e27de76ec209aa88b08d7e5a701b3d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-01 10:32:03.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b2b970d7747634f2b94fb0a5dfe46b24a056f466","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 11:41:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fdfe16164c54b48b40531693fd8b0d9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/295820b82d8647e085b6e3647589a8c5 : SUCCESS in 2h 16m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ae4f8a9e5664d32a4a16cc7d5ef1a65 : SUCCESS in 2h 18m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e2a27188c4247aeab51efb1c2e7a6bf : SUCCESS in 2h 56m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97d0cbc37a4c459ca057b705c5c3bd50 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46deb53aa9654a0faf71d8a07d9df5c0 : SUCCESS in 5m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b0bac0170214455a885d4b006a8e8413 : SUCCESS in 2h 13m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2b5b1a1e1e444157a130872969bdd28d : SUCCESS in 1h 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23a3a336a71e5479a0787aeb051dd7e51d1cc61e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-02 13:09:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73c989b804d4a7fe755c5db7e7526425a612f500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 13:10:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed57d5691b5f7bae4760e07794d5c3084c1a37e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:24:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121674c24198400b8b32f7d80ddf3cde\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/70c6aa328d9e4b3dbc791c49e7fc9f6e : SUCCESS in 24m 45s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6335f5623117449eb2a2a9291feafff6 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fa8e27679e124a7a9971c6f71de9dd4f : SUCCESS in 21m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/afb2830493c74f078b3673b706f3c9d0 : SUCCESS in 53m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3e21221c6566467686e55b07c9d8ed92 : SUCCESS in 1h 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527ada2fec584bb48dc159804cb0cffe : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9800d54de5b54e5fb653d71e672e1bde : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f58fd96981284810baaaf06e1eafd3b0 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1c023bb644474cbc75f52c505e49c7 : SUCCESS in 4m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3b9dba382e12459888ee8257a17444ec : SUCCESS in 22m 58s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ec69dcfaf826493e80e185f8404bbdec : SUCCESS in 1h 03m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29b14384bd7a456ab344e51752e81e75 : SUCCESS in 28m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5848c23878a245c9bf473e45ff659be6 : SUCCESS in 40m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18965ccfde8e4dd89c3d7d3f257fac74 : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4dc7e4da9b23f9f534149a8435302cb717a394d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:24:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"891bfedae5655089888966e5e99f10bcdafea7d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:25:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb75acc21314f9a8599fb9de5997ba8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6421e6ca5dba458b8360146dee528c1a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1d26c867493800b83827e991483c447b89ff87a","revisions":{"d1d26c867493800b83827e991483c447b89ff87a":{"kind":"REWORK","_number":1,"created":"2022-12-01 08:42:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/866304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/866304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/866304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/866304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/866304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/866304/1"}}},"commit":{"parents":[{"commit":"6be0068f376b0ae67bc81b50a97e042a88317d28","subject":"Fix plugin builds with sources using type\u003dgit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be0068f376b0ae67bc81b50a97e042a88317d28"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-01 08:41:57.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-01 08:42:11.000000000","tz":60},"subject":"telegraf: mark buildable","message":"telegraf: mark buildable\n\nChange-Id: Iff646c87a5c8d8d1831fea00ae18457f93cca2e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1d26c867493800b83827e991483c447b89ff87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1d26c867493800b83827e991483c447b89ff87a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
