)]}'
{"id":"openstack%2Fkolla~866674","triplet_id":"openstack%2Fkolla~master~I886da6215c20a3d0227b4ee5d603a7e416929618","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-06 12:20:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I886da6215c20a3d0227b4ee5d603a7e416929618","subject":"opensearch: fix footer name","status":"MERGED","created":"2022-12-06 08:44:08.000000000","updated":"2022-12-06 12:21:21.000000000","submitted":"2022-12-06 12:20:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866674","meta_rev_id":"ca7748014bb6d4461c01b7214ae54af591062ef7","_number":866674,"virtual_id_number":866674,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-06 12:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-06 11:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-06 12:00:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-06 11:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 10:01:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 12:00:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"07409992b553fbefef58313a1603ed52cef6b793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 08:44:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a05e4d97a75cf8e66ece4f08619485139503ad59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 10:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183f94eb94d54bc6981e703480296420\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1047622247074455a2fece6b52e02bd7 : SUCCESS in 24m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/45e731e479b0476382d5afa3c5c7dac1 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/35fd9e036c3d4dbdb573287e53073987 : SUCCESS in 19m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d166665a4d1e471e8f15c2cd568e918e : SUCCESS in 1h 06m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c22ad03e8069466bbb17fa966cde981b : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964fb7d57f5547e38390dabec8ee6fb1 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/720645c82b4f4939954f285362151e61 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54b2cbb03aee4255a6600c0c5b830b13 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9df6b0d9a4074f5fa8a2370a430022dd : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863c1b350d4c44fc9eb874bcc5c1fdc1 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dde8da2c89f943709cfb516a6039ff49 : SUCCESS in 32m 44s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3c2a3c6825784246bb017ebbcceb7de2 : SUCCESS in 1h 00m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b1ed65d9e5964a359d88f166cb3a3299 : SUCCESS in 16m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/38e191db79694b20b0c53aeda21f4b7b : SUCCESS in 44m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c320f334c7d4c4fa17c96e912cdc5a5 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":1},{"id":"371709157c7eedb8f123d42d05745cbcf31744e0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 11:07:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18bfa1a69384ebef08bb22dc98098fe5f1f336e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 11:07:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5743a3671aee9759406cec530f27513043bceb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 11:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18574622d452484cac056413149f431c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0c37dd3246a0433b8d8988747b0e1996 : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b60f06facd564f8d8ce99c5f7902750b : SUCCESS in 2h 40m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c69a9b672e74599be84fda443c65d72 : SUCCESS in 2h 14m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ea8bd98af42441599a61c511bd633c1 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2af6a38ceafb4edc8904997d3088f260 : SUCCESS in 4m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/45090ca072c443daaa77fbb7672c4551 : SUCCESS in 1h 44m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5e7e16395eb146e5b1ca52037822ff66 : SUCCESS in 1h 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b6921817e3161a830d68d8494295bc2ae294967","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-06 12:00:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be564e1244920cfbac932bb5d1269d58bfb5917c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 12:20:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43624902a1fb469c8169064c1ebee859\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ab87929445b14840a678d0d1bab60a29 : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/81d4db3b75a84baab08a93627600c85d : SUCCESS in 1h 01m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/019c6a3d0c8640999a9849882df08f05 : SUCCESS in 18m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2d1a9b41eebd44cd9121f83a0327fc09 : SUCCESS in 53m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0927b80421414d57991f814da191f844 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbad190ad50b43a8a977aa44ed3c22c8 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b41dde47f6024b30b3699db8dc956519 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bff307890a4849a49fa2c7b15e2a832d : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a6b1e205674d589544023c1b94e96a : SUCCESS in 4m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0ea366f5a30c40fbb7c4d7338834d815 : SUCCESS in 28m 40s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/be515512d0144920bc0983048316282d : SUCCESS in 1h 12m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2aecbb516ed04cb5817b58011b06d042 : SUCCESS in 20m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/751af61bf015467c8e56708d6d1f49bb : SUCCESS in 44m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec0e369e3dc94ce3bc74cb70219b7fb9 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4d52215e832dfd622fd3edf5a4b91461fa3d1bbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 12:20:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca7748014bb6d4461c01b7214ae54af591062ef7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 12:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20dc35810f67439abfe97d540f7f61d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b024a91f06834e96b95049d6406cef47 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf1ef50d87b874d8cb4621e3199a5346a1753907","revisions":{"bf1ef50d87b874d8cb4621e3199a5346a1753907":{"kind":"REWORK","_number":1,"created":"2022-12-06 08:44:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/74/866674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/866674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/866674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/866674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/866674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/866674/1"}}},"commit":{"parents":[{"commit":"2a4c3092f43fd79ba9c22947fb5c690c629296ad","subject":"Merge \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a4c3092f43fd79ba9c22947fb5c690c629296ad"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 08:43:54.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 08:44:05.000000000","tz":60},"subject":"opensearch: fix footer name","message":"opensearch: fix footer name\n\nChange-Id: I886da6215c20a3d0227b4ee5d603a7e416929618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf1ef50d87b874d8cb4621e3199a5346a1753907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf1ef50d87b874d8cb4621e3199a5346a1753907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
