)]}'
{"id":"openstack%2Fkolla~866707","triplet_id":"openstack%2Fkolla~master~I6f32aacb4e0b7dd1eb7206e02425b16f5beae4f0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-12-06 20:38:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-07 10:47:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-12-06 16:12:08.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}},"hashtags":[],"change_id":"I6f32aacb4e0b7dd1eb7206e02425b16f5beae4f0","subject":"Drop elasticsearch, kibana and logstash","status":"MERGED","created":"2022-12-06 13:11:03.000000000","updated":"2022-12-07 13:31:27.000000000","submitted":"2022-12-07 13:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"866707","meta_rev_id":"7fede9580c98c2b6449752c501c970da491d356b","_number":866707,"virtual_id_number":866707,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 13:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-06 20:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-07 08:53:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-07 08:53:09.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-12-06 16:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-06 13:11:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-12-06 13:11:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-12-06 14:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d2e098561022f8a0b978e9c34306e74c9edb6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 13:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e34987440358d3a1520a9e9e36f13a9833193a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 13:11:41.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"1537d7f993ae694654f07a87de67aa2bfc90ffc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:30:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06087faf001b4d8c9fafe7c103e0c462\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6c3bef2cb71c4db481639ed5220dbc3e : SUCCESS in 25m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b04451a3f1e245ad9f125d6a08fa66cc : SUCCESS in 1h 17m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d0785764656c4d0fb9118b7ed47afacd : SUCCESS in 18m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a2d6eae0fe47447f91ac01a63bcac4e2 : SUCCESS in 1h 08m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1b5248586b73444e87178631d929b850 : SUCCESS in 1h 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0697ec7a54324bd7b88edb43c501115a : FAILURE in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2c5c8a40dbc4377a8abca3313644c44 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70730a60b012401296bd49d505ad0717 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79cfd690f6544cd5a115246c5a46716b : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d601f44f10e34bae943680634028c6f7 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e34ccff111941bfa05acdc4ec75b5ee : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/41a7c6806bed4bb2b41ad625e51ea746 : SUCCESS in 25m 59s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/916c40e02d3346c5ba9ded39023485e0 : SUCCESS in 1h 11m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cf4054ce54944a26a55c4df62cb9b58e : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf3ec687b4be43438a4c4b25c503fa91 : SUCCESS in 47m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c97e860e44a4c4bb76f8a487aee2f8b : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"19982b21917e5a866e879b1be110238f825756f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 14:58:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"edce9ee587e1fb5a3fa0ede0255195813165e603","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-06 16:12:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"19016805b5dab4a216f90fd052433f45860c58f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 16:33:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90a05ca2431942e8b4e92c681dfa5fb5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/59111ee819964f51b1f89ad2e92774b5 : SUCCESS in 36m 49s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4644f8d6b3e7450881903a2802605064 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/183a84b241af4418a78d68b6ef82497a : FAILURE in 17m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e2c7e1fe618474bb8f4adbd0936f21d : SUCCESS in 1h 03m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/48ff05ab43414d9c96e8226f83eb8a02 : SUCCESS in 1h 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e272708aadf49188ea596bf643a5fac : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51aa7e852bd346fc852a753a4fd0ba68 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e78962bcdb44c8d901bb7f46b07e690 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c7fb6f9d4014ddaa31daf7a9de873fd : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960ee6bbfa634468a6cffda7e323179a : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3422b4f548346b18019126f43a09765 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e667e27bdc434471b8446577d945ff70 : SUCCESS in 30m 54s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/eded9dbf62134f3d94c81c8512e6e9dc : SUCCESS in 1h 09m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/daf8bddd2b134a19a77a16501763bc18 : FAILURE in 34m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af3ec510a07c43809d6d72f8a67bdab5 : SUCCESS in 47m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/071686b687c74ac3b9234ea097095212 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ddb0c51cd29a340a9757b7398e4c9e8d3df830ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-06 16:34:55.000000000","message":"Patch Set 2: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c9602b2444c1051915638c07f1a883a84dbeab52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 17:27:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b96040ad3f5848e6b28783e91ecb29c8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d9c586ce2eb84fe0b4ee2b29b58d2baf : SUCCESS in 1h 48m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4cba4130108c44718161e6ce7d6c11df : SUCCESS in 1h 36m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1297437b574b432982b1e9aa7944b5f3 : SUCCESS in 2h 26m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b38d1027675e42e2b3822d6175ea2d6e : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3a4b8eac5cb490aa39c0d117b7e64f1 : SUCCESS in 7m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c4c64c80bd7140a6b16cace97b5f7d48 : SUCCESS in 1h 35m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db80386f19a94750b9c5e00fe4bfb4dc : SUCCESS in 1h 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be339dcb533ee4271237caa35b096c9328f3aa8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 17:52:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6daa15a788af4bbba47e47cb379f0489\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d31e0b3e6f7241d0a7d0324b7e38a99f : SUCCESS in 31m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2e7c3c18e9804f6188360479efc32ae4 : SUCCESS in 1h 12m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3729ff2d0c745fa93cf8ee795e7a0fb : SUCCESS in 22m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe136fcd1a614fdd818a3682b46bb4d1 : SUCCESS in 1h 10m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/64506a6a3e0e4b338405d693571bc953 : SUCCESS in 1h 08m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bccecd565ff46eba26cf4859ba8511b : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72641f0fb43f4ee0aff270b4195559ad : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bff1b3fec638468c80f1c54975158f49 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54f62febc3c1422897f7026a545c037e : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930dca6e63b34e2bbfa32599769094c8 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f47c94c4396046e29652181f65da6735 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d3199d9075ac4569b9843f68833a512f : SUCCESS in 20m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1803c5d187f84b438e1893e2242b5583 : SUCCESS in 1h 17m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/12b0311d765846628598db3c597900f1 : SUCCESS in 31m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea504f051d0e452cb9e37bdade8fbb51 : SUCCESS in 45m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00d3ac489e4846b684de481d9d734fee : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e953385692741bb1ce9abab7026db3ac7eba5eac","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-12-06 20:38:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cab9b00d1eb3bd677639173174cf4d7b8835181e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-07 08:53:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee808312da7786d9bbdfc33904ea3ac252d4e121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 08:53:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"20c31d430e81a8655579051dccaca451b9767789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:33:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f28cb9a085147bca4d10d44b37588a8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c04ffea23a2c413a89f974f5f534c07b : SUCCESS in 23m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/dc76f8a5aa034159b49cb30518067183 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d35c0c14053f45079e40b11b9dfd66ab : SUCCESS in 14m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2ad34262f1864fc9a16ae6c3bfff1ff2 : SUCCESS in 50m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/831724ce6b7c43e0ac006e938e0fc6c2 : SUCCESS in 1h 00m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f8d7e1a9fe48df90e036b5140cda89 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cceb321a90ec4cd59c92b1d1e0226b92 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8e166d637444dfe8aba96332ef5595c : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be94f9ee634e4bceb75c546113bb3e04 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06cd3afa01d4045ae2ea8c229825dd0 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/91a109b712674c8b95cd122789888c39 : SUCCESS in 31m 11s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ac1a88eac45a4f119bf28b0decb4d1be : POST_FAILURE in 1h 38m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cc575448107e4961b78c78d39f7cddcf : SUCCESS in 20m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e640c62ae66b46edabaede6f889ecfb7 : SUCCESS in 49m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/934319a69704470d9e5dba6f04ee597e : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":2},{"id":"cb153b765a89a2ca7befc4dbafd4d4d2e23be666","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-07 10:47:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"deac06ad5731881b83ce65a412f37d9a58aadc3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 12:04:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe514d9687a406c99a8b01aa42b2e81\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5aa0e0384cd5419c9486fbbc48797f8f : SUCCESS in 22m 47s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a354071ac4bf4613b9555853223f4504 : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/511d6bf606bf425b89d4e4d807896cd5 : SUCCESS in 20m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6483d87c016e464d904ac22762be9df2 : SUCCESS in 1h 06m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8e005c302c9b41e69c1e065a1fc9e084 : SUCCESS in 1h 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55004233fc824590ad1e1b336aae078b : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/545a1b95ca7f4c3692b0063222fce01c : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c45cadc7486e4806a35c1b45890acde6 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fafb06710174b6892317ae959bc8a12 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbbc7e1b31943ef819b82e3878d8ba6 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab671cc02634be998f272b9d6391c6a : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ee40d453ecb74f8cb21f407b0bcfefee : SUCCESS in 22m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/180bf2aeb114482481adc07e640fc051 : SUCCESS in 1h 00m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/10160508380f43b0985e7df65f1c75d2 : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/473f61eebaed44fdb76627373a29ab5a : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aaf5323f24884a978e80520d47e40f18 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ca8b672483d1cd1f5f9d276740ec457176a3427a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 12:04:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e369015c91835dc8c2d7559a43ac94e29e2363b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 12:53:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47eed8eee5be4063a1736fcf2f5dadba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/98f82e83c2af402b8a721d91fc2b2179 : FAILURE in 6m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/21c14a87d6844dda9f65256afc8a636e : SUCCESS in 53m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/950fcb2874f34898bbc316322829695d : SUCCESS in 2h 03m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63f1184de7e04b7cacd124ef7b4be224 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d08ba46e4440420790f0f81cfef2ae88 : SUCCESS in 4m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9a8e5de56b244041adb4b96d4fc58a9f : SUCCESS in 51m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c00181ec63474d7da0b7956e5ce543df : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e3035733927d8fb088512a4df16d3b60c091647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:30:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05b409abfb1744cd9b9e7ed2cbb8d475\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/39c26c993770447bacc695e2c47a5d34 : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/94b2b50373d44d8eacea87ddd77c7a48 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/926afa5bb4964d718d1abfa428d87bbe : SUCCESS in 18m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9fe7ec987e73428cb46407f50a241913 : SUCCESS in 53m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7509f9e9766e461ab61f16bfdbdd2b8d : SUCCESS in 1h 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9a64ffd1d64b80aceddadf6ff363d4 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfd3880a796643a9b395c7e4a88d01f4 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55916441bb44a81a1a28dbecc5d0c75 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f9db91234e4efa802199c3717b5887 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bd140e62c914d27ba76eb24f9dd1375 : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5a58e7694c5d49efafeea30d7c004085 : SUCCESS in 24m 49s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/6ca1cc01072643b29457a3d9465a8d3f : SUCCESS in 59m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/87c86965b60e471b80a79fd993236608 : SUCCESS in 17m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e036b6d0a924515872386b189f9def4 : SUCCESS in 52m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ffecd3f5ac9f4f40a8cbe346110ec70c : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":2},{"id":"58f83da90986041f1a082ae0d2ef2f72654f52d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:30:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7fede9580c98c2b6449752c501c970da491d356b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:31:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5623255f7c448a19e16141aeeae0ac4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8bf0dc6ec44db693ec6122e93e7450 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0fed98e64a340499c79ac56647bd14d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e221ac8f2704a21461f9b36134816329a43272cc","revisions":{"1dd1480a21fbdb3ff6ff6634873ddcf442bebde8":{"kind":"REWORK","_number":1,"created":"2022-12-06 13:11:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/866707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/866707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/866707/1"}}},"commit":{"parents":[{"commit":"bf1ef50d87b874d8cb4621e3199a5346a1753907","subject":"opensearch: fix footer name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf1ef50d87b874d8cb4621e3199a5346a1753907"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 13:06:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 13:11:01.000000000","tz":60},"subject":"Drop elasticsearch, kibana and logstash","message":"Drop elasticsearch, kibana and logstash\n\nSupport has been dropped in kolla-ansible, the upgrade path is opensearch.\n\nChange-Id: I6f32aacb4e0b7dd1eb7206e02425b16f5beae4f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd1480a21fbdb3ff6ff6634873ddcf442bebde8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dd1480a21fbdb3ff6ff6634873ddcf442bebde8"}]},"branch":"refs/heads/master"},"e221ac8f2704a21461f9b36134816329a43272cc":{"kind":"REWORK","_number":2,"created":"2022-12-06 14:58:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/866707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/866707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/866707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/866707/2"}}},"commit":{"parents":[{"commit":"bf1ef50d87b874d8cb4621e3199a5346a1753907","subject":"opensearch: fix footer name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf1ef50d87b874d8cb4621e3199a5346a1753907"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 13:06:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-06 14:57:51.000000000","tz":60},"subject":"Drop elasticsearch, kibana and logstash","message":"Drop elasticsearch, kibana and logstash\n\nSupport has been dropped in kolla-ansible, the upgrade path is opensearch.\n\nChange-Id: I6f32aacb4e0b7dd1eb7206e02425b16f5beae4f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e221ac8f2704a21461f9b36134816329a43272cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e221ac8f2704a21461f9b36134816329a43272cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
