)]}'
{"id":"openstack%2Fkolla~866852","triplet_id":"openstack%2Fkolla~master~I885cee9ad92e7a78b755caf590ccbb8adb2dd070","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-08 06:08:06.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I885cee9ad92e7a78b755caf590ccbb8adb2dd070","subject":"CI: cleanup kolla_python_version and remove centos from gating","status":"MERGED","created":"2022-12-07 13:11:37.000000000","updated":"2022-12-08 08:35:59.000000000","submitted":"2022-12-08 08:34:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866852","meta_rev_id":"d0f07fccaded4587d6f93180e7c2fa64f6bc11ec","_number":866852,"virtual_id_number":866852,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-08 08:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-07 17:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-07 13:56:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-07 17:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 13:11:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-07 13:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 13:56:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"9bb21a5835a7c56adb2da9248a6174eee66c2549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-07 13:11:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe957ec0ffb87c3070bd5be8b7cdc13d973f3652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027kolla-ansible-centos9s-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos9s\n          - kolla-ansible-centos9s-source:\n      check-arm64:\n        jobs:\n          - kolla-build-centos9s-aarch64\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos9s\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c6b0ab2b2913885cbe5854e708bd62f9eb4119ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:12:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027kolla-ansible-centos9s-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos9s\n          - kolla-ansible-centos9s-source:\n      check-arm64:\n        jobs:\n          - kolla-build-centos9s-aarch64\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos9s\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6b6de8213aa79f07ec1aef0a29cf3bf0fddaf374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-07 13:14:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"713adf41758e91b231f7bbe01942d7362f326261","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:22:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444933d887ab48fda408c9e4fb9b1a74\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3db63dbf61e44ba49a9e832fedb5f60e : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4f05584904f4b35bb21dc4d04389395 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38f3c2404c80acf4567fd3e485141215cb66b6c3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-07 13:56:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"42d211c6a31e79714087e17a3342d3187329e439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:34:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee94da41e184fb5a629459376880aca\n\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b900261241c7433da2c153159b5e51da : SUCCESS in 1h 17m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05520bfd523a4ee389002c5a7fb117bf : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fbe6807f72e4642b599ae2f2fe6ff6a : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e3a8a5e963649ad9efe5ea17361223f : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef701ec5a8134c65bd45d18504f614f0 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07af7a7dd084eb3b33dacc20a86c2a1 : SUCCESS in 5m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/43821563675443b29f2abae24cedcaa4 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f728351bc30c5eeee4fb0cb7bffe2cfdca6c6ac0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-07 17:30:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e94207fc1e48a60d8d76264278cfca2077788ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 17:30:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff07ee84bbea87c8433d2bacde8c7a8650d92902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 18:41:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b04f45b89f54bc1b0dc10c18ce5e661\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538df7f8003643719c6ad2846af22ebc : FAILURE in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/659aa195e5384b8bbf8f3a2fe51e7a09 : FAILURE in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce7ad84b021b4742b64b2adc1d7578f1 : FAILURE in 2m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc2b3b30ec440469093ef61291fd832 : FAILURE in 3m 25s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/2ce4f5ceed7244908385857f256812e5 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7e03a53727e9d0ea51e11b0b5c769eccd301b7d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-08 06:08:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b14513191c6df8be2a7a8fc988f4553ca487fac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 06:15:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9e5d619f76414c8f4653eba09a0360\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e9c60edf5564dafa5d21afcfe7fbdcc : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31a448183033408885722bf15573672b : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"beda5f967cb92b18664d50c6a2e54690c70a6460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 07:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a456ebe57d6f40248578ab3b4d5aa0e5\n\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/38b5a3ffd73f4e9c95907929b1ce0cea : SUCCESS in 1h 09m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92bae59642a54735affa998807c3ed8a : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af5b85b9e1c745bf80fd968afc691af6 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56ed23cbdf4d4ec194205162aca3cc1b : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b676bf0b4c434216887a65d8a7606219 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020d180fe49a4494afa3937dbcd4cb46 : SUCCESS in 4m 07s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/1ff7234bcda04afcb59017a7c2da8768 : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d8db515f049b83dd2c86dc21b8eddb56544f2e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 07:27:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb3a55c9edddfa127b49d2b78174cf9f02c5f207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 08:34:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20dc57cb2b94a5dba3f8c6e6143f440\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7bbd7f849d4b3a9a158150d585af51 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f016d015da4a4411b130d5c76f199576 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbf0c1215f9846c2b783adafddd7a672 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f599c133eb45109c248974b75c7d8f : SUCCESS in 4m 29s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/5898b2439ddb48e9af50d27ed9ffb57b : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":2},{"id":"96cb047bc2b476efcd23f0d1e224489797a8ee93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 08:34:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d0f07fccaded4587d6f93180e7c2fa64f6bc11ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 08:35:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e81a5ba4f781407dbdb505d65dd6b931\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2146b2478af44a548baaf73741a9fd36 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c878de004c6c3c7016dfd376a371d9d04c570b4","revisions":{"f173c6fb08b66a178b9ac6f40ceb8bf6d7bf50b7":{"kind":"REWORK","_number":1,"created":"2022-12-07 13:11:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/866852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/866852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/866852/1"}}},"commit":{"parents":[{"commit":"74e0d0169fb6eef02877e83998578a8e7d64ce71","subject":"Merge \"mariadb: update clustercheck branch to match db version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74e0d0169fb6eef02877e83998578a8e7d64ce71"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-07 13:11:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-07 13:11:35.000000000","tz":60},"subject":"CI: cleanup kolla_python_version and remove centos from gating","message":"CI: cleanup kolla_python_version and remove centos from gating\n\nChange-Id: I885cee9ad92e7a78b755caf590ccbb8adb2dd070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f173c6fb08b66a178b9ac6f40ceb8bf6d7bf50b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f173c6fb08b66a178b9ac6f40ceb8bf6d7bf50b7"}]},"branch":"refs/heads/master"},"3c878de004c6c3c7016dfd376a371d9d04c570b4":{"kind":"REWORK","_number":2,"created":"2022-12-07 13:14:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/866852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/866852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/866852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/866852/2"}}},"commit":{"parents":[{"commit":"74e0d0169fb6eef02877e83998578a8e7d64ce71","subject":"Merge \"mariadb: update clustercheck branch to match db version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74e0d0169fb6eef02877e83998578a8e7d64ce71"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-07 13:11:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-07 13:14:41.000000000","tz":60},"subject":"CI: cleanup kolla_python_version and remove centos from gating","message":"CI: cleanup kolla_python_version and remove centos from gating\n\nChange-Id: I885cee9ad92e7a78b755caf590ccbb8adb2dd070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c878de004c6c3c7016dfd376a371d9d04c570b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c878de004c6c3c7016dfd376a371d9d04c570b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
