)]}'
{"id":"openstack%2Fkolla~867800","triplet_id":"openstack%2Fkolla~master~Iad83132b61efadbf09aa9aa2edf96235085764c6","project":"openstack/kolla","branch":"master","topic":"bug/1999778","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-16 15:55:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-16 17:29:24.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-12-21 14:53:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32814":{"account":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"last_update":"2023-01-16 17:29:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad83132b61efadbf09aa9aa2edf96235085764c6","subject":"relay non-zero exit codes when creating openvswitch bridges","status":"MERGED","created":"2022-12-15 13:00:41.000000000","updated":"2023-01-16 18:45:06.000000000","submitted":"2023-01-16 17:29:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867800-bug/1999778","meta_rev_id":"37cce8d73e21911ffbd93fe21cb9186a7fb8e210","_number":867800,"virtual_id_number":867800,"owner":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 17:29:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-16 15:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-16 15:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-16 15:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 14:18:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 10:44:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-12-29 07:46:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-10 14:19:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-01-10 14:19:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"c6a7365fe8a09826a52c72478516501621994932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-15 13:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2c46bc89bd76df67883d6d32ab3b2ff6ce0dceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 14:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0017ae3a5b5045a68ab7d9070070724e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd2e6d39c2d24c50b840eca1fe3cdce4 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/41c93ed20e3448f88728e62f941876e1 : SUCCESS in 1h 16m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/927d3e03b2fb4966bd7c2f76c2923092 : SUCCESS in 25m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f08e6651a6f64a71adb111c548ac126b : SUCCESS in 59m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/897f9fb0de10479b9f7639dd778acad4 : SUCCESS in 1h 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05173c1b6964d85ba2efa541422243d : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1015a3201fa4cedad7b50121c7dfd9c : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db9d5f497a7941ef9068fd33cbe5d600 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82a6d64c89ac41c0a66a11bd0c4b9188 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a94a9cf21a64d7d914e647831b539de : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c156a6e3dca4f4692c7e036d04f5726 : SUCCESS in 23m 16s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/da4d56113f8b49de8f605d02cace0d37 : SUCCESS in 1h 05m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d033a77af9ce425089d0833af067c519 : SUCCESS in 19m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bdab896252d4938bbc9a4cc747bfcc1 : SUCCESS in 47m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dab230621265474d9593bc3f29834024 : FAILURE in 43m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9644e6bde4fbe1c4d57ed2fbaa8fa243a7108ec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:04:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34c2a0cf4654d08849c9e1343f31f28\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6882a20ea4bf4abda934cc05163a2465 : FAILURE in 7m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3641a1c6c674a93b6c32e2a2a1cdcd8 : SUCCESS in 1h 47m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24c3c14b6c1a414c99a9e2a37a8e5fee : SUCCESS in 2h 48m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6529cdf6fe944344b406b51576dec4ac : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fdfbebcd5f5f413098e32728ed928ea0 : SUCCESS in 5m 34s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/63b710bfa84f407592ee79dd8840861f : SUCCESS in 1h 33m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8a22ace611a142a08af7af0539cacd3b : SUCCESS in 1h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"420b32e5bc5d8c614fad5b06f9877f946897809d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-16 10:44:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"df34316710258b96fa8acbfa8a61307d71fa0fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-19 09:19:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b6299cf0047d97dbb666662965d989e149c4eaf","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-19 09:46:09.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"72b98b5ffb8cea9ef1b509715496c58746317d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 10:50:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaea0449cd324bf98b53ec5e8712cb5a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f9489e6d0bb74166bb026ccb41efec5f : FAILURE in 4m 16s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a5c5dbeb58804e7c8df660e65270ca82 : SUCCESS in 1h 09m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/76e29d6263e14e38805071cceeda8748 : SUCCESS in 19m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f597564caaa41e29e7cf90ba3e4184a : SUCCESS in 55m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e2bbc40253e1483c89b98404a4b95268 : SUCCESS in 1h 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060929de392240b4985c21068838fde8 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4223b44cd6d4aec9d6e9c2ee88c8e8f : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6c5a4f4d91d43e1b188974da041ab83 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96f9a71b2d2747a5a1a52a9d48416dce : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e1a7eaa9e44beda91e0355a8e71e6a : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f858fc2876494082a078f93654b0caf0 : SUCCESS in 25m 45s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ff6eb79c4cf44b579786b01822891a93 : SUCCESS in 1h 12m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35a8759b096248e0aac3abfaa5ca22a6 : SUCCESS in 15m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/decfb180f3c146bbb28c3e5cd787961f : SUCCESS in 46m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1888e2d8ee2f494eac46afb9caec9abf : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":2},{"id":"86b17a32a1a13dc026a76819f101c7ab0a035ecc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 11:20:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf15abef08b4f8282e2510f153a8060\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/05a85547b9694340b954b55a7c8cf636 : FAILURE in 7m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afb2763634234b949801ac96e89b2ec5 : SUCCESS in 1h 06m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/074fbb53b52341e38ac6bcf11a2152ff : SUCCESS in 1h 59m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7387f2b1ba504636aab6824054c315fc : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5610aec3a65d4b5baa343f0f0b7311dc : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ffc6aa7226c740dfa53d8624b726e30a : SUCCESS in 1h 18m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6384077263ca46c8babdb603895b46dc : SUCCESS in 1h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"986ba32548e81716fc7add1e4efe5d7ee71eb373","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-20 09:59:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d07aa688f9d1a37662d39026f221c2be37e6450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-21 14:10:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7697c8e3d58c2840ee6030b6789f541d1887734","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-21 14:15:28.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"507a37144fe235c816beffa85e77b6f1f5a36914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-21 14:18:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3d04c59836cc947919ec77ef22ea9a7ce1e2d8f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-21 14:53:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"368434b58bedb52259619b39b7edf31d70a3b411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 15:36:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe363921adf641a2b4bb948540a543e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6a3162c76aa446fdbf5ca3ccf5df900c : FAILURE in 4m 01s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fb4b52d1d4fa40c2b9701483afc58083 : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8cfb906bf82647c0a1e960de8e8b9e23 : SUCCESS in 18m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a74a6d43c5684ab49b37ae72848b7f18 : SUCCESS in 1h 07m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5f689a11ba9949dd832771f11a0b3d0e : SUCCESS in 1h 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b6b3508ac9484eac7c457a51e3d0af : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc549fc2b3b54ae1b59d826e511c24b1 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ffcef9c0c5d4859a3b007b0bb19803a : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ae58f5425e4ec58a2c5674a5874490 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca55a1bd04f04c9a995ce346b0299e00 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f29751681c54316aa933681832c4d2d : SUCCESS in 2m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bb0308e28f404b2f88a94c6da403bcf7 : SUCCESS in 27m 03s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/37f4e284ae584a879592e0d1e663796a : SUCCESS in 1h 04m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/90c45e396c8c402ca90989f4282669fe : SUCCESS in 20m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9cc092c904f4e96b9e81b77c2d4640a : SUCCESS in 45m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/580c447e96c446b691e889865b8f14c6 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":4},{"id":"8239f4fe76723f9e3efd992de92dfb9070be8b77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 16:49:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91700830eb9d4fbaac5f6250b992a985\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/09e7407b6538428a80cba2ddaf0b8ae8 : FAILURE in 8m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d8fb19d8f914bc18dcc4ef5ff090c14 : SUCCESS in 1h 43m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b79787f3d334764b9f21d7d51dbdb02 : SUCCESS in 2h 28m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7523469fda442628ab4ead6657b500a : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41a8bdb2735348c18fd4c5732405bc18 : SUCCESS in 6m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/746ee388be0c44b1b9cffea288919510 : SUCCESS in 1h 36m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a74518c8920e4b02ac6e51d49a9dd492 : SUCCESS in 1h 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6248c08e210c807a1dfc0f92d639657320599f7","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2022-12-21 18:26:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b3122eebfc23e696a788d52b169394b0017d5da7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-29 07:46:37.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"21ed7562c97ba467f319759e80a60863ed07a7d0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-29 12:03:17.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"53083f54ea0aad97b8734001febe15f1d410b67e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-03 18:20:41.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d4b899738abdbf715a5f1ad99197e8fcfad113c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-03 18:20:56.000000000","message":"Patch Set 5: Backport-Candidate+2","accounts_in_message":[],"_revision_number":5},{"id":"5e7a11378191437aaf4a4821eba192be634218c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 19:46:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10aa2087e834ef59b7174eebc5f37af\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9c14780b593044a4818fb282c71c3cda : SUCCESS in 19m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a2fd43e419664c5b9d1204b7d96fbedf : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a6e3c697b6248e785bf820cf4b2b2a6 : SUCCESS in 20m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/043e0fac502d4fb9b81dff546b5a9f37 : SUCCESS in 52m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9a300bf0aa944e8a8e00944653073371 : SUCCESS in 1h 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcdbc7e392147c29bb12ac2c5e2b520 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d42eef811abf4cb28a4a54d30c9e2ace : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a1582101d1a4f828ee11a3ab2b80cc8 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/487de0ea344a4ec382be9cb2957e4caa : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081f1c45e97544f9a5750d86777f48ee : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00fa3ad1888c4b0497c82c4981b7b44e : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1bb7896d448b4f66a7786f2514a28978 : SUCCESS in 24m 46s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/05c70ba358dc4e0491bcc14e4422ad81 : SUCCESS in 57m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d48e5948761431488648b2c2e360b77 : SUCCESS in 23m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbfbb4027766468caf43cbdc044ae86b : SUCCESS in 43m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21d237455d404423b4b48cee09e56668 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":5},{"id":"99a8f1a11738bbbf1c400979579688a0d116efbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 20:44:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f25d63ea774c58b24cd141ad02dc28\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d6bc167121c145c295da5ecefe24fc30 : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9aa0b5a4bf03419381291565cb4a2f42 : SUCCESS in 2h 20m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/060dbbdc4e104a058234eaaa8c01ed0f : SUCCESS in 2h 17m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b82f3bfb0b4e4a14bacd810614866133 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c6955ae105a41a19fdb9e4622fe2fc8 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c69f9b8f42224a2a943b375ca784e19b : SUCCESS in 1h 46m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/806b5810b1bf48f88976b81f19906984 : SUCCESS in 1h 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"df2de1596578388f4f641d454f0143cf5496ac1c","author":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"date":"2023-01-04 17:04:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ca865b3318900f53270a221305b6110db2a649e8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-08 21:50:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"25379c45529a3d6b813a8ace38e198da56cf8b85","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-10 14:19:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e3cb2e9683a9a23f7c06a561bbea9516002e65f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 10:31:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"423330be461bf20b88b48bedc48db87bb86acbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:32:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5288d270dce76aaa80c58c919729d7dc3ee0aa77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:36:40.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2b45c22bfd441fe91700d3151bc2052\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f060dd6c29a94b77aaabbe79d122735a : SUCCESS in 18m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f5d1d22790c422aa2f04ae359f8aafb : SUCCESS in 55m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dfbd023bf3aa402f889902a45551d224 : FAILURE in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6a06e0c61f43138de7b8394b7469b8 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec0369a6fae4b5b9891c3e087b5e468 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/993dfab9754b4fd7914b2bf509314257 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244602c3474d48769b084e88354aea89 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75940ad1c15d479bbb418f917de35e71 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b50eb600e27846388ae45b3630e7a9e1 : SUCCESS in 22m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0dffe77cac644841ab54cbccf543feba : SUCCESS in 59m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40b6c2e4b926487bb5fc9784066aa400 : SUCCESS in 18m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dac0f10e64944157918dfda7b49c0434 : SUCCESS in 44m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f90a503471c041c6b9d1c97ba256a063 : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d145279516449a8becc927c009eda21c2bd3f2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 11:42:52.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cb7f8e5e3fd92016379b589074bbd67631ecd01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:07:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e2c7d4f3c504044ae51b3067d2cf091\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/20376b4febdd42e687dbe8d196d0bbe5 : SUCCESS in 22m 27s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b7c8ddfcedac4d079f034007533dca1f : SUCCESS in 1h 03m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b6719c634a4442068c5d47a3db00aad1 : SUCCESS in 21m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ac5f7ae552046a9811479cb24d91de8 : SUCCESS in 1h 01m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ee27a078597e42fb8d61d3e1cb99d4e9 : FAILURE in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/190099aaa3874c4ab23018e563cc202c : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d2ed1b5d834dc79c1dfa5e628b3ab0 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6905792c5d04209b7f49dd8418539a1 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09573a5632194da59063eaca0860ac8e : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c585f60c9974dae80cac08ebf831268 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b733b6e42c4c50b76a621773421768 : SUCCESS in 2m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/75aaa58fd14d474ba3174f2bc8b8a972 : SUCCESS in 21m 56s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/da2ba861895647ec87f6272a7e3eba33 : SUCCESS in 57m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9e547e8853f74872a5f4d86529f8c083 : SUCCESS in 17m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/033826f322e44eba9a62928e2b8384ae : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f6822aa6d1540ab838b42584e1656d1 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":6},{"id":"bf62e208796d699f9f997249e4f51d88e473ea51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 13:25:40.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"1bd523eb5cca11721ae10264721c171efa341188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:44:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23fa3c24b5d0430b8cead069b9b2767c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad8f438087944d23a2254d5a32665190 : SUCCESS in 23m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ae19d7914dad4728a78d04895cec4589 : SUCCESS in 1h 15m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/81bb9df3128e4918a466db8dbe0c63bc : SUCCESS in 22m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36c1b37f8da645bea0df49f2bec3ce0d : SUCCESS in 1h 01m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/13adf6c5015d4e8f801fd8ce5ee5deb9 : SUCCESS in 1h 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ff7051aa064c42a465359199274a85 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f1d59df9fae4840a35968d634d00dbe : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e53e832225d4b9d8d6f8951cbe42669 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b7404eec8ba4c6a90a6e11873308a9c : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e341b13a2a444b8f50b346053c4624 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93b897bb486433b80da000ab470bb63 : SUCCESS in 3m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b128c1c6bdb740389c2ef29156ee9723 : SUCCESS in 24m 20s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/79b54cdde819459a91ed071cee6229be : SUCCESS in 1h 03m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/afd7eb64a5dc4affb02fc21623bfc6a9 : SUCCESS in 17m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c31d7b49077b4744b40defae77ac3fab : SUCCESS in 58m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c351fa0bcfeb4a03accc8e892fa9bd8f : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":7},{"id":"7f2884d61e261a0d87cedb9d381044f10603a78a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 15:55:43.000000000","message":"Patch Set 7: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3e1f04362446f59d1c702a6fb44699cdeb5e5ccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:56:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1cd9095e34e62c911a36833e6c38dbf87cec548c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:29:24.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e087b8807e454f6a8b8c398f47fee6c9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73f69ee92bd1485098b1e9587e9f24b1 : SUCCESS in 20m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bab02bbd3f264fd081d88bdfb14d196c : SUCCESS in 57m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/67d926645d974c1994bc9216c9189616 : SUCCESS in 1h 08m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f38c5034b404618804ad08eb44e23db : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e14e088906194b878b9e1f31d8dbf9f8 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbb7581d92f4498bb1aa3f53b6e08288 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cd5687cf5e441dbff0149156820c9e : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b01baa2bdc0b48418e09bdb9063f2080 : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/904e1fd931f64e8b976db7b230c2e250 : SUCCESS in 24m 30s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9163555be3c840a5a0444dc399dbb22a : SUCCESS in 1h 13m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5bb6bd34f9784ae39ca1a7c8251a95a6 : SUCCESS in 17m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f662915fc464d74abc9aea02e295e9e : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b311938ec7548ea82500f80b092a63d : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":7},{"id":"f218faa9c36f9443497dfff1084f7839f05974b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:29:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1dc9a0e5418ca2e645cd930b2a7d6306a28db3ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:30:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77dd386c37844a78a791e1dd677b0cda\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076cd566f932426795ce5f1670e8728a : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65468d385f4e45a79eb7ac6fcb0970ca : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7},{"id":"37cce8d73e21911ffbd93fe21cb9186a7fb8e210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:45:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59515010d0347aabe93f61fc1ee275f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c2a29813b5bb4095873c99ed203e5ac6 : SUCCESS in 1h 33m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0b7a16934e264d2ca600a80192cc5e5e : SUCCESS in 2h 06m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/963be38198564d2abf83b332d820e54a : SUCCESS in 2h 30m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ef3978c2c134de9b2dd79658e31bfde : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e33bbc0f9f314669be6feccad1b52a2a : SUCCESS in 6m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dfadcab73a7242a2ad91b7ca923bc32d : SUCCESS in 1h 39m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/adb56d2a452d4df6bdee0fa3cf47d06d : SUCCESS in 1h 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1","revisions":{"78f1b94465eae148483e193e125c571f5e07e26e":{"kind":"REWORK","_number":1,"created":"2022-12-15 13:00:41.000000000","uploader":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"ref":"refs/changes/00/867800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/1"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nfixes: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f1b94465eae148483e193e125c571f5e07e26e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78f1b94465eae148483e193e125c571f5e07e26e"}]},"branch":"refs/heads/master"},"0f07cdea28586d3636793b166d5e0d87f972302e":{"kind":"REWORK","_number":2,"created":"2022-12-19 09:19:06.000000000","uploader":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"ref":"refs/changes/00/867800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/2"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-19 09:17:20.000000000","tz":60},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nfixes: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f07cdea28586d3636793b166d5e0d87f972302e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f07cdea28586d3636793b166d5e0d87f972302e"}]},"branch":"refs/heads/master"},"191c74e32039d839e6ef1e9314e4090fcf477d86":{"kind":"REWORK","_number":3,"created":"2022-12-21 14:10:15.000000000","uploader":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"ref":"refs/changes/00/867800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/3"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-21 14:08:45.000000000","tz":60},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/191c74e32039d839e6ef1e9314e4090fcf477d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/191c74e32039d839e6ef1e9314e4090fcf477d86"}]},"branch":"refs/heads/master"},"ab742a919be9e551d26ed945e4e95e530996df97":{"kind":"REWORK","_number":4,"created":"2022-12-21 14:18:23.000000000","uploader":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"ref":"refs/changes/00/867800/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/4"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-21 14:18:12.000000000","tz":60},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab742a919be9e551d26ed945e4e95e530996df97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab742a919be9e551d26ed945e4e95e530996df97"}]},"branch":"refs/heads/master"},"dd96379d93c6902e3c0e90e96e24cc0a903d0601":{"kind":"REWORK","_number":5,"created":"2023-01-03 18:20:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/867800/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/5"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-03 18:20:41.000000000","tz":0},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd96379d93c6902e3c0e90e96e24cc0a903d0601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd96379d93c6902e3c0e90e96e24cc0a903d0601"}]},"branch":"refs/heads/master"},"db614fb3be0ea3b9e6f92a48d235491b3c7fdd2b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-16 11:42:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/867800/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/6"}}},"commit":{"parents":[{"commit":"7b6206294cb88609fd6ea19350d63b940748ba8e","subject":"Merge \"CI: drop kolla_python_version from Ubuntu jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6206294cb88609fd6ea19350d63b940748ba8e"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 11:42:52.000000000","tz":0},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db614fb3be0ea3b9e6f92a48d235491b3c7fdd2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db614fb3be0ea3b9e6f92a48d235491b3c7fdd2b"}]},"branch":"refs/heads/master","description":"Rebase"},"a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-01-16 13:25:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/867800/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/867800/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/867800/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/867800/7"}}},"commit":{"parents":[{"commit":"7b6206294cb88609fd6ea19350d63b940748ba8e","subject":"Merge \"CI: drop kolla_python_version from Ubuntu jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6206294cb88609fd6ea19350d63b940748ba8e"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-16 13:25:40.000000000","tz":0},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/870540\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
