)]}'
{"id":"openstack%2Fkolla~867831","triplet_id":"openstack%2Fkolla~master~Ic78fe52b19e38dc540e4a51fdacfcb26c1ba834d","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic78fe52b19e38dc540e4a51fdacfcb26c1ba834d","subject":"CI: use Tox 4","status":"ABANDONED","created":"2022-12-15 17:32:32.000000000","updated":"2023-02-07 14:14:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"36661f1c3636f43d713395e5603230ce04c9e240","_number":867831,"virtual_id_number":867831,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-15 18:54:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-12-15 18:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"984c6cbf603ed78d09a8193967d4ed09e43fb670","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-15 17:32:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3d3aa0b5ecaa2658653627f66c4d135e8b78210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 18:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40dda5b9c4664162b33d0f9eadc219f5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7a63f802a0f64baa9263beea456fb653 : FAILURE in 4m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/83ddf23668954bb5b0fb4c0d8af89735 : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/38c896acad2a4ab3b993263803da80d4 : SUCCESS in 18m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd2fa8c4516f4d48b8be389de76c5bd2 : SUCCESS in 56m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/085f25c537034cd496f46ab80a9d008d : SUCCESS in 1h 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24cf828d48ac4f3cbebc6a49c398c10e : FAILURE in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a63f79a21f9b43c0807175cc382ed459 : FAILURE in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d83f459d73c548948b89517198408916 : FAILURE in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb495f0646e42dd9405cd8caf99a212 : FAILURE in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec2aafa8d98343b98f3fd3edf41d8244 : FAILURE in 1m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7ae567f2c8412e8a08377d801192fd : SUCCESS in 5m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/82484ff00cbf472aa43513e390a21441 : SUCCESS in 21m 02s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/cae09ab3f7de4aefa111492b0a2fc87a : SUCCESS in 1h 04m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35f4ca271b7c4d86ad8dc06e643d3f53 : SUCCESS in 23m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d340867f8bb047359e8833e6155efc4c : SUCCESS in 44m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/148d1d63ca794d3e8935b0d7a97637e6 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f90725ed8a5e195fd809514ad3c730a27bb11966","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 15:48:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eacc365a4ed842309a3acc8ec7b3cbb0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ff96822f1017412ea4d9f24c57c61a06 : SUCCESS in 1h 21m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/801e44058364452a846b9a0f9686edbd : SUCCESS in 1h 45m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f11b7efd9cbb418bb4b0e5393bfe586a : SUCCESS in 2h 13m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d30bc4e7ae2246fc97ca121294fe4883 : FAILURE in 4m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65ae1a6a14d6450fb3a0cfe5e764a00d : FAILURE in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b1d55211a59b469dbfe97c0d131e7fb5 : SUCCESS in 1h 48m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6cd061fae39446b3900edccffa88eaee : SUCCESS in 1h 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36661f1c3636f43d713395e5603230ce04c9e240","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-07 14:14:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3713cc667c6e97b7d475560f8b0d07f3b7b429e6","revisions":{"3713cc667c6e97b7d475560f8b0d07f3b7b429e6":{"kind":"REWORK","_number":1,"created":"2022-12-15 17:32:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/31/867831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/867831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/867831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/867831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/867831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/867831/1"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-15 17:29:55.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-15 17:29:55.000000000","tz":60},"subject":"CI: use Tox 4","message":"CI: use Tox 4\n\nChange-Id: Ic78fe52b19e38dc540e4a51fdacfcb26c1ba834d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3713cc667c6e97b7d475560f8b0d07f3b7b429e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3713cc667c6e97b7d475560f8b0d07f3b7b429e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
