)]}'
{"id":"openstack%2Fkolla~867836","triplet_id":"openstack%2Fkolla~master~Icf7d1988aa7208cda1b6f70269b9b54cd201b730","project":"openstack/kolla","branch":"master","topic":"no-binary-in-zed","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-12-16 11:57:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf7d1988aa7208cda1b6f70269b9b54cd201b730","subject":"CI: do not check for error in binary images","status":"MERGED","created":"2022-12-15 18:00:12.000000000","updated":"2022-12-16 11:58:19.000000000","submitted":"2022-12-16 11:57:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867836-no-binary-in-zed","meta_rev_id":"f3fac2bce90085cc5c72769eaf2f04de39902d1b","_number":867836,"virtual_id_number":867836,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 11:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-16 10:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-12-16 08:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-15 20:37:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-12-16 10:42:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 19:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 20:25:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-12-15 20:37:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-12-16 08:31:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-16 10:42:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dc6c7becf5f7b09c209fe040ab82a78276ace95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-15 18:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffd17b88f3d309138f3926a1b930d2f9daefb433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 19:19:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fe5c50003574c8b91a763ecccd09a32\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/479decb96b5645e58fd55fcd5edb7e1e : FAILURE in 3m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/21039808e3ee4172b4898ccec811b479 : SUCCESS in 1h 17m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/51114ecbc02c48a89584f1acc67e1e4f : SUCCESS in 19m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4facbba1112443988da68b3a7e98a372 : SUCCESS in 53m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7064ee21118e402baf5815c042f13150 : SUCCESS in 1h 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340a25928bd544e08e115417f64dfc93 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07b894df030245069b43a9b12b121e74 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6e434029fce474a9231be3aed9aed1e : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34f9742bf88246a5856c069ba60339a3 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fcad8d023a0418ab949761b2128ad13 : SUCCESS in 5m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/34b37247b87c49bcb058d0de61d1c7fe : SUCCESS in 22m 27s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/a7f78e3659574fb691ef05e8ddae4929 : SUCCESS in 58m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d82a6d879c7d4d85961643af7cc8b507 : SUCCESS in 29m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cc97658992444c08ea324368fb18de0 : FAILURE in 24m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/400b909c349d48b5b8250338bc96d8e0 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7263c0762658906f97ad3f34d32f8463c4b3953a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 20:25:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ff4082caf61f33702b2ccc0c2ead9d2cc8c0185","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 20:29:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd44cef01e747b689542ffaa061263c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/310c1208259c484587b81e378ef5b2c8 : FAILURE in 8m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3ec26fa7d784dbf972d5da919577645 : SUCCESS in 1h 21m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ac2573b3251404491e501809e7aef15 : SUCCESS in 2h 26m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32cdbc96dd9d41efa009729df5dcb143 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45db84ebf8ca4711ae1605eeb066c436 : SUCCESS in 5m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f2c8d8ace1cc4d0aaf3d33131b5d2430 : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cec2225b6a61495faee0a73e17a0b3d7 : SUCCESS in 1h 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bc34ac26ac60090965a4dda018f5c939b85faf7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 20:37:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"de7e5f306e643861738d78e17fb26206947520d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 21:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/341713005c7f4db39ca196a9fef9c63d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/648ae96c279a4289a04bd80ae5c516ac : FAILURE in 3m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b31fbef3fddd428ba1dbac74f99bd2f6 : SUCCESS in 56m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/18af5e445b1d4603a853f29c859c7843 : SUCCESS in 18m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6aa2be2f4b7d40658fd2d25841c0403d : SUCCESS in 44m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/776a02e354ae4b48a3ca6feb62f65b78 : SUCCESS in 1h 00m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e884f70cd1234ce091984486ccae9193 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d3a7a2dbd79470291fb6deb786a5dcc : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bab7031d25f4bf1a0ee6adec9a75dc6 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec65b59807e54e859a8bc19b334d9c36 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a04d06d9a84364ad28059f3c9b9d54 : SUCCESS in 5m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5f21bc6a5cb54aa9a77671c2c520fdfa : SUCCESS in 21m 18s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/c659afc75a4546498195f09540d7f2bc : SUCCESS in 1h 05m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/56c00f046cf34c21a004e7ffa302794b : SUCCESS in 21m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0e364f388614c5facf1b9f0d58fbaa0 : SUCCESS in 39m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/819e81ed71df432792b5f33f5e231c26 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8429ce0d48bd0cf93b37722313f4de53a6cf3555","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 08:31:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d82f7526d9cfcfad40e65242fd71f357f5d55a16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-12-16 10:42:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"161f518ed985fef781c7f59f5b055bac08aaa79b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 10:42:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdb7e7979c3cd9e8170b1ca9339a78e5c00bef68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:57:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de08d78a4ad84253b0e82aae1512fc71\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c084690bc1bc4128a401fe8c2f527904 : SUCCESS in 17m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e1264f0024204caea1057e8ac01b3cbf : SUCCESS in 56m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2092ea205e594cadb885182af343fbde : SUCCESS in 1h 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42228d63de95457f8634841a8dc6f80e : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81fd9b0418e646e9b1a33864de94d72f : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8e996d5d3a14e15a88a4b86222997f9 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0576908f93a46978d4df8a5b8ff5432 : SUCCESS in 5m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/555dc99fb96845abb6f658025a262058 : SUCCESS in 21m 15s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/cd68c118476c49cf89ed550c6718476f : SUCCESS in 59m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4813546ce3154719bc10e00fd8eacd72 : SUCCESS in 16m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15ee1dbe888349188526b777da026f7e : SUCCESS in 41m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0c3ef8a341cc46b9a14b32fe8cd4318f : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"80c22c1852d7ff9b57fbbcef3115a18ea790d7c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:57:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f3fac2bce90085cc5c72769eaf2f04de39902d1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:58:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d12954d3c0a4dec9697d3b60686a05f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc85d9a505a4411bb952713e3a8ff099 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0b9a9704d5d799f99a5cfa7923cb1209da16ac3","revisions":{"a0b9a9704d5d799f99a5cfa7923cb1209da16ac3":{"kind":"REWORK","_number":1,"created":"2022-12-15 18:00:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/36/867836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/867836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/867836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/867836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/867836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/867836/1"}}},"commit":{"parents":[{"commit":"9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e06e70ebb7f4afb07a5fa5c193e2a9c04e577db"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-15 17:59:44.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-15 17:59:44.000000000","tz":60},"subject":"CI: do not check for error in binary images","message":"CI: do not check for error in binary images\n\nThere are no such anymore.\n\nChange-Id: Icf7d1988aa7208cda1b6f70269b9b54cd201b730\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0b9a9704d5d799f99a5cfa7923cb1209da16ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0b9a9704d5d799f99a5cfa7923cb1209da16ac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
