)]}'
{"id":"openstack%2Fkolla~867919","triplet_id":"openstack%2Fkolla~stable%2Fzed~Icf7d1988aa7208cda1b6f70269b9b54cd201b730","project":"openstack/kolla","branch":"stable/zed","topic":"no-binary-in-zed-stable/zed","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-12-19 15:13:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf7d1988aa7208cda1b6f70269b9b54cd201b730","subject":"CI: do not check for error in binary images","status":"MERGED","created":"2022-12-16 12:07:06.000000000","updated":"2022-12-19 15:14:17.000000000","submitted":"2022-12-19 15:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867919-no-binary-in-zed-stable/zed","cherry_pick_of_change":867836,"cherry_pick_of_patch_set":1,"meta_rev_id":"3e02b1dc884840bfc40dbfb10c20a56c63ce0449","_number":867919,"virtual_id_number":867919,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 15:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-16 12:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-19 14:00:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-19 14:00:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 12:07:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-16 13:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-19 14:00:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d66c47324e29fd499f65b006d712aadd2a21418e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-16 12:07:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ff53c80f8f28a8ac59c7e37ac95f5e401fda5026","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 12:07:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23e853a5741d89542e740b26c60f79d3223babef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 13:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a04c00891b746179d59ea0cdea77ece\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6a4fcd892b9b4317824adf97ad8622b8 : FAILURE in 4m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/058440442e21464eb2ffe77095fea8b4 : SUCCESS in 1h 09m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1a507717fbf54562b816b93bc60a1b97 : SUCCESS in 18m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/feeb98f7b6574bc684a2ba82dde50068 : SUCCESS in 57m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/edb5e7d254424945a3f1ceba0aefa234 : SUCCESS in 1h 09m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19d7f1e4fee42ab818a700821c5781e : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4e95593b4574fa29c48f1798bc9d726 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f3dbe41b68746aa919f9cd38cd0d4b4 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3afc5f509c74bdcb5b5e9e90c3b1792 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9af347d234f4255b02ca492ffff38c3 : SUCCESS in 4m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bb7aa9f5eb4d4465a95fae9aed37a918 : SUCCESS in 24m 00s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/e584b6cfed71407a9a9e776cd10b53d0 : SUCCESS in 1h 00m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/05fba51cfcc64993947a4b285a5e51b7 : SUCCESS in 17m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a84d2532962f416f86cd1fe6b5a3a3da : SUCCESS in 37m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e74da9b31e744d36ac8f19605abb2322 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"057aae486edfd410c57b57cc755db47edac84550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 14:00:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41ae447fe3f4910b26107595a2236a2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/807c6c615c5b4304ba326b0103909f8e : FAILURE in 8m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c8c4a0c151c446f8dd9053556338c2b : SUCCESS in 1h 22m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/efb6650b5e9747f181e696e1ed00800d : SUCCESS in 1h 50m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69700aed949d47ec9073cbc6df426ac7 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/866992ab37aa4d93a255201f0e95cb06 : SUCCESS in 4m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/71b4abaa30dc4f8db18aecc7b573ae41 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b35d561c40324523a08ab5ee69294788 : SUCCESS in 1h 06m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a97e18adb22780f5b350b7f0a45a7df1bdd665de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-19 14:00:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e69080fde6e113bbf0560ab11ab8ca4e2c7d6408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 14:00:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4168ea78d998e081a6bdfa7b6f73c77edff8d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:13:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399abbd1eeb5460da62f3b65530bd6f8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/befba0682df94c6e8dd803e8a4d22703 : SUCCESS in 17m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/66d2750c001a4f159629e85634bb0f19 : SUCCESS in 1h 10m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/58ab5ac3998f46b68090a5dc00c2500c : SUCCESS in 1h 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa70e8540434b32b1f76d621b9c7cc4 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80fcb970630145fa8068f163aaa50a0a : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88c8013e42f04395b011b48f7f60de7d : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f0a2a231d4840dcbcc45c836f41a6a6 : SUCCESS in 5m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9fb2a5ac48424204b9f126a01e4dad7c : SUCCESS in 19m 17s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f74e020c20b8408d9b11dfc7f1000582 : SUCCESS in 48m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc27d4fbf1d74ac09717a2d48c00dabe : SUCCESS in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/afbe2a49932f412d9b6aeca77ddf87e9 : SUCCESS in 53m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cad4810572ec44cdb56099bcfcb38493 : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7a24afa5297e2e53c212ed6be8e0cf89663e2d58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:13:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e02b1dc884840bfc40dbfb10c20a56c63ce0449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:14:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25bd9b3ded144d79aeb8a73cfb58844d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2f92da134d4332883ae60765b57c3c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5b8d5198e2e8f68dc8f6c4947caede64149d202","revisions":{"b5b8d5198e2e8f68dc8f6c4947caede64149d202":{"kind":"REWORK","_number":1,"created":"2022-12-16 12:07:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/19/867919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/867919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/867919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/867919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/867919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/867919/1"}}},"commit":{"parents":[{"commit":"920de01f127388b7620240518cf801afe1060fec","subject":"Change default base from centos to rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/920de01f127388b7620240518cf801afe1060fec"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-15 17:59:44.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-12-16 12:07:06.000000000","tz":0},"subject":"CI: do not check for error in binary images","message":"CI: do not check for error in binary images\n\nThere are no such anymore.\n\nChange-Id: Icf7d1988aa7208cda1b6f70269b9b54cd201b730\n(cherry picked from commit a0b9a9704d5d799f99a5cfa7923cb1209da16ac3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b8d5198e2e8f68dc8f6c4947caede64149d202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b8d5198e2e8f68dc8f6c4947caede64149d202"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
