)]}'
{"id":"openstack%2Fkolla~868564","triplet_id":"openstack%2Fkolla~master~I84b4d006fba4aaedcdb324c24e42a9c0b441405f","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-29 19:50:29.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-12-29 18:25:17.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-12-29 19:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84b4d006fba4aaedcdb324c24e42a9c0b441405f","subject":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"","status":"MERGED","created":"2022-12-28 14:07:36.000000000","updated":"2022-12-29 19:52:09.000000000","submitted":"2022-12-29 19:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":866577,"submission_id":"868564","meta_rev_id":"02c92b2711a6124e3c67bbe12bf618ab74a75832","_number":868564,"virtual_id_number":868564,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-29 19:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-12-29 07:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-29 18:25:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-29 18:25:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-12-29 07:43:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 14:07:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-12-28 14:07:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-28 14:07:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2022-12-29 07:58:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6f7f4c718459bc95868b2ab3bfa5a6404caa9c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 14:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bbd5caedd5ab361fc89318dc92095fd9053e82c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 15:41:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"910d1e3eddffebedda19f4a22561a0aa488f9a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 17:02:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ef5e9cb50aa44108512a0a6acbcbc77\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ef824fb2c4b84369925c6c1b087f3e79 : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f4868eddf682437db480aeff23216280 : SUCCESS in 1h 11m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/61eebdc93da34889a213f175fb46a420 : SUCCESS in 22m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af81a7b344f2435e8e8fd6311a6fd804 : SUCCESS in 1h 01m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/37d52e45ef8c41098d2c380aebe4e36d : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2faf0db632940a1a394f4c3d1bed267 : FAILURE in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b32f9622ae4d988a07c4b67b2ce387 : FAILURE in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/629e94f06a0346b7ba16662b06816ba3 : FAILURE in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8af66657e44acc838a615bc2b39fbc : FAILURE in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a456ce4ba71f4c89ad08bd0f98306774 : FAILURE in 3m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/daf5986550e64c46b90260600ad5b471 : SUCCESS in 23m 41s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/538735e2de5a49ee9ca998ea98eabe87 : SUCCESS in 1h 01m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f064f239af2477984607746e911e4cd : SUCCESS in 19m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28eb7bfe2d4942e8b05c024f96dc57aa : SUCCESS in 44m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/198bc458803a499293de09dc4a3e3df7 : SUCCESS in 55m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3b8b42668c52628e99e41f5a15cbb40c349c336a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 17:42:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6c7c3fd3716578847ca3c7976e1a441e3abfd369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 19:05:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/663debc840b341e0992d23c6ca43726e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/63fbb09487484b23a28a93291f39a0b9 : SUCCESS in 23m 07s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3497e272496d4fae8670a8a6e2cd66f2 : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2e41570646344c61bbf092a9ea412700 : SUCCESS in 21m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be3808457772452d96a2ddda616464b5 : SUCCESS in 1h 11m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d16cde259ca148b6936d742da3461421 : SUCCESS in 1h 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff41a0562cd84183b0fb5d144fa9e761 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7082b32a4eec45b6b00f8d03ba71239e : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a485f86a5c1c44f48f839b5026641223 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c3dda4108a472fa3684fc958b4b3c6 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce17b746ca94e2286136ded6ff83b4f : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d08c4cad1c0346159d22d5c9bf369be1 : SUCCESS in 28m 07s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/bee4d1c206d64d10878be7cf242c8291 : SUCCESS in 1h 14m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ae355e1b4a7b435c8d6af1a7b4c3e7d2 : SUCCESS in 21m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/136a0256a3d94d8c9a51e17e719f9aba : SUCCESS in 35m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01a930bc10bd422cbac5f5910d40b800 : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":3},{"id":"3ef2750179882c9cc67267cf9acaaa8cdd26b850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 19:54:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb66fac3980740379904b2fe0265f48d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/30990643aa45486fbae14717caa2c649 : SUCCESS in 1h 15m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5d139cafeac1459fa26f962e8c4eccd8 : SUCCESS in 1h 30m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b48e1151d36f4b81b7b153d85c134e38 : SUCCESS in 2h 09m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e1c27098b9342889da0ef185bbf14bc : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/673ccf69b05046d6b2a4978c4a38a909 : SUCCESS in 4m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/078a4d3670804616aa860a205ada2830 : SUCCESS in 1h 23m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/66833744718c4690917edfdb9ca14c63 : SUCCESS in 1h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93e119bb5c214812a1fdfe57f5f749a9333f698e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-29 07:43:15.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"44a64583a4a85a3e9cf26c0e2f848aa8e0aeec1b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-12-29 07:58:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3358faee06422460881a58ed15070c9cbb761d26","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-12-29 18:25:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06128628d824e7a3ef941939c291c2c18a38d2de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:25:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"feaa11c599dc67f7e01cb903180c15d703f545ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 19:50:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beaf726443dd4840af85ba9d31555127\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bb6457628f24441cba86014c7b6366b2 : SUCCESS in 20m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6e328a569fd74e4785c8f5653db9427e : SUCCESS in 46m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9dab943755954cd2940d2c0b51f980f2 : SUCCESS in 1h 08m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c506d4ef1c4f460589bf386178ce38fc : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c01464931a194d1a8c0661cf07b83131 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de18e4577e934d7186a95c281cbe92cb : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e847af5187478380df62ee7b601e0f : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/98949f4d9855406c948c6660daea583f : SUCCESS in 18m 51s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/256b0deba39345d19176c534a14f403b : SUCCESS in 1h 03m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f92f451508fb42d580d5e3c590e496a5 : SUCCESS in 15m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0720cc0ae8c845a4b4c47a8878e89913 : SUCCESS in 45m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b475ea993b7c480880021b4e4b9aaa24 : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":3},{"id":"9885e26f71961586dd1693bb6550f402ecc6e916","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 19:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"02c92b2711a6124e3c67bbe12bf618ab74a75832","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 19:52:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d9b088a1db46b0b2b2c690feebd053\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da6b846ec34442ca59797fbb4f23470 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b3c64dce820601cb2d9a9ae1a141392c2abe95d","revisions":{"233c8f9ed1ab80535466c673903d878fd364057d":{"kind":"REWORK","_number":1,"created":"2022-12-28 14:07:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/868564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/868564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/868564/1"}}},"commit":{"parents":[{"commit":"206521af2881425cb91fe0365fe87d35a4b2de3f","subject":"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/206521af2881425cb91fe0365fe87d35a4b2de3f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 14:07:36.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 14:07:36.000000000","tz":0},"subject":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"","message":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\n\nThis reverts commit 206521af2881425cb91fe0365fe87d35a4b2de3f.\n\nReason for revert: Reverting after Zed release\n\nChange-Id: I84b4d006fba4aaedcdb324c24e42a9c0b441405f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/233c8f9ed1ab80535466c673903d878fd364057d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/233c8f9ed1ab80535466c673903d878fd364057d"}]},"branch":"refs/heads/master"},"142a2f989d205c4d94d9ee45a30c0500df97baad":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-28 15:41:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/868564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/868564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/868564/2"}}},"commit":{"parents":[{"commit":"a0b9a9704d5d799f99a5cfa7923cb1209da16ac3","subject":"CI: do not check for error in binary images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0b9a9704d5d799f99a5cfa7923cb1209da16ac3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 14:07:36.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 15:41:56.000000000","tz":0},"subject":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"","message":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\n\nThis reverts commit 206521af2881425cb91fe0365fe87d35a4b2de3f.\n\nReason for revert: Reverting after Zed release\n\nChange-Id: I84b4d006fba4aaedcdb324c24e42a9c0b441405f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/142a2f989d205c4d94d9ee45a30c0500df97baad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/142a2f989d205c4d94d9ee45a30c0500df97baad"}]},"branch":"refs/heads/master","description":"Rebase"},"6b3c64dce820601cb2d9a9ae1a141392c2abe95d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-28 17:42:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/64/868564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/868564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/868564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/868564/3"}}},"commit":{"parents":[{"commit":"a313ed46dd5ac9c15100bb5a748322c2840ee5c0","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a313ed46dd5ac9c15100bb5a748322c2840ee5c0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 14:07:36.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 17:42:38.000000000","tz":0},"subject":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"","message":"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Zed\"\n\nThis reverts commit 206521af2881425cb91fe0365fe87d35a4b2de3f.\n\nReason for revert: Reverting after Zed release\n\nChange-Id: I84b4d006fba4aaedcdb324c24e42a9c0b441405f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b3c64dce820601cb2d9a9ae1a141392c2abe95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b3c64dce820601cb2d9a9ae1a141392c2abe95d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
