)]}'
{"id":"openstack%2Fkolla~868996","triplet_id":"openstack%2Fkolla~master~I08fe2827bc58baa95d396daf28de1079e3df4f77","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-18 10:33:13.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-18 10:33:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I08fe2827bc58baa95d396daf28de1079e3df4f77","subject":"grafana: remove SHA1 workaround","status":"MERGED","created":"2023-01-02 11:14:35.000000000","updated":"2023-01-18 10:34:19.000000000","submitted":"2023-01-18 10:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868996","meta_rev_id":"2dc8f93eb043608854600f7af8efa6948e15ba1a","_number":868996,"virtual_id_number":868996,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 10:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-18 08:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-01-18 09:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-01-18 09:00:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-18 08:12:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-18 08:12:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 06:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-18 06:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 08:12:11.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-01-18 09:00:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b6e5ffa8adbe7d9a753e1686c600b7d63d7834b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 11:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc6ce0167ccb76594ca07abcc6d09b6f235c8462","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 11:28:34.000000000","message":"Abandoned\n\nstill fails","accounts_in_message":[],"_revision_number":1},{"id":"5cbe8dfcdf692af9618ef6308934bb230f3eed20","tag":"autogenerated:gerrit:restore","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-18 06:38:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a5bdb52f94b51a84fd6ae0e66d7c5204e97f49bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 06:38:37.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 868996,1","accounts_in_message":[],"_revision_number":1},{"id":"90676b230c51e51b10828e8640f8b9ac77a7ff2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 06:38:39.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 868996,1","accounts_in_message":[],"_revision_number":1},{"id":"cd76108cd4ce645cbbac1db886ebb34f0ca82397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-18 06:39:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b54aaa2b93db16a4ef9a194d34ebe9907782c32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 07:58:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c30393f089cc4cc2b60f149861661b8c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/daff290c294a4c049da2397317961543 : SUCCESS in 23m 15s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5fadeaa55b03470c8a18e8dd00915bce : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8c506d2709444ba4a4b69526e24e542e : SUCCESS in 19m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f8690738d4804195a25b8f0e4f9e0615 : SUCCESS in 1h 02m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8a46625ea2b84eeba1dd18a4693d6943 : SUCCESS in 1h 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ab66aa94394b3ba486bd1f4ad558ac : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fab196d45064c8ea377092f4673bd48 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ef36ecb0cfb44adb10dea0cd4b4a654 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec3add9f2b064bca9525bc5216c71b28 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacc64f323c84cd9ad9c7060cc36a206 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f76552c1a4024575b291d33696ee8572 : SUCCESS in 22m 35s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/85b6457eda0d4dd78470f5b9c1998101 : SUCCESS in 1h 02m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d675943790df448c989aebdde1b3f80e : SUCCESS in 23m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/672f5fff5a30450299a01187a9b4b662 : SUCCESS in 44m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/43c5d4ca3c9743b099a83328e291a33f : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":2},{"id":"43c59c3861d62438111833d56ec404420c9705c1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-18 08:12:11.000000000","message":"Patch Set 2: Backport-Candidate+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"f4692580652dbc177668d34d79d4ba0d328000f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-18 08:35:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7feb25870a8c68c3f41100d5572302890448e2c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-18 09:00:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce81a8c00b4bfd650bdbccbeb90979a9efb3be52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 09:00:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ecec0070735f3f56f35c44dad30ccf85cb661a6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 09:01:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190f35b80c834dd5ab1e31cbe037a7da\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4ec954146d8c4021843940952880a749 : SUCCESS in 1h 55m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5648681e1171483bbbb70a1a17d48ede : SUCCESS in 1h 52m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07a1e904d31b46db8902b127db386649 : SUCCESS in 2h 12m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26af397867ea421aaca297e72397b046 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14ed6ebf01704de080239e1a8f918dce : SUCCESS in 4m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb4bd64703f8424e8cb27fc78fd05933 : SUCCESS in 2h 19m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d17ef351863e4a26a2df312660ebc4e7 : SUCCESS in 1h 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3f7dd183e3d50aa96aaa6e218c73b79c08513da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 10:33:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5088451c78644add9457bfb28f5d3a4b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2e0e886dcc244fe3b4447aeb68b1f2f1 : SUCCESS in 20m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eac683b60c9a481e9c413428da21b963 : SUCCESS in 55m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d273cc5a6da445acbd3a779d756f4447 : SUCCESS in 57m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e168d205d38f48ddb3d12cbc6fd6818e : SUCCESS in 2m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001f2c2fb85a4265ad40a9171d5dc160 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa40def3587a4a178b383265fdd20746 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12ee99bcb324fb185de9f35911f65bc : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2eb4621e1f0b49c4be43ad21a0989fc0 : SUCCESS in 24m 43s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/9103b69e27504fd59d969e91d45cd663 : SUCCESS in 1h 17m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2a7996d693384047a38a528300fb2276 : SUCCESS in 25m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc2f78e5f7124c6d97470c87b97cb159 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/595a75a6b63347ef8d561ab2b9700192 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":2},{"id":"267f8b483d418c33e66998a210dbf43f953c0188","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 10:33:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2dc8f93eb043608854600f7af8efa6948e15ba1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 10:34:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d3f67d093a94d219bd9a06a6e71ab58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6976ff1bad2a4dbb8ee32ce7f5cbbd6f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47a9a818596b27f658fb7b40aac2130fd04b66b6","revisions":{"64c19840029431f88d2146327be625406b406164":{"kind":"REWORK","_number":1,"created":"2023-01-02 11:14:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/96/868996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/868996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/868996/1"}}},"commit":{"parents":[{"commit":"d079dd67e3b6f3cca560444809e0d6cedad9d9d3","subject":"Add locals_base option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d079dd67e3b6f3cca560444809e0d6cedad9d9d3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 11:14:12.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 11:14:33.000000000","tz":60},"subject":"grafana: remove SHA1 workaround","message":"grafana: remove SHA1 workaround\n\nChange-Id: I08fe2827bc58baa95d396daf28de1079e3df4f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64c19840029431f88d2146327be625406b406164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64c19840029431f88d2146327be625406b406164"}]},"branch":"refs/heads/master"},"47a9a818596b27f658fb7b40aac2130fd04b66b6":{"kind":"REWORK","_number":2,"created":"2023-01-18 06:39:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/868996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/868996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/868996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/868996/2"}}},"commit":{"parents":[{"commit":"a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 11:14:12.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-01-18 06:39:30.000000000","tz":60},"subject":"grafana: remove SHA1 workaround","message":"grafana: remove SHA1 workaround\n\nChange-Id: I08fe2827bc58baa95d396daf28de1079e3df4f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47a9a818596b27f658fb7b40aac2130fd04b66b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47a9a818596b27f658fb7b40aac2130fd04b66b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}]}],"submit_requirements":[]}
