)]}'
{"id":"openstack%2Fkolla~869802","triplet_id":"openstack%2Fkolla~master~I757482e0c9df099197d0e5f388bbc8541aa4c651","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-11 13:39:43.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-11 12:16:11.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-01-11 11:42:36.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I757482e0c9df099197d0e5f388bbc8541aa4c651","subject":"CI: drop kolla_python_version from Ubuntu jobs","status":"MERGED","created":"2023-01-11 09:45:19.000000000","updated":"2023-01-11 13:41:34.000000000","submitted":"2023-01-11 13:39:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869802","meta_rev_id":"b126e5c7079b5381436e89ade6034c4410a74de4","_number":869802,"virtual_id_number":869802,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 13:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-01-11 12:16:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-01-11 11:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2023-01-11 12:16:11.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-11 09:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-11 10:50:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-01-11 10:50:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"0242d4d325dd7e1219339ddb235abec19c65d7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-11 09:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46691a7844ac757789edab283892dd96484ae9b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:45:41.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027kolla-ansible-ubuntu-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu-source:\n          - kolla-ansible-ubuntu-source-upgrade:\n      check-arm64:\n        jobs:\n          - kolla-build-ubuntu-aarch64\n      gate:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu-source:\n          - kolla-ansible-ubuntu-source-upgrade:\n      periodic:\n        jobs:\n          - kolla-publish-ubuntu-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-ubuntu-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-ubuntu\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-ubuntu-source-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-ubuntu-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-ubuntu-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-ubuntu-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-ubuntu-source-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-ubuntu-source-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-ubuntu-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a8fdcdf273dc874d86bab1f3288272b47b2884ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:45:46.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027kolla-ansible-ubuntu-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu-source:\n          - kolla-ansible-ubuntu-source-upgrade:\n      check-arm64:\n        jobs:\n          - kolla-build-ubuntu-aarch64\n      gate:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu-source:\n          - kolla-ansible-ubuntu-source-upgrade:\n      periodic:\n        jobs:\n          - kolla-publish-ubuntu-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-ubuntu-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-ubuntu\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-ubuntu-source-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-ubuntu-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-ubuntu-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-ubuntu-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-ubuntu-source-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-ubuntu-source-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-ubuntu-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"13de56d4daf183ef95d438721a7251d58be49beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-11 09:59:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2eb97418d82459618f91a0d457332e4324e6af4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 09:59:58.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 27, column 18","accounts_in_message":[],"_revision_number":2},{"id":"f35934bb357bcbe92cb43a84730c8433f52876eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:00:04.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 27, column 18","accounts_in_message":[],"_revision_number":2},{"id":"3c16a03cbf2d223a4093eb120589285c19283fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-11 10:12:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"649301038f81b0f931c44de4af334dc9ea04de77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 11:03:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4374049c28d443c68124214f08eb0afc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a92b7c000f384749adb619765abc7c5c : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4a1de830f62491f93f46124e9b61115 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37820a253af771a1135266b0b855b75d1a2612af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 11:29:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d9f1fa0206431da59015a13c0b5a35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a7742c446d45c39f7c916db64f1ee9 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38abd84406c46648aff05edc3c228b7 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d1d90e97939418dbdd6241c9a90b316 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f26387bd2fd445a96c5875e0913d31f : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9694ed60e54f949e6790722ddc1a19 : SUCCESS in 4m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/448a992712a24e5da3e98d1a43f1c69c : SUCCESS in 39m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32e7b36e09eb423d97fab5a0b6c62dd5 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":3},{"id":"fb38e27bea116321c389cc741c49133085886028","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-11 11:42:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"af62027cdd6957d3a9bf9593a8e3b48ae821d31b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-11 12:16:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"402c17214b1f7e81ddce74c5ed28f6fba56af4b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 12:16:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"61145427e9dffd310f5f121d53767bfd12abb024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:39:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c8e3aea4ae4c188c7525e94244ca3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ebd2b41b00140ea850696655ed55d80 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ea730f371b4050a1b7b3e46f3e8cf2 : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80bb8cb52e06496ab1af7edd69406c29 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42e32a3eb35480f9aa4cbc879547d4a : SUCCESS in 3m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba37f3ef44a94a8fb337a0587d14c961 : SUCCESS in 44m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8037587506c94efea5ec6adffdb66adc : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":3},{"id":"ce832e963b12a571c476afb1a6b7653a3fb03ab8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:39:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b126e5c7079b5381436e89ade6034c4410a74de4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:41:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d81c8d768343aebd6bc6efd1f64d89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1d9c73febe4ddeb91d43cd50ba2a64 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"629b1241fc5b53bac3920e88d9a97e401a053b71","revisions":{"c898dc9e634e012762c9ba29dd9618cd768e9e00":{"kind":"REWORK","_number":1,"created":"2023-01-11 09:45:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/869802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/869802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/869802/1"}}},"commit":{"parents":[{"commit":"9fd6e3608ec1b79ec72199c63af77141cf7e0003","subject":"Drop skydive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fd6e3608ec1b79ec72199c63af77141cf7e0003"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 09:44:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 09:45:17.000000000","tz":60},"subject":"CI: drop kolla_python_version from Ubuntu jobs","message":"CI: drop kolla_python_version from Ubuntu jobs\n\nChange-Id: I757482e0c9df099197d0e5f388bbc8541aa4c651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c898dc9e634e012762c9ba29dd9618cd768e9e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c898dc9e634e012762c9ba29dd9618cd768e9e00"}]},"branch":"refs/heads/master"},"6be918d6b45c27dc4c12545e73f9afadb5bc5997":{"kind":"REWORK","_number":2,"created":"2023-01-11 09:59:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/869802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/869802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/869802/2"}}},"commit":{"parents":[{"commit":"9fd6e3608ec1b79ec72199c63af77141cf7e0003","subject":"Drop skydive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fd6e3608ec1b79ec72199c63af77141cf7e0003"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 09:44:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 09:59:27.000000000","tz":60},"subject":"CI: drop kolla_python_version from Ubuntu jobs","message":"CI: drop kolla_python_version from Ubuntu jobs\n\nChange-Id: I757482e0c9df099197d0e5f388bbc8541aa4c651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be918d6b45c27dc4c12545e73f9afadb5bc5997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6be918d6b45c27dc4c12545e73f9afadb5bc5997"}]},"branch":"refs/heads/master"},"629b1241fc5b53bac3920e88d9a97e401a053b71":{"kind":"REWORK","_number":3,"created":"2023-01-11 10:12:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/02/869802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/869802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/869802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/869802/3"}}},"commit":{"parents":[{"commit":"9fd6e3608ec1b79ec72199c63af77141cf7e0003","subject":"Drop skydive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fd6e3608ec1b79ec72199c63af77141cf7e0003"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 09:44:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-11 10:12:18.000000000","tz":60},"subject":"CI: drop kolla_python_version from Ubuntu jobs","message":"CI: drop kolla_python_version from Ubuntu jobs\n\nChange-Id: I757482e0c9df099197d0e5f388bbc8541aa4c651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/629b1241fc5b53bac3920e88d9a97e401a053b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/629b1241fc5b53bac3920e88d9a97e401a053b71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
