)]}'
{"id":"openstack%2Fkolla~869906","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I113b9984294cf8663d3fc0c8840320e1d40ea731","project":"openstack/kolla","branch":"stable/yoga","topic":"kolla_yoga_rl9","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-02-17 15:12:43.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-02-17 15:12:43.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-17 13:39:54.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I113b9984294cf8663d3fc0c8840320e1d40ea731","subject":"Add support for CentOS Stream 9","status":"MERGED","created":"2023-01-12 08:51:38.000000000","updated":"2023-02-17 15:15:01.000000000","submitted":"2023-02-17 15:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":64,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869906-kolla_yoga_rl9","meta_rev_id":"cd7b0c2763ed9b992b450e537290db0e4546ac2e","_number":869906,"virtual_id_number":869906,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 15:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-17 11:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-17 13:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-17 13:39:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 08:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 20:57:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2023-02-09 17:11:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-02-13 13:12:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"787f9fda1fe4a4af979865469d484d35380c1825","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 08:51:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef38edabcfa09babcb335e865c6c707f34499c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:51:58.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/yoga.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/yoga\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/yoga\", line 157, column 2","accounts_in_message":[],"_revision_number":1},{"id":"a6214140063ada51b6f74649d9c2f3f719a907c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 08:52:04.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/yoga.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/yoga\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/yoga\", line 157, column 2","accounts_in_message":[],"_revision_number":1},{"id":"f7dcd3ce79ba6f1b525f57beabee03f178f6f247","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 09:02:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5d46d8b8c6fb21a0d02becc737c76c4ec5edc14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 09:20:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d59e70c082668150a258618b5500a791a4f1bdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:47:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dd3ed0fd06b4810adec097cabd5ec10\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/877a4668f80043fc9eeb4d651d27b851 : RETRY_LIMIT in 3m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/263593e2262142c0997abb01cdb930c3 : RETRY_LIMIT in 3m 58s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/64ab063873bc43a5bb4b83fe80b0706b : FAILURE in 6m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/874511acef7a44a0bfbbbb2565c70c42 : FAILURE in 13m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6a8eea9981854d06b514ad851a878359 : FAILURE in 12m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/838b3f273e6c4761886389e9b7df1631 : FAILURE in 13m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c704a2770df048b1aa3a757247b712ff : SUCCESS in 22m 52s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5bce6dbd4a44558941eb808980a720f : SUCCESS in 20m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0c01b1a20604413b4ad68dc68247c53 : SUCCESS in 59m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/58aaf5ca7bb34ec199a4299e1583eead : SUCCESS in 1h 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89766b91c5a4295956f11bffdb775a4 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa256a22bbe44794802ff9e56b2556bd : FAILURE in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4dc0fbe95840738a69d8f194c61ec8 : FAILURE in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310eea7968f24ddf91cb40da9b398ba6 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfac65c1052343ed8eee4767ca73251f : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc2d004947dd4c3aab6d856d976643d7 : SUCCESS in 16m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bea6e918eb6d4abb9cdfe4f5c3e64cac : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":3},{"id":"bc937e16e89019cc595ad1f1cdda5b8c0bc4f406","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 10:48:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a78773d18b37d0347736d074d8ab7f9e65537368","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 11:24:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af9dec926a091bea26c57a262db64e7c1aac95b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 11:34:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"939ce1114c1b7b610ad2ad5a841599f3754e9c38","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 12:01:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a59eb586666b1e06a1d47b1e8a22ea1ce159cdeb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:05:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"890488094aa02939ef55c6539ccbe7ea09f2a9b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:05:23.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 864920,12","accounts_in_message":[],"_revision_number":8},{"id":"240c8845235c08465e3b8c657d465b85b13d3086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:05:28.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 864920,12","accounts_in_message":[],"_revision_number":8},{"id":"c3c3f1f8dd242d7437f7b0652a3df811aa5afac4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:30:12.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"93dc780899c0af03e598d7dc88192f949933b144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:30:37.000000000","message":"Patch Set 9:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":9},{"id":"8cd6efd8e2a1f8425d24197a2dfeeeb6dc9ed33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:30:50.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":9},{"id":"af99317d3c05435f0587975da4cde143899d3344","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:44:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2487537f5eba3a1317e4e469c7e80329f5e66f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:45:07.000000000","message":"Patch Set 9:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":9},{"id":"968183eeb88bd359a73a3ca987719886a3813530","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:45:15.000000000","message":"Patch Set 9:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":9},{"id":"6773f410af128ebd67a34046d175411eb86bd382","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:46:42.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d28936ccd4b34183e822ce36b7d99d55824ea963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:55:56.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07922f29edaf47b4bec8471f1342043e\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b1e2f35198b44c1cac5fb27ada928988 : TIMED_OUT in 3h 01m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/92fa8dbb434d46ffbcc77b7f8584fdb9 : TIMED_OUT in 3h 01m 28s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/5608dc0653b74cb995a649ef8a23e52f : SUCCESS in 38m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/afbe831caad84abeb5c166f8e73d7043 : SUCCESS in 1h 32m 59s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/484c7a7a375c42c3882ab8e984b74451 : SUCCESS in 1h 18m 49s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b00d1274cf984e968e9d2042f42ebbc4 : FAILURE in 1h 11m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/81463fc2d3bb4eb4aebf59f103781bd8 : SUCCESS in 1h 32m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe5747ecbff641699f6aa69d17481f7c : SUCCESS in 24m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5d9b7e25ba34e7387b025f2e33f3a7c : SUCCESS in 22m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4fa699e0fcbf4bcd97d3781b229dfc14 : SUCCESS in 1h 15m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5b639901fad546deb4e4946c245cebcb : SUCCESS in 1h 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76001b356a04028890f046f59ae354c : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ff8cd98bbe947f4a155af0e4b3cdb58 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9838becf88584b71b0c27867a4b018f7 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e95c18bb27e47b4b378eab18a4542f7 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/925df6af03bb482ebded3af679768478 : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8490df8c0c847469d20405359d10048 : SUCCESS in 19m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/278c870dbbce4512828b6aa9798f9edd : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":9},{"id":"cc3e6f2621fec48980f6382cb4d61ec1e837cfb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 17:02:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8649cd3074c34036b45418a425a14bc9\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e166f6db19f845d29819a06962c898cc : FAILURE in 14m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08dc29a94e59413c8a0e562e90374549 : SUCCESS in 2h 20m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6ac1101764c4dd2ac0dea84841f3a21 : SUCCESS in 2h 49m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a0a8b30aae8467686d33836bd6189f5 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63eb571e3ac843c0a2cbc35afd4eb20c : SUCCESS in 8m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f87ccb3687e4e8d80b53b7575bdfdae : SUCCESS in 2h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d7ea8fe72cc97c3e7387a42d1fca2510b8234b43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-13 09:14:42.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c2c9f4ba3d3f4a9a8d9007381b296d7b62aa4a07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 09:27:01.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"40c3d05340a03ad53f3c5bb28fd16720969cc43d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 09:27:04.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"d36ee240851b3cb1bf2e93c5e4ed86d4186cb1d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-13 09:27:10.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"74c1b5f9a235098bdc76568b1a115559254b9610","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-13 10:03:01.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"5a78879012c4e0b3f082022446ddc9895fc79eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 10:50:25.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d7c408969e64351be0dd4cb4b357981\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6f8240f33be0411a806858ffd7eefc0d : SUCCESS in 44m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3bf2da952b5446db85823236b0b39592 : SUCCESS in 28m 46s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/6ae35bab4f254ead8aad04af38fe4a59 : SUCCESS in 29m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/59aad8778be249a29bc9a37bd29dfb38 : RETRY_LIMIT in 4m 02s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d33def4bb8ed40ec8d29e0307ba72be3 : RETRY_LIMIT in 5m 13s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/28b6246ab885458abc138be4064e96b0 : RETRY_LIMIT in 3m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6c4557ff520844f8b303431ba8109874 : RETRY_LIMIT in 3m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/630df07451104dca865be0856c70f237 : SUCCESS in 21m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d3900c43b13b4e7d958aca98977571a3 : SUCCESS in 18m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/31a4dda5792d4296b23d78b8652d0237 : RETRY_LIMIT in 1m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1cfcc5c1997146b48b9d549b601f2266 : RETRY_LIMIT in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2048bf4856a4865ad747909c44ec3ac : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79041180634c43d8aa132db9702cea25 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e87daff795d645c98059d0b7ccb6f0fd : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505e5d112b774000bed08d5efef015c3 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4123b4dcff9740b090bdb326c5421272 : SUCCESS in 3m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/219f5127f74e4e3f900cd2fa8d388ca6 : SUCCESS in 27m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/139a50da59984e68bef4a1631974e8a8 : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":12},{"id":"bc824ccbea00dec19358f66d0e716d18a0a2418a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 12:35:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414e21cd9d8e440599d54f0fa7f1dc1d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b188eac9eae41adb8816cbdaa4e02f8 : FAILURE in 13m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f62ac92011f4ee18b89a88cbc470410 : SUCCESS in 1h 58m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a32a1730d1ca45dea9ddb3b43c4c03ce : SUCCESS in 2h 13m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/caa24d60fef54996b9b77e150765685f : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61af6d0209054badaf3296cda2385422 : SUCCESS in 8m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a132669e94d144629a6df3cc719ab307 : SUCCESS in 2h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"449585d3f48ddcd5941af61c6896526f556c35ad","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 11:34:30.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9d809d27adcb83da2acdd2d094445b231b66f08f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:57:33.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"bea2116f5c7aef40c9a05a3c1c093e1c048fe1a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:57:37.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"99ed725e9f00e4ca17b6ad61d6dcc4fe3bfbf96b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 12:58:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ef6f366d56eb68badf34f7ed122194d210151993","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 15:35:05.000000000","message":"Topic set to kolla_yoga_rl9","accounts_in_message":[],"_revision_number":12},{"id":"75873e0de0256d20d955e3e4d63dc665159f0823","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:12:23.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"68596337846cbff4c65602a5f3f00544ccdc690c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:12:29.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"49921e0fa228d56452be95fa50682a44a6c57c4e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 16:13:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"385854e6a7859f33de95386eab50f04caf798f22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:14:35.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"f8f0d08a4fd06210fe997a9b94fce2355a341989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:14:43.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"c0ddb5879db98e3d5db25013474d9c5497003a43","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 16:19:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cb5dec1d9c58408d7d43e0edcd94eeb3cb447125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:09:16.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b86c10dbf1f422899898dccb8c6b445\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6e473d0e00144f46b484e7443efe454a : SUCCESS in 43m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6055b10427884a128ad6307ac6618a66 : SUCCESS in 51m 47s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/d9dbc1dfd8f449c1a5ba94fdcf0657d7 : SUCCESS in 47m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5eab4cac575b46e99a97f54193e9a3d7 : SUCCESS in 1h 20m 53s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e085a629521e4115831e7c76174d263b : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/2213a21f9ed64df3924dae951c538bea : FAILURE in 29m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e3ca9781c1a24aeb9137bed1398faa63 : SUCCESS in 1h 41m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/db21c2a95ebc4a1692820d5afa93fc62 : SUCCESS in 25m 35s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ca6d73635e97427e8c58eba58e2ae5b9 : SUCCESS in 15m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/137cd9ce7c464a62ac972dde27948a96 : SUCCESS in 59m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/74758e85dd22416c8657297cb8fc4f3b : SUCCESS in 56m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd2d54510154f7c9cbf836104f33ede : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66cc2c4bf7c548b8bff8dfb7acfba6ee : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e13c8516f734e8d9b44b91b76b92783 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/471e2afeb5c8458a85161fc09ebc12da : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf4df276a0b1406fbc5d8a8ff5634274 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/46b3aab4123848d385ce79e08b938466 : SUCCESS in 31m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d065cc5b92d34a148672f43651f90c93 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":12},{"id":"7f97d5bf7a3df28f0f6cc22e135928399c08f311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 22:14:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a8985484b148c7b9c25d87e2e047b9\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/28e4bed7644940bba1f2b5b3549457fa : FAILURE in 11m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c10e56ef58a64ad398ee0dd9500be220 : SUCCESS in 1h 59m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/69e0b950b688479b8e01e5856389f14e : SUCCESS in 2h 54m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/018b5c0e176440d9adeae369e07ff3b0 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2516b1a69e94853910b03b69dde8bef : SUCCESS in 7m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b45cc9bf17046b4b47b7fedac8bd0bb : SUCCESS in 2h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"afb3c9f7b2a4268143db341496b2409934d48292","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-18 20:57:39.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6a9c671914884338644c833803cb3e5183538873","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 11:52:13.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b11940e530153c20c89a63c1fa074c4a7c33a85e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 11:52:29.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"35ecfe1ed9463570a705d847f8b802ca3328b090","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 11:55:08.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8965e339849ae1f84ced04a241e182c5bae382a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 11:59:56.000000000","message":"Patch Set 15:\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"e0817c2167e6c89a9b1d6a3031959fdd09f7f188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 13:43:31.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa3034e34eab497eaf7d13439fef6721\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9ccaae01333a4b24b1ab1c1bcfc33bf6 : SUCCESS in 30m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/025ca7f20be44bfc82421b5ff155f371 : SUCCESS in 32m 21s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/43c1edf51a9541aeac6d3127e0fad6f9 : SUCCESS in 28m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fc516dc5bd754e10a15d89623f97028b : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/67d64e93c8174d29a5a75153ad517de5 : SUCCESS in 1h 21m 53s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a05de1bc37d54e7889ffb0eb587c4e9d : SUCCESS in 1h 21m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7102464092704f24a48c5059c4ff68f3 : SUCCESS in 1h 10m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c7508630e25344c1ab65230f82762497 : SUCCESS in 15m 09s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f297abae9a641b7a49dc2922464eb9f : SUCCESS in 15m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eb54217b07ee40e3b036330175894188 : SUCCESS in 42m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b3a851b18dac40e3bf7f6345fac0516b : SUCCESS in 58m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0c1132ee8e4e1a9b04a19804b39aee : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc5059d5dc44a86844732cefc96be6d : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6b53a9901764d2da73cca162205bb58 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3674ce8d504535bbe4221ea8b981fd : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a23e9b5a29b34de1a6cf4b5184d0841b : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9a630dc7bea4949ad4b7f0aea959bec : SUCCESS in 13m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e18dd1fb4a348b78999b439cb346027 : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":15},{"id":"f0180d1907a2c8b6b7ec25b57cddec29e76830d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 17:56:48.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70807084fe84e1ba41b66d3a2b8a322\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/264b240edf0e4dcb8225ba62e36df864 : FAILURE in 10m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d3afc20be3a4ea096efa6dd29e83eeb : TIMED_OUT in 6h 00m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cb58512cecb497eb123bcd3023ae001 : SUCCESS in 2h 39m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/074a8305df824bdfa1e164079226cf96 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c63aa53399704687898f11ba78ce99e8 : SUCCESS in 7m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4bce9425b41546deb79e7d359a602a91 : SUCCESS in 2h 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"791953a633bff281235a07a80c3ad05dcb439140","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-24 14:51:01.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"36c7df8abbcc3bffc9d1423e9f6d4baa0aafe352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:43:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79b56c336880425c8d50dcbd4958c458\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/341d0d1369ee42e78c6bfdd77326d884 : SUCCESS in 33m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4d2a61700c0c48f4b0f98556edf6d828 : SUCCESS in 30m 44s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0fe58e9f6763402d98bf2f9e56c8cc25 : SUCCESS in 25m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1ec563e93a1148ffbc304a3f10a914dd : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/28a57a47b4174d49b6a0b8c534dae280 : SUCCESS in 1h 15m 05s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/6491880fb8bf4381ac6f97ced67d6166 : SUCCESS in 1h 27m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5d3fe9d68a504207a41603906f881423 : SUCCESS in 1h 49m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b27ae7d3a2cb4de69deae7ed6f3b301d : SUCCESS in 17m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cbca26c6b2314d789e42fe3fbf364f7c : SUCCESS in 15m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3aab899f425444b2978a74c08c8418e6 : SUCCESS in 1h 04m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3e7bf66a05b545cd829ef70b6602d49a : SUCCESS in 1h 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01451e33a84b43d6ad1f73d3baf9204e : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/489457c0f0cd4d71a5a685fea706d4ac : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82625147e29f415dbf5afc5e4684c20b : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bbdfda15363471084a5cc36c6007a56 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da5096d7c6de4e6eb4c8294fa565d222 : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7b0af58f1144f64a815a1fc13757690 : SUCCESS in 20m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c96b2933792c47998766784c009b9d29 : SUCCESS in 22m 52s","accounts_in_message":[],"_revision_number":15},{"id":"a66fbcc686e2047e8d565830c369721ee9459cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:30:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04e1fd70032f4a37a31458bca52443c0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/937434d3f82b438db17eb3b0ef314afe : FAILURE in 11m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6232bd78df14f44a4fed31181f9d992 : SUCCESS in 1h 33m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/604894ff2d46424e83d5c174bf4d1491 : FAILURE in 2h 35m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea5e10dcac304a26a9d4ec5c416604b7 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e626d773f0b24fe4a3b595007f7ef93d : SUCCESS in 5m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff5eb49ccde449a0a3eedce3bb1e7dfc : SUCCESS in 2h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1bca98e072f2065d62761361860eb789433eb442","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-24 18:13:33.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3fe23a7d04b9eb97edd050c4330a56410d5bdd2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 20:06:38.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea78de584d84bdf8d6f1525be47741a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/17443372e8f648369e5011a1815e01f4 : SUCCESS in 49m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/558e8f8d9da4432e82d7f03ed403b9d0 : SUCCESS in 52m 17s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/608be16153314a269e220c000d79ee9f : SUCCESS in 54m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9a45e43982b74f6297cb8a19282b31c3 : SUCCESS in 1h 27m 45s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/24903b968d7849ba96625e337834141b : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/9b298b5ba1094e00b3b95049a7e68e7e : FAILURE in 10m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f4e86137154743f1942f38d81966e357 : SUCCESS in 1h 33m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2471ebadafd64324baaf8a00cdb206ee : SUCCESS in 20m 54s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49eb33905b8145d08c9a755b89954a39 : SUCCESS in 17m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c4feab9b6dd14a9d9522b720440cd503 : SUCCESS in 57m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6c4f6be58d0846d6a2aed5cf428e8094 : SUCCESS in 1h 08m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3472d8d24d4699ac65eae7a56308b7 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15f83bd13f3345d59c08d3032b091d8e : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc39c730219b4f92b23e6cdc5edf09ad : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11ee8d80f984bd8a4ea5b8389cd1e18 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5f358b700b14d5a807f50ed401cf331 : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ccf7f6fdc81e4c489284490dced2b24f : SUCCESS in 17m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/46156c4da34b45d9b3bd992a3e1f0911 : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":15},{"id":"eca5e43586e2f8a6ee036f1bfa48eb7b385b4bce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 21:03:46.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b33a512d314effbb61065460c054a1\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/147eb7e9164d459799b3e360c4f95047 : FAILURE in 10m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4496b886a60d483fb571633974465471 : SUCCESS in 1h 48m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04dd695898f041e790b9c4bcc1fd2c6c : SUCCESS in 2h 27m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1483a28b658c47478131a71ca4d105eb : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8047869b32cc421db0faf43cdbfb3923 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c654fe5240c34fa0ba8ab3adb56dcb54 : SUCCESS in 1h 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2e659d8459208cbb7fe4f2cb24650a3706c8754c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 09:41:38.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bd25e11f2f5df431eb6dd3a79b5b3d49f83cb264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 11:07:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f74cc7353145b1a58a5271cc05346f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/39fa97fe88eb4671a6aab0498bfe5d9a : FAILURE in 5m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e1b0b4bbd96410a86f1f3aad7bec3d4 : SUCCESS in 19m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1522b6266ed47c9b6383b51e8575dee : FAILURE in 1h 03m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44eac086e96d45d9863bdf21cb8c9de6 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28152003c9824528bc9326e61d63e0d8 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd75147eea294510932a6a4996252f6d : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b23e1be9ac17ab2a0d577d7e54219884930462b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 11:19:00.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"c06476744eaebc01ae28240c3a3dd4997bcb9a8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 12:50:45.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"4b6963157437d2451cc23ef51d1bf550011b1ba6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 13:54:26.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"bab350750e37c626796e86b4f0b10dd10ceaa4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 15:45:04.000000000","message":"Patch Set 18: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47333e87244245b7b569ccc23139c276\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ca3515d2f4c045b596959e64ba5e0fac : SUCCESS in 42m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/404fc14c4e5142749874220c27d04599 : SUCCESS in 43m 09s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/0f39289396524156aab1d5fc611179f5 : SUCCESS in 30m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8b86f7cc01eb4ae1b3a19938d2e130b9 : SUCCESS in 1h 25m 07s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/98edeebe8d134196a3e56476284eb851 : SUCCESS in 1h 40m 16s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f0989f468ec04b729616ff2f331dafdb : FAILURE in 15m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f1dcf318d40c4a96968089def2b3aae7 : SUCCESS in 1h 31m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/437ab46bc01042829c46c6f388594703 : SUCCESS in 19m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/51383de21662406e91aef142f86809aa : SUCCESS in 15m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd1db070b5f849c7b6863db8aa8137a6 : SUCCESS in 59m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/94cce7f4aa0843c1944137407c4fe1e3 : SUCCESS in 1h 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8388fbf8b06a4b948ba1f784c9dab1de : FAILURE in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01f56b69fdc140dfa4486e5c8efa73e7 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d99cd1c4863d41d98fac4019181d60e8 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec3b62c12e845edb6740e93a411a509 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb85fe0375454ca7bf6d3365808dfa8a : SUCCESS in 3m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de6df869f67e45a38d369ffda10361ee : SUCCESS in 16m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd2f078c43d54ce392e6b957e9b0bbf0 : SUCCESS in 22m 47s","accounts_in_message":[],"_revision_number":18},{"id":"f65e52a48491322c5e17cdbbb4dddef321871622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 16:25:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ecb3462f2464411a907b4bd044718b8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/50568ee6a898444aa935a1c528a3f245 : FAILURE in 10m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1c76aa6e34e4f16bb05f021e2b97b46 : SUCCESS in 1h 54m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9bdebfe70d4d469ba7524be811a90d83 : SUCCESS in 2h 28m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cc03f218df34a58aeaf9948895dc2fc : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77af04f837a04b578713c81317afc398 : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e4242af7a5240e8b445fb045cf2b3c2 : SUCCESS in 1h 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"669f8c3936e0c63b9cd116f039cabfc02c7a294a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 16:32:43.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"d931a8c3a3d316fa82beb46ad71c9a823d3eba77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 17:32:52.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e73d9e50e23a49af849f8b1cfa8a2c94\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bab97f11426b4b65b9aad26c1697e1af : FAILURE in 7m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c3e44f486c846dd9d7bdf73ef630813 : SUCCESS in 39m 21s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d80df25c25f4176bf58ff6cc5e1c508 : SUCCESS in 58m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f92a8e44ba7c47698770976d1e506231 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf1aee7b701a4416ae8dee9739bba9f8 : SUCCESS in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0eba0bc05dd34cddbdd234f3934c9818 : SUCCESS in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"09069331dc3eb90570e1c5d61ab1deb92491f011","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 17:44:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"1a51bd938e1557d43491845799a5607b082f98fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:01:07.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3806f266652e43f9b26a0e864e5b1307\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/929b7c2d333c4016a572dfdead067836 : FAILURE in 3m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b71816d3ea44e8294c423be8b099a2c : FAILURE in 2m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/553e6f608f03439eb1cde585818d5548 : FAILURE in 14m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acad3edf4fcb4272844b3cc8f38b6586 : FAILURE in 3m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0a96cbd3ed840ef9c74294a7686257a : FAILURE in 3m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/196efefd34ca467295323531f297eb00 : FAILURE in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d901cfea810706130286a25564cbad42bc0abf79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:08:27.000000000","message":"Patch Set 20: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b823e96086674c75983aefc74a6c4da2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e21dd24fba86403ab15b57081d562143 : FAILURE in 5m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/42b767be9f1f46f3a80d8018fc7fa632 : FAILURE in 5m 52s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/3dab3c8125e34bf8b1583aa0fd5c316d : FAILURE in 5m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3eedc18e30fd4655b55c8590035a2feb : FAILURE in 6m 21s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/33e6087985f74aa4987645a92380bf98 : FAILURE in 6m 14s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/05927775166d4aa785fccf8c982bc3fd : FAILURE in 6m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9c9d35899707408c874f4556e0e60188 : FAILURE in 7m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/804ec1df7f5643b28a2d6f1962090985 : FAILURE in 2m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e834e488b4e84cdea8daba892fc2af8b : FAILURE in 2m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac122ec00a124420a84d50c9b6375754 : FAILURE in 11m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a210e58e262848ccaf9ce09b279e695a : FAILURE in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee80d797f9e4a879513e2c593e26168 : FAILURE in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/367c1e67cd304cf8bdec190c33e21673 : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9120402ade464bfdb012b3ce99622c07 : FAILURE in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c6802765ed407d90010b1a47bbdabf : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50982e18fc24338817341b7770e232f : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/623d975d2bbd4ebb87259d521a807762 : FAILURE in 3m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0594f58515014f2993aafd8a6412983b : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":20},{"id":"ea90c402a4abcc64701c57fe4084f3c06ca9e94d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 09:37:23.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"853f9ccb2e73baf8471a4088c3ae5dd3aff5fd7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 09:49:22.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e6cc7c7ee34acef6578ce48bf858faa710b1cb2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 09:59:25.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedad84eab7e4e808336e5e9422b8496\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/95e06bafada145678f8982178c7d5aef : FAILURE in 7m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42b3aff8b4ca4c10a0b382166c4e63c4 : FAILURE in 4m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/60ae6c55c48d413f9e7942efb9d91ed9 : FAILURE in 7m 39s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b648b9abe6914f4fb02c52f2cb1e227e : FAILURE in 4m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/527c5ece1ad64731b856831c210d9264 : FAILURE in 5m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b41d5dfd99b0406790a5b483d50c3f42 : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"95def7b66ae242da918ef029ccdbe69a6a0cf7ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 10:07:26.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"cc8d45fb90bb84503065a7ea1edaf062750f7921","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 10:20:39.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ba0643d9c910bc418809f8638a1ed89dedc76e36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 10:30:40.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"07959c34044d90f2e15c4f81a8ac08b4e82f3af0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 10:46:22.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"15c36ef9e1b6a636e32a68dae9725223495ffa68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 11:28:48.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"b86d813406c9a9729963ded03fd2c7270b749114","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 11:56:32.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"3bc68a5de37b03241e48384a0b406fccf81a158b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 13:03:10.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"a22145bebbd511bccdfd04ee43e6160d5776ed6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 14:05:46.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"1b73a235938d3f348d92e2951c7995f976afef41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 14:20:35.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"23f45c1385a4f497166be6fcfb1f5f02358764d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 16:24:46.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851dad0014bd47afb6dfa3aa4bba636b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a85e7af5c2104ac9aed72c2da195c6d4 : SUCCESS in 55m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c31055d6024f4229948b6296f5ec2a95 : SUCCESS in 52m 58s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/58213308e9bf46d485de069650f5c079 : SUCCESS in 51m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/417e57591958485b9e5051d770b878ce : SUCCESS in 1h 22m 01s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c5b640ecbca14b49b33866d58f17cd10 : SUCCESS in 1h 57m 40s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ef4bf0d689fd42f397d55885ecc32743 : FAILURE in 59m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c87298cc91bd4cce97295217e2723655 : SUCCESS in 1h 32m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d722e6d3a6214560a41f640dec394e62 : SUCCESS in 36m 21s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f03228bf79d481b94d62f3a6690885d : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6775bdd6dc243c69df8b08c7e303b3c : SUCCESS in 1h 18m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6b0f44d665114a9da5e9a88ee9fce8ef : SUCCESS in 1h 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ac2128f6ee46c2a8da44fc250306e0 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad63433adbf4b419d155740bcd1e21e : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb6c061b86b4fc8bf9808afa4bdb710 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743f632c8ddd4394b42761cf2936ff2e : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3badf7807cf341d7945425ec6dec65f3 : SUCCESS in 3m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bebd055bd7644fc7996c0dfa30f19b71 : SUCCESS in 18m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e5199d49e4e427491707b2c3d05f4a6 : SUCCESS in 21m 43s","accounts_in_message":[],"_revision_number":31},{"id":"c6bae852f2067e421a29e4e111b62c3962fe062b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:06:23.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d63ae604a5b5433faf9c951204e8a4fd\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d1850084bdd4835b599180a44ed960b : FAILURE in 11m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b09f3618b42b481ebaae36aeddc36faf : SUCCESS in 1h 44m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba51390f65094ebc9bd2b8495694312a : FAILURE in 2h 43m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/652bfd4385e54aeb9e462968f225ef49 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0f9369ff8eb46499740df69e6970b7c : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ee044a02b6c48b0868e330b71259607 : SUCCESS in 1h 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"61f8e3ba2a180e545e2e60334d01a31e1bf0621b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 17:45:06.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"dfb088ed873bd1d8260f9dfc0d9770f62233954b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 18:10:33.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa3d19600a6542ea925f938811467bf5\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d47f32bddb74e74a7ddc28b2247cbb2 : FAILURE in 10m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a1a71ea684f481296e503d8c901e95d : FAILURE in 8m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da4fce347cce42d088192cb2f384f074 : FAILURE in 22m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8640294d57b4e76a0048b26aca1e3b7 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fd48e2038204b5da25ed8b3a3a7517c : SUCCESS in 5m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ca6cbca0b0b40eba501a22e4ed3ce07 : FAILURE in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"be7bdfbab891ba2c0c4a01d341d6ae63c1d8cc5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 19:09:48.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a49064822204517ba8d5d54322a1916\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/46a235648da14f71bb1ab6ef28c011ba : FAILURE in 8m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/21fb8b8f213b41bd800ce044f4b67dd5 : FAILURE in 24m 17s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/777db044d30048aea17efa6bcc8529ad : FAILURE in 37m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1c0724bde03941999672ed2b0c984aa8 : FAILURE in 54m 46s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/37202b8b992e4b428dcbaa4221c6ec7b : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/b6bc16ebdbef46fd9b2a97f4172b9a07 : SUCCESS in 1h 18m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/26699b87afde4503a7e0e63f6f63e83b : FAILURE in 1h 11m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6a02e19d24a4355a0f79f327471f02d : FAILURE in 19m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0aa6cc78c55243b184d579ec8a685c3f : FAILURE in 16m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ed30b5444d444b387b1fd0c4bd2f99f : SUCCESS in 46m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f57be16acbfc4f7182142fa8fd7eb095 : SUCCESS in 59m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca96e9bdb754196abcab3589943ad4a : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984fe292a5a74a4caebfa7a6f048ee6d : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f0d78e450947bd9079d270f5cfa818 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04797ef9954a4ce49649b2e637be70fe : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d47e2ded2814571b93980cc3f78e7ce : SUCCESS in 3m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14e8315e2ac64c699b5d5d9a442d6924 : FAILURE in 16m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9892c9b0a79643018316e2e859357b8f : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":31},{"id":"8e19cea158df6bb38324b69b02235c633fa65c72","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-27 05:34:03.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"121ff10b8bffaf6b8d7f32d1164ece08cf0a2acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 07:02:00.000000000","message":"Patch Set 31:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2115ce8ddab439ab1bac7c697794730\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0475555378b47eaa4a56ec500cbf5da : FAILURE in 48m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/81c9ef6526a24fcc84f39f28cbcecdbe : FAILURE in 50m 16s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/2c296c1e4d3448a5870f61cd67966be1 : FAILURE in 36m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/746da72f4d8b449ebe5845121e35ace3 : FAILURE in 1h 04m 44s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/be69f0c5c66f45e0ac3cfae6baa0bc4c : SUCCESS in 1h 19m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5b723e3bccf84fa8a3294f529f2f375f : SUCCESS in 1h 15m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/294188edf560454884a49864e845566b : FAILURE in 1h 12m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/792315d48bba457c9f7532723502e99d : FAILURE in 23m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9613dddad4cf44659e7483cc39b30e12 : FAILURE in 18m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d3b8fb8887104dd1a31d4b48aa2df00c : SUCCESS in 46m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1773e653023b4ec590bf9c4519c3125a : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d010d1cb154cc6a165f1c7f6eea989 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff759c6ac95144beb9996f8d3f43f630 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b84c8fb245749319a74bb5f5caa6215 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb621d08199e41028af9488dcd4e73c2 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5651ac9a94794a60a8ecfd28b7a22f1b : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01ef2b9c84824733af22e14d5b8c726f : FAILURE in 19m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae125fbb10114f069a2cf5fc26e30381 : FAILURE in 48m 07s","accounts_in_message":[],"_revision_number":31},{"id":"362c1b823101f9b6bd0bde354043c7990f887751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 08:01:19.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41505c4657c849a7b1466c627c18817f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc87bda527f34529a6b21e4c6db54e9f : FAILURE in 12m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85804d52b5a44bfab1352d428d60096d : FAILURE in 2h 07m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f318480065a54e19abda567a7b606e9e : SUCCESS in 2h 24m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71b0b45f046442d5903e6ce9076e2bcf : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0839006ba81e4354a870de02f8869940 : SUCCESS in 6m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e042eac4be694b15bda2472d4a1f68a2 : FAILURE in 1h 59m 01s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"cba2f2c876336a7bee55a907dfb4bc572113be4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 08:29:30.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"c55a0af52497b98d6df9a7901927601151f61389","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 08:29:36.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"87bb65dc01557be3646154f7263d4f7e1f648935","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 08:29:59.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"71e2f74da5935939f7230a13aa2aa5c45fa476c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 09:41:01.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5111fa4d80f14da9b4db971e19d90af7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8add76bd51394511854dda3493e12bb5 : FAILURE in 4m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcb4d4c6f7834bea8daa8f55f9350ee4 : SUCCESS in 18m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f25ea71945374f0c9dbdca6a23d3bebd : SUCCESS in 1h 08m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/753166d29d7a456f80b95e7cd1f7ac75 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/383d4388a5604e09a19c0ef5c802c40f : SUCCESS in 4m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/57d40233c6d24094a96858b8c644b217 : SUCCESS in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"3c429d7e6890d5450deb1bd443f3777d789ab8bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 11:46:28.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8589522e45f40fea73f4863ecef3f0b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4fac4ab979aa46df8984ec267ea6d0cb : TIMED_OUT in 3h 01m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f5c36aeff2534003b5b06c736ae231a6 : SUCCESS in 29m 55s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/004a046f98764920975a277f607209d9 : SUCCESS in 26m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/09bae9af01b54dd4855b7a0acf32556d : FAILURE in 56m 11s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a480a2e23ab14f37940bf4267e5fdae4 : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ed428688dde54b31aeef824d9cea58c7 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e6f9e353b58a430abd84abc862215030 : FAILURE in 1h 29m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2be1dff7ce604231a0677c8076c5e212 : SUCCESS in 18m 59s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/374432d4bcb646bfae4d6b1a52c87d3c : SUCCESS in 17m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/83328c25f4ec4f47a32486010afd73e8 : SUCCESS in 47m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/335866ac3436420aa71c47bfb3c43eed : SUCCESS in 55m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4bebcff65143e6b560c3675ee04d98 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb53ba593e0f4eeebb78b840772c1e20 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be619ccc5aa648efbdb1758a3c77b13f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad804c8c19940b884435fc9c6c0879f : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/801df9dac4c346c1aec84b47269c3c3f : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acd7cfa2576542f9bf34463fe62bc434 : SUCCESS in 16m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbe9c9ca9f0c4e8191af59c998d746ba : SUCCESS in 22m 07s","accounts_in_message":[],"_revision_number":34},{"id":"832651a3676fffc27a7846eb500bb8a7b9b9eb9a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 14:50:42.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"cdb2efd9114f6b533349a3f7b5c1d68a5ad80654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:06:24.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d4b870ccec4760939766221f40712e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8172eb90abd4897ba2812ca3862a54d : FAILURE in 5m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b7fae6a2e4e46bb8f080da180a46c77 : FAILURE in 4m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e97ef80795534521b366606d81720ba5 : FAILURE in 13m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77b19df10ec14948850b52a6b36ad8f8 : FAILURE in 3m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7285a8be766e4611a842fd51d98fe148 : FAILURE in 4m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/98b559a6d2664c378f591ca5f2fa7ff5 : FAILURE in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"3d4d5f4143a0cbea37178e44a9513dc13b156e98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 15:07:20.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"076061d4973e666ee8656efe842348a8139809a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:30:24.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c0a9b09016c493baf4e4862d05c3bd8\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/61318013557348c4aa55a5b629a634ab : FAILURE in 10m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ea2b44e9a114d9a97424298a2a76ed3 : FAILURE in 3m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e15ad1d8d8342a98b43da1595299ae6 : FAILURE in 19m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d88cbc1be38c487ca7e63d446711a7dc : FAILURE in 3m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd9702ddb9be49a6855af4c563d1c489 : FAILURE in 7m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b38b240307343b598a1c1f713b3212d : FAILURE in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"f270130cbaf20da5558622af0d93a7a803c86aa5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 15:41:36.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"35a1cf63690503a146e2d975e23f03cf302ece7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:12:48.000000000","message":"Patch Set 37:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a690f382635435199d7cc3556a46106\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/74a337a6529d42b4aaefe251dd2fe1b3 : FAILURE in 5m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6624212bc6504ee793337e6e08e52206 : FAILURE in 4m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d373cd95c2744dbf92c429e238898896 : FAILURE in 27m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df85aa830d96431781e3e3ad897bdd09 : FAILURE in 6m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06ccaf083fe64194bd1d0be4f2cf0cff : FAILURE in 7m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2179227cb86a4e84967b9660dbf04670 : FAILURE in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"8727463f5477c66c4d243549f8d8514254b68a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:25:34.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f21451f127264f378e13481bf62ada58\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/592ed3695bd54b8b9e50073af7c4069c : FAILURE in 5m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bbe7b17981d045ffabfb99ca63136c0b : FAILURE in 5m 23s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/4fe598393c0c4892b67d313e8884b825 : FAILURE in 4m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/75c9a990020b452f9a3b7e7f251cf0e2 : FAILURE in 9m 01s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f266a55d50724e34aad4b730bf3fa6b2 : FAILURE in 12m 28s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/21e530dca21942758b167063ad6da2d7 : FAILURE in 12m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c732bd60dbc741d9b36359cbb95885e5 : FAILURE in 11m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f3a8e7a3f354cc7b3930293d4cb9abc : FAILURE in 3m 42s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d32a50f15ba04ea49c33719a52305a5f : FAILURE in 3m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac605ac10bfd4e9d946e0776498a12cc : FAILURE in 13m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/44f77486351440a188db563addf094af : FAILURE in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d24980dcac4b4b8b32d3d0ff844b64 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/842ce40992444b5988e6c2e808473dcb : FAILURE in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/052d0283e32f46dea241afdd36bcdce0 : FAILURE in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92290ccaaaa142b08a5a5d349dd2e3a6 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df0605bfb41546e484b79bfd531405c4 : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d221f565894844ea86b04b00e6ac1530 : FAILURE in 3m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ab81789f1b904f63a5267a9fafa4d964 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":37},{"id":"44e6bcd7b307ee916f1d77b4825ddce9dcdb4c4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-31 17:24:38.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"151ab206887de83ae374756a9227b3a0dd86e2f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:47:33.000000000","message":"Patch Set 38:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/427d1f6371804b0c83d5b5ada3e65861\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/163305f3e8624d3b84ee680f08af2dc5 : FAILURE in 9m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/24b80e6caeed482c9f6d4d0d8a6ebfbf : FAILURE in 5m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f85e3a4e51f244e69f7142cdc88bbfea : FAILURE in 20m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5345e55d1c7a4a3eab0dbf7a585b62e7 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2782d5eee4f24bd7a7be488ecd493107 : FAILURE in 6m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/56dfc3a53ad64712a2cf6b51839b1985 : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"7e77863feb979656e35cbd4de468839dcfe18b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:55:42.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88f50830adf2409b8639e3e00eac3adc\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a44cbd9c0214412794d79f836663db15 : FAILURE in 6m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7fad40185b32467ca40234120c07ac48 : FAILURE in 5m 33s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/21b268db08b24f6188ffb00a7b95941a : FAILURE in 6m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bcf69cc858cc4e3b9c34f0e22bc49632 : FAILURE in 15m 54s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2842e7fe9c9242cd84ab7ee77eba1ba4 : FAILURE in 14m 36s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/d9d3491099634b41aea8a66add462572 : FAILURE in 12m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e709d2c9c807464e93df389aae3dae68 : FAILURE in 14m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2e99b7178f7d43989b25c0e70afeb8c4 : FAILURE in 3m 13s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/11a8355d761f4c27ba99d37113d19718 : FAILURE in 4m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6a33cde04dc24a2dbad850e946183174 : FAILURE in 23m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8a8b6262b2c44c58bfd0397747c2b423 : FAILURE in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5fcb759823448bb37210e7659c1c80 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee4eb7aa03db468fb6740b1803837e77 : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d581ee072349ceac6abbfdc1d73b56 : FAILURE in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6053711b68444f509e1be953af7f3168 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2dd0e16ded94a3484f1ab700949c76d : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbc54b9f9d314423ae3d8cd59296e95c : FAILURE in 5m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ecde2217d96a4cfdbedb43b0d01af88a : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":38},{"id":"59e585fcb63bd2c793452882236b66a6e429953f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 07:57:45.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"68d477cdfc21388a5ac051a2ddf1a5bf524f9d37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 08:52:05.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"33c84734ec720bab92472e72dae765a4c56fab25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 09:39:39.000000000","message":"Patch Set 40:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":40},{"id":"2c2a82ef1ffdec4de2eab8e4d0b863f12080d384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 09:39:39.000000000","message":"Patch Set 40: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":40},{"id":"8728574e856b7d51d595ec63928be1e2fa59f414","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 09:42:52.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"9d26bf18750d0f2c63a8ef403e2207e76ec4cb84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 09:46:40.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"0965f1395520b746b7bf28d8fdf196f0a2718b2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 10:09:21.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"c5e7f997808ea9514bda715efecc7e3d862d0073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:34:45.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f612117c0e85458d87e7fd2f80737d36\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7944af6de2e469a928a85a2940725af : SUCCESS in 48m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6cf940fb2907489aa31cb758382dbb5e : SUCCESS in 59m 55s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/775d572105d94d7584f25ef733aa2986 : SUCCESS in 1h 02m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fb0f72dac7ec4bdaaaa47b7578da6e28 : FAILURE in 1h 01m 27s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7502dd0fb25e4381a241b5c133f354c3 : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/4f57b922aeb2459db7afb18a4f83fd8e : SUCCESS in 1h 09m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4075a8e0306d4c789c384444cbba8970 : FAILURE in 1h 22m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9182ab016cae4c53bd6d9de123a4207b : SUCCESS in 47m 39s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/58dde5e1caca4f22921509ce3e0737d9 : SUCCESS in 15m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11a6ba414ca44985abc11c1998d5134e : SUCCESS in 57m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8c60036c5d9b4c94a667650e483e7b62 : SUCCESS in 1h 04m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfa816e0faa40b3bcea3907cdb2ca5e : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db67033070324d3493c6dfd4c5bd2657 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbf34327f24e4cd58710ef5b06e9fbd4 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6cfcf260d440ada874023881cf2f45 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6541e247881473d92c4d5a3287f36eb : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f697e7cb3fac4d7db009d07fec70cff4 : SUCCESS in 17m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1bdc859bdfb84f54acc32016ade3fe06 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":42},{"id":"d50cdf60bcfb22f007709708de68eafae7363632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:41:40.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2182ee7befe14d7da745b8c6c62ed554\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c049b496cbb54b7499b8626805e18882 : FAILURE in 7m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1542bd5b52048a786f715b0d9d2c661 : SUCCESS in 34m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0e887a3260449bb81d45386cde46527 : SUCCESS in 1h 29m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa9fbfee26be4bd0aa8875b6708a5cfb : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3084d62880ad4fcd9142dc2339ed7c4a : SUCCESS in 4m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/949a1457f50244e28efddf4708507f54 : SUCCESS in 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"c0e75ad03b09ab726a2819107f7a833961e14254","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-01 13:24:14.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"83c67d6d13cfe736b4a501e264103dfa72f66fd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 13:54:25.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f28d7bd38d44c58899f48dec3e6097\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/94905122ec144d95ab17aa8e2f928c14 : FAILURE in 7m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f20dc68800104c8c9e11b0aebf7468f6 : SUCCESS in 27m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/60bdaf54b5ab4ec292634ff7c1c2f5fe : FAILURE in 26m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9dff73c03984dd585679db79c052ebf : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67321c4b41ca4af1a7eb6e845dff75cd : SUCCESS in 3m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8151476430e2438498d651cc44329f97 : SUCCESS in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"5162143d7752dcc0ae594f068c9f8b249110a661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:06:24.000000000","message":"Patch Set 42:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":42},{"id":"20e90a33081166007f1145f221bd4335decad7b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 07:30:58.000000000","message":"Patch Set 43: Published edit on patch set 42.","accounts_in_message":[],"_revision_number":43},{"id":"e2d547b02dd3077b61a75248532424642fd719c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 08:34:32.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2281c8e93dd4a5b86c8e71e3bb8d53f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/28b9ce2cd74d43c897e2f70bf171ce1f : FAILURE in 9m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6dbadd2d24254d2a82a5c7ef236cb766 : SUCCESS in 27m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aec9909b16204b7abfd0e1b1849eb9b8 : SUCCESS in 1h 01m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a84c91ff0ab45369bac53c73402c97d : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8eec5dc0a99c4798ae4c92b7f61d88f8 : SUCCESS in 3m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a686bbece6134d71a07d35f35fc8f0a2 : SUCCESS in 35m 38s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"8945ada2e1cca0160ee9bbb6676797a579e436f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 09:28:23.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b775db746c481bbd24ca8de11a0a67\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b855e80e59b54e02935654c6cf21ae2f : SUCCESS in 49m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6fbde1b62544d89af134bf75cc5b938 : SUCCESS in 1h 11m 06s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/5ee777719b274318a72e59a307a9e8b5 : SUCCESS in 1h 05m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2f50cc9b72734265b52c72357cd5bcfe : SUCCESS in 1h 28m 08s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a99213d91094b6799dea1f81177568c : SUCCESS in 1h 22m 32s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e6cfa293b29a459f9da25f7e6a33c642 : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/502ee592c4184cf3aa2b7e381f020bce : SUCCESS in 1h 50m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1019525b509f45bda10c4b34583abe9f : SUCCESS in 31m 03s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6076a649d2e642e5b8de9e1765a5932b : SUCCESS in 20m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d607d88e3a574433a00ee2173b79f8bf : SUCCESS in 48m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e18a80a2ee494dd293252c52c01dbdd4 : SUCCESS in 1h 06m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c3bc284bf945b9b5bf7d09148c2c7a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d653376e997490682a2d8976c469aa3 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13b3565172924bc7b2261df8f87cdfd9 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9764048a61e41bc81a7b7a4f4069fb3 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61a5e0a20b8c4d9ba3374c9d9ee12668 : SUCCESS in 3m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c711c17226ee44118283e210f19c5e54 : SUCCESS in 34m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8515ec4906d744368d03455baaa3baad : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":43},{"id":"deb22905d27269abc238210b993ce49babbfaa52","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 10:45:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":43},{"id":"9b275ad8ab9603dc6a96eb18803cd055cfa46f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-02 10:46:59.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"7ee47512670fcd61142518785c18d3e4983d1d95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 11:57:07.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7241ff5836694d4e970f3f8b4eda39c5\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/05190158122c4bd3ae5d5db861722c4d : FAILURE in 7m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdaf48cbe4a04f70beafc79269932512 : SUCCESS in 31m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f42dd9471d648bea99d356d03143f6c : SUCCESS in 1h 06m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2a12010e43b45979cf89946a61a79c0 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a99e6e13f3e45a982c1ed0cfccdfa20 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c921b0e14e0e4f54afb6ff63c4c865c3 : SUCCESS in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"dc0e71b0006f1d10ee31db7775ff00f34c4cb20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 14:10:35.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f62b7d1877f49fea9ead40c8285ded2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/98f6bb51c4fa4cf19b71937803d0fdca : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9451a1de0a9e47e2bf3b1738a4e7f79c : SUCCESS in 51m 08s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/e9587404ce444a91a91efdb18b08aef2 : SUCCESS in 41m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/372cdebef4434bfaa64c05457d221066 : FAILURE in 1h 04m 08s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/270eaf9672c84c26b66a00877c96f8d6 : FAILURE in 57m 24s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fe4c9407703746c684028a3ce2527c34 : FAILURE in 54m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4aba76f283e64f73bb4975c6067468f6 : FAILURE in 1h 13m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1e9379749ca405680fc47727319ff16 : FAILURE in 3m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/55406dec381243de99f675bfd9d48f87 : FAILURE in 3m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/66d137e5671749ad85a5339255de3184 : FAILURE in 13m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bcca63a1205c4d86946786a153a68dc7 : FAILURE in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425ed13f964b4271b5e3a8bc1de0abbb : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa46602752c459a9ef6fb6cc7b36883 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20a831abd262475e807cf16bfd223609 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1d1fb4b32b44658216e1e0c2f99828 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1213d5152983492d95c982c55f7fa123 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42bce882604f4932833a2600cbaa0435 : FAILURE in 4m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a97fa2347858408e9cb07d76f83a33de : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":44},{"id":"bacbbe6d7a64b5e57482643efaddcaa127c66fc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-08 09:14:16.000000000","message":"Patch Set 45: Patch Set 44 was rebased","accounts_in_message":[],"_revision_number":45},{"id":"6bc55b56ff2a439270ec3a3b1e72148b2b7b9b2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:14:40.000000000","message":"Patch Set 45:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 864920,30","accounts_in_message":[],"_revision_number":45},{"id":"16ef283bda94caf7b1f07e49b93ac34758147332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:14:50.000000000","message":"Patch Set 45: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 864920,30","accounts_in_message":[],"_revision_number":45},{"id":"ff5246236e5819522e2e761e181cbc662d2900db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-09 09:39:25.000000000","message":"Patch Set 46: Patch Set 45 was rebased","accounts_in_message":[],"_revision_number":46},{"id":"b30d25ff3c8cf5b05b6da91bd418066b973e0993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:18:06.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bbc40dc427d478a821a2008d969ad22\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e1ceede4b90847aba3dafda3db131ead : SUCCESS in 32m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e648515004be4345b2c2f55c5f6d2c95 : SUCCESS in 32m 15s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f1a5db5859b246268ae164d963f9854a : SUCCESS in 30m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e159534d0f144bcc9a9422dea9aad6ea : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0c4e3b2fe144b69b95c57a9f45ffbaf : SUCCESS in 1h 12m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/576f24a2abb24a6bbfa0efe471be19ee : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a36eda77e42d4495b54363063d0d7d51 : SUCCESS in 1h 31m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ff6d58ccfca474daebae9b19e2cc75d : SUCCESS in 23m 24s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0cff03f678d94bd48ee6b542a103ad3e : SUCCESS in 19m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2439f4c4f1794800be73f7f6ab32c732 : SUCCESS in 1h 00m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3935793488c94cbc883ae0fcb4ac2f7f : SUCCESS in 1h 06m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48984c814fc1417aa2b6ac500626b4a1 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b2ec63f1b8444c28b5ddaa6a03d8ecd : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62812030aadc42c28be789da2cf819b1 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17928d47bbe34895b6211b764fe6844b : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901937e796ba4e3d8cbdf6aa00e786c7 : SUCCESS in 3m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aad76526af5448acaba64090d447b6a4 : SUCCESS in 20m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0b2c09f3af94d40ab5d3dfce8173715 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":46},{"id":"08154f6f31d4b2752253d56d1f2360cd0a96ec44","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-09 11:35:29.000000000","message":"Patch Set 46:\n\n(13 comments)","accounts_in_message":[],"_revision_number":46},{"id":"897a80fb64917fc2c8701f89aa8ea7155fe3430a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:44:21.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9595d7c60837456b84043eab1493d5d0\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ffd9d94a438d411b9ee6a84577ed9703 : FAILURE in 12m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65ba361051844fa497e9eb349f82de14 : SUCCESS in 22m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e2a6f64b31d4162a7d92eea5a7fb329 : SUCCESS in 2h 02m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb5920a7f7754f3c8bd40255e192ee58 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34c09f60788440e28a1e0cff14aa7b26 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea78897f607f41d68bfc2129a0188ef6 : SUCCESS in 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"acc6226e6e796aedc7044406e090e31fc00e9b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-09 16:08:25.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"edbf206416e0395a01c3522f400d6d57ed382543","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-09 16:08:36.000000000","message":"Patch Set 46:\n\n(9 comments)","accounts_in_message":[],"_revision_number":46},{"id":"5bbbd718b557246ec6ce366c56f662b1c18dd886","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-02-09 17:11:27.000000000","message":"Patch Set 47: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":47},{"id":"12a20f033fb34e096c00fdb79b04d75ef71e14cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-09 17:32:16.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"45dc9530cdc996cb76160b0da2de67474f956233","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-09 17:32:43.000000000","message":"Patch Set 48:\n\n(4 comments)","accounts_in_message":[],"_revision_number":48},{"id":"3b07b352d0aa70be2e36f7206061f1815ea4e944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 18:38:39.000000000","message":"Patch Set 48:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6a720a6a1341b998bd55eb2767e969\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/41c7757be7844d148d0f59567b22d2f6 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3eac2e9a50d743278505accfb7228ab2 : POST_FAILURE in 25m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8ac2a9ec8ca488c8426e6be1961ca2c : SUCCESS in 1h 03m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4baeb89b395c44b1a932d155af266cb6 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/822460cdd6264bcdb2af6b516a60ce8e : SUCCESS in 3m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d3af11a013d4d36a0f3958822253ae5 : POST_FAILURE in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"2853e3d04bce5ef961eb1e9a95f6191f61002765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 19:22:50.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab0ed7bb7e7e4da496f5f287adb4510a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4c21124a83e491095dfe09958773c57 : SUCCESS in 33m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a0f496bff2034027b16273b26299f228 : SUCCESS in 30m 15s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/10890a4d485f4b4aa5e806e5ca4df8b5 : SUCCESS in 35m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6becc1717f5e4b5ab0b4d4af6eb820b9 : SUCCESS in 1h 30m 59s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f6fd4260f57a48ec87a2779d1f8b9d4c : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/30523df707cd4f1da57ec0b3861b118e : SUCCESS in 1h 18m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0f79b1845be74c9e9e26bcec06c2f936 : SUCCESS in 1h 39m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fea7979da83c4d8fae56b5f36028dfdd : POST_FAILURE in 25m 32s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/be0038178379496a92ddd1fb7e507a50 : SUCCESS in 18m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2dc4bca03ebd47a99023ada12f811fca : POST_FAILURE in 1h 01m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/780a2c37ce43484db7d22a105e5bb74d : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982084abaed245c3bc63cfd06e13c19e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/525d019090324747b36b3d130466d511 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bad6ae5b8eb142fcb8da07b7e1f9e9a9 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7726346e9d3243d1a3fdf43b67510811 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e3a1e363a444b268e95b9a45d6f310f : SUCCESS in 3m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db624b7beb1d4d718f5c02c07b9c11dc : POST_FAILURE in 22m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32128c9cf59a41829daf54a0e31392c1 : POST_FAILURE in 24m 57s","accounts_in_message":[],"_revision_number":48},{"id":"e5cb2f96ff4b2b8198fed3632d77ceaa65f055aa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-10 08:01:01.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"015e0c174f65bbff027c5c8105cbfd4a9c6b5f67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 10:39:05.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c3f99ceb1774d35a2411393884568ab\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e12d8e78745d4b8b8a8a278447486aaa : SUCCESS in 2h 35m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ab17f74deed486fb7aaf52b1b17c29c : SUCCESS in 1h 49m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d61b4faab874672a31ea430ed090273 : SUCCESS in 2h 24m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74fe8d60023344b9aa5c54954f319919 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02fac5791f674c8ea71a0cc12b4a9463 : SUCCESS in 7m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a378f8749bd4bb8aa1eadc7060c555a : SUCCESS in 1h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"7cf1708949b4ff005d4ed183eb8f049680855d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 11:08:39.000000000","message":"Patch Set 48:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dade10784ea04dfcada5b48cfb9cb2b4\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e2e0940df39241fead6d9e0f1cb0e9e3 : TIMED_OUT in 3h 01m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/778715b07e3446848c46b23746aa04ab : SUCCESS in 1h 08m 35s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/05ba1c7f195144b39deb6dc98c7f0a58 : SUCCESS in 1h 09m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aedbb37358c547168e3f338c497eb902 : SUCCESS in 1h 57m 54s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e1e3870f8214d1abdb3c264d130f18a : TIMED_OUT in 2h 02m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/f8f52142f77b4cff82ed17ad19e44c7e : SUCCESS in 1h 15m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8a700f2a03b24cd1a3977cc8fae7b495 : FAILURE in 2h 17m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d35af0a0b7734bfab972d5c1204dde99 : SUCCESS in 21m 30s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d662506514de4cedae3fb3075efe77d4 : SUCCESS in 39m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dd510221d83a4769af0c1a322599d6bf : SUCCESS in 58m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/891c32ca42d74e7c898b62c059d7e3e5 : SUCCESS in 2h 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d53ac357c9460b8df5e2b7a7771429 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26f847860cd24c7198c309afc903d693 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daf095439e4345ebaedbbac7e5cf022f : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77ac9072b0b45478ce71b6e1375daa1 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/918964fd922b4ce183edd7ed9480762f : SUCCESS in 3m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b183de09f754d85b6ee9057b06986f0 : SUCCESS in 23m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f2685feea814a2bb0a5f134ddafd6f1 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":48},{"id":"8a7d6e9d7fc471446d85dcda4712f0ed21e7b28d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 07:28:12.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"d928ccc9eacb8a37a40ed23032416cf864a4985b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:47:16.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1210a20c61494c478dc2baadbfb5de30\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/db1a241cb0de4ca1934e1bac3f3163dc : SUCCESS in 22m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/88ba5c5fd3df46c2919ad75e50382b7a : SUCCESS in 2h 16m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77315ce4f8284557b62b9ec90ea0e331 : SUCCESS in 1h 17m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99e5a7d52e5b4b00896e304b5dc10e3e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ff03d7bd3f443d5b6db9f4ce5b10425 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9faf4e57fce94f32a0e8163a2d22df5a : SUCCESS in 1h 56m 43s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"795f9cd273e0b4336beebe71a0e73064e3b2f322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 09:49:25.000000000","message":"Patch Set 49: Published edit on patch set 48.","accounts_in_message":[],"_revision_number":49},{"id":"352f710775640598309af91c7e0961fcc3a62578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 10:33:54.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"88913068994f52a2c5f6ae56d8cf7b002bbc2429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 10:42:32.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"c76d3696c91e12dcf884f181531687504b4d738f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 10:44:35.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"3848462faeba54a39dfb5be1337fc547f2637c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:29:57.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d532e82e7daa48a1b5ec6006c4cc93a3\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/26677e851aa94b92a063b3c08947bcbc : SUCCESS in 28m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9070f00aae9f4212b031e6c2439f94a0 : SUCCESS in 28m 38s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/f9aa213affd849b58d9a0cc1269f1e3b : SUCCESS in 27m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3fc6cc8425ee4923bf33430af112805d : SUCCESS in 1h 12m 37s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bffeb475f9aa459e875f1f725768feef : SUCCESS in 1h 06m 03s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/fad027e3a02243aea2901d37d3b25939 : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/40703e917b42419faef7d7834776d059 : SUCCESS in 1h 11m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5204bd21d04747788c8b7f47d2ec3d69 : SUCCESS in 17m 17s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/44e1b01789f741f5aec195efa10fb2f5 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6ee3417f79914d249752be4ef5a7f1ca : SUCCESS in 58m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5edd7fd120c84311ab9aaca46cdd6ab8 : SUCCESS in 1h 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41dffcf2c1bd45ce9fd92a33652f6bb2 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10222a87ee0947a195c578a808e9321b : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c06dc5744f45c088a4e6cc5ce6277e : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1087a7e4dc44c93915aaabccf123943 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e94583cfbc874834a8146189a34abe3b : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/495902ab06674288a9b0e6b34d4febc8 : SUCCESS in 18m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3870a4f1f2b47539cc7f2c763b48141 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":52},{"id":"9f73d9a9fba87607d79577cb3ea0136743923a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 12:35:35.000000000","message":"Patch Set 53: Published edit on patch set 52.","accounts_in_message":[],"_revision_number":53},{"id":"a1d5b4ca127702a4838b130081240a5cbf2feb0f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 12:39:11.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"c52bbd322024dae108e32bcff9318f55ac1978d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:44:57.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fadd3a7b1e44fe595dc5df955bb8b3c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b34729efa2aa46fab8e3557086bedf7f : TIMED_OUT in 3h 06m 14s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc385e6deb1647f990f72acd9d03ab12 : SUCCESS in 3h 03m 10s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f73c60bd3ec43afa9c8f08b92a94e18 : SUCCESS in 2h 12m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/308624faafd444d2887fc6e879a1f700 : FAILURE in 2h 46m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/751a1713a36f4f7ab62a2f953a08cbd3 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a5c9178f42945bfa480c5a9fc707a57 : SUCCESS in 3m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e90b1b647d244154aa6356d6c774dd4f : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"1616101f5a1266921e06bbde2032d7d0259cc27c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:48:28.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0c3d0c871d049a6bad74f5823884bb7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/65d71b5cdff049688ba365ca7beefbd4 : TIMED_OUT in 3h 02m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2200bd9009c24432abe1e42a36a265ba : TIMED_OUT in 3h 02m 48s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/64a91ca5be0f4ddab7a473b1e711bd58 : SUCCESS in 39m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/045b90c79278449f9cc58e19afe3b7ce : SUCCESS in 1h 26m 54s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/203e6389ce4e421d9941a4cb47948fd4 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/11fbaad89c8e4bb28028e3d081d35cdd : SUCCESS in 1h 19m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/138a0313841647d8b25e5f7aac81d21e : SUCCESS in 1h 37m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1a74e34aaef64b7fa1bb75a039938ccf : SUCCESS in 22m 51s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4a2ff5f51f7c4fd9a39b46c855a66da3 : SUCCESS in 24m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e61a0d87801431c862347d152a6517e : SUCCESS in 53m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b929f4cd5ef54d1590846be7f18a61b9 : SUCCESS in 1h 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edc43f0f94a46e49f12eacc8fc94d0b : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cae4e30497184dd99b9807e7161ea24c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc1f712eab7a4336b782ac69f9923b4a : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62732deb47340359ce40506d362efe0 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f7262c0753041b9bea999521fdf4fb0 : SUCCESS in 3m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e10ee00d7d8485aa7abeefacd8e1323 : SUCCESS in 19m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/067d4133d30b4cf4871bd645b7ebc238 : SUCCESS in 25m 20s","accounts_in_message":[],"_revision_number":53},{"id":"de656de07e7d3a70e5547d22e30b9a0b510047c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 16:34:35.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"c1f13d44d45391aba49ea056e955c5eaed4a39b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:58:06.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86fca4e58cd14c4bad506af4568dfef6\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a83db48b5c22441cb98ab53b7bc7cabd : SUCCESS in 37m 05s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/12fc9cbe34b0466fbbc2f835015fd5d3 : SUCCESS in 27m 50s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd57191282d74853bb1bcb5c5ea30194 : SUCCESS in 21m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b33762e40bb4241ad204615ed69971e : SUCCESS in 1h 21m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2fe54ed0b704189b5851d037d3b1c2a : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/713bc2418ba4480c91d8aeef940852e0 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/715112e1e98147188152ea0bda63a5a0 : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"b53f3b0a502c0986515e5995f79b822611cd2dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:55:42.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a307feb85e5473493d97625a5ef7d9a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cc79d4a9a3174234a5ef69d54badf641 : TIMED_OUT in 3h 01m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/82e1e8917bec4edbbe6ff91e7fc40fed : SUCCESS in 37m 11s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/df86b7eb5b204874af794a80d240e834 : SUCCESS in 30m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cabb2677ef314b90bc2bae387b5bc117 : SUCCESS in 1h 42m 46s (non-voting)\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f087f3ee3d654685851b611382bc5763 : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/dccd8ec631fb48199f6a30f482f110be : SUCCESS in 1h 13m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a44ef9ceb6a94d75bdebd750cc28fe4c : SUCCESS in 1h 35m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9abf168d5c5e4e0ca5ea20244ac066c0 : SUCCESS in 23m 17s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bf47e2836c794430b15f5eebff71bd3e : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bb808b8741e0424b9af24e76f76782e2 : SUCCESS in 1h 13m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b060e4d2402549b7b1f87760c83f9735 : SUCCESS in 1h 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f63ff67366744eda49a0cada0663022 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fabfd90cd9aa466d9591b7c25cc340b5 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9529ff7d1a184dca86654e55e4a2f419 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19403e62a63d415289138e089ebce023 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d11ff2769947a4968350ac7d1c415a : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cba184690f24ce291589635995c5d18 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74478512668f4e5e9b3f123dbd83e65f : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":53},{"id":"60748af97fbc1a146f82025da404264915211e6c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 10:04:13.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"0844c5907bde2cd96f28defb5067087df6658972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:41:10.000000000","message":"Patch Set 53:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":53},{"id":"133a30541f99569f0e8ac6af048fc013d03084cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:41:16.000000000","message":"Patch Set 53: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/864920 is needed.","accounts_in_message":[],"_revision_number":53},{"id":"0b7af57b2c24eda6b6768f29e1fd59b886c82a2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 10:51:58.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"cc04a9a5b0b7c246a6f63fbc9bac92cb45d11378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:28:39.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75664b1a58a345fd82e3990b9cba88d6\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fcfd6d6a0fe8467fb0dfab3b0ab325eb : SUCCESS in 39m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b15b52970da94c5d9a12fb820c8dd955 : SUCCESS in 46m 43s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/068454d1dae5439a8a11542e240ea288 : SUCCESS in 44m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/97cc564fee4747f58283cf31546177aa : SUCCESS in 1h 22m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/47b6c6de6bfd466e84c8d39eb56a52ba : SUCCESS in 1h 18m 13s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a8ee23f8035149e3ba44b9cf254a6394 : SUCCESS in 1h 12m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2bef4a573ce475193a16458bcfaba92 : SUCCESS in 1h 26m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/21393a1c368c4cdba788cf1cc6e1b866 : SUCCESS in 43m 19s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7ffaac33eb334966ac3496d20be80921 : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2985849f4dd645efbc23f232adadc22c : SUCCESS in 59m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a51e85f8ced9428abf5edab44426edc9 : SUCCESS in 1h 29m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b112726b7b641e79e47a14f7f3356bb : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7400a39c0855483199b335339b48a045 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f892a4443970409d86c62efd1c5505f5 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ea3ec1525a46e0a8492d1c591921e3 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc095a580584298bc741cd42a619216 : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b83dc207a79e48c68d8d72c7f0d1cb77 : SUCCESS in 35m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce44f1bc8082474fba6bd97da6f05022 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":53},{"id":"6b3be1c0e8f191f779224d2686eeb630e653b57a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 12:29:00.000000000","message":"Patch Set 53: Code-Review+2","accounts_in_message":[],"_revision_number":53},{"id":"43a3f26a89ee2fa7c0f6ab7691c1e0c0db5a3490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 13:59:16.000000000","message":"Patch Set 53:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bde1feb7a584917aa26345a4557d366\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/74a2193dda2e4626bb601a30cbbb274e : SUCCESS in 46m 42s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/398262ae95c44a27a7d4ef3caa4ccafa : TIMED_OUT in 3h 03m 48s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e149ee15221409b85dcdd3d18d295ac : SUCCESS in 44m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17946776a3cb47bcadf7afd89f86275b : TIMED_OUT in 3h 02m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f18c788a5431460194324b69f30cc9aa : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8ea498c85354e8988a78be0dea78b3f : SUCCESS in 3m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6add0dda9fc472397d1434bebf38c13 : SUCCESS in 2h 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"678b9c915cf39532aa5fab28ad1fd74431955743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-17 06:28:58.000000000","message":"Patch Set 54: Patch Set 53 was rebased","accounts_in_message":[],"_revision_number":54},{"id":"d4c32e5f69e1e03f1a8b5136e1400c12ae1b2049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:37:01.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f84e076b862b4cd2a57fc0a1d1ada0d0\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/07874d03ae604dfbacad82fda038eb6c : FAILURE in 36m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0805271ede2493783883e47928656d0 : TIMED_OUT in 3h 04m 17s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/262cfd03e1454fb388d94ce2a3fd25ea : TIMED_OUT in 3h 02m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3aa0db510c6c4ee79a1180e9d8fb3d55 : SUCCESS in 1h 19m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ac37d03d20b548029bc51fd3a677df3e : SUCCESS in 1h 08m 30s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/7bca29dbbcc74051be2bdb81c6ab6a00 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dda364290efc4ef582b0847e6d8d9b93 : SUCCESS in 1h 39m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cdce3ab4b57c40ec88b8fa8f9eee65b5 : SUCCESS in 17m 39s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a515629c4b524474a1250246a91311b0 : SUCCESS in 19m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c48ca72bdb446559d6e9c891866b9ff : SUCCESS in 1h 01m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a2705420ae35422b95832fbd70665bc4 : SUCCESS in 1h 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4605bc9b0cdc49caae9c5dab23595a8f : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7abd4ce437074b2e8f57d425af0cde67 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16130f355cf4453a8121416a48d8dd0e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9136648255894d9699efc000d7e3da95 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af7f1f5850574e5cb60be774540b6117 : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf05a43ce06c4362be7f261370b514e3 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49e0c89017084592976415515552b50b : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":54},{"id":"279c08f2772af11f1549ffde364b5936fbde5f0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:38:59.000000000","message":"Patch Set 54:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2655e3e3e32448bbb37a64b69d10dedd\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3d9a3e9c44d4dee8d384ad349bb3941 : TIMED_OUT in 3h 05m 22s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0463d79559434b3b910c9df3772fb74d : TIMED_OUT in 3h 05m 33s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4a3ed838fd64deea9d16ac88b112230 : TIMED_OUT in 3h 03m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ac5cddb7f174a71a5f9af134eccaf18 : TIMED_OUT in 3h 02m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2eea0225e5749da8570444113d937a3 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da71ad7d735540efa872c0d4c786b35d : SUCCESS in 7m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b3dbe7b6a7f49189ee57da039d73b32 : SUCCESS in 2h 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"8de2ba25530b5c90359ae29acf9501909b3f7f20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-17 11:16:20.000000000","message":"Patch Set 54: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"9e7ce369db04024a88476725ef9333a7826d4fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:01:17.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08db8df541e743408570afa565be6f7b\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/91c5603c33ea46b3b6bdc108ea5f913d : SUCCESS in 31m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2464d61ae1724139acf58ce85ac6ba8d : SUCCESS in 40m 56s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/be43d514a59d4b4ba0bff88170bc3244 : SUCCESS in 39m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/573155ea7a3a478085d5beb88e4bbf13 : SUCCESS in 1h 14m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6867466c3a7d4b048b49cd82769e191f : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/350db25af0c743e596f734a81f268433 : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/47d45debbe15436e918bf65478fcccdb : SUCCESS in 1h 30m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66459e3983cb4b23b71181a2e530ccb8 : SUCCESS in 24m 42s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/140f4023d1a54f04a1d051751911882e : SUCCESS in 16m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/64f0a41b47a64ebba55eae9c0e98fd98 : SUCCESS in 1h 15m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9ae5b982d5ad4afb9e8775669f091c63 : SUCCESS in 1h 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a9df49c5cd465682446a082f29b2f8 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccecd33bb44b45dc8dcdd31d12335017 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06bdac0c93f2471fba7e7a10adc52054 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5e61e4848a4d4798907cbf7a51b2f6 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5d9526d7b424e8788ae3dd55733604d : SUCCESS in 3m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64aea59ca30a45a0babc8aa51ee4c661 : SUCCESS in 17m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c6ab7901ebf4f9786c148f8c39257cc : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":54},{"id":"7f7463f46570cb6a7c6ae1d97a12c5a29b18c1bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:39:28.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be607acccf574cf39f8d14ba449e8be4\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/31b92d1fa8004372b361f0029c52cbb9 : SUCCESS in 31m 14s (non-voting)\n- kolla-build-centos9s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f92968e7b0b4ac1a5d6cc27a04b2439 : SUCCESS in 31m 24s\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1181af2ffb54a5999414dccbd6b0e6c : SUCCESS in 19m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84c4a0cecdc14c70956d0583e4b02591 : SUCCESS in 2h 20m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/320ef952958c4a42ad6a0a4f04949df6 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a853d9f4cc14150a807999517ceb6c5 : SUCCESS in 8m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d87b2cdecf24f579507c78ec06f39d4 : SUCCESS in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"cec77bc45f6c4338c22baf5f6f2a07addcf92601","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-17 13:39:54.000000000","message":"Patch Set 54: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":54},{"id":"4c5594e948305bc2d934cb4ae824648dcdbc932d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:40:26.000000000","message":"Patch Set 54: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":54},{"id":"7fc6c823e1ffbeaa407273d4fb3b2afa096f2c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:12:42.000000000","message":"Patch Set 54: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/723d2490c974463c8b4d584b84d49fcb\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cdd5961c6de54be980e73c38f2d9c4f9 : SUCCESS in 49m 37s\n- kolla-build-centos9s-source https://zuul.opendev.org/t/openstack/build/95d1a986d85e44b596ac4f6c43a4fea4 : SUCCESS in 55m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6951a9df85a9474f9d46e5e20701dcd8 : SUCCESS in 1h 18m 31s\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/a6f60d045fec46d69bdc72fa0eb365b2 : SUCCESS in 1h 31m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7235e9bbe1e64ed4bcffe3795791964d : SUCCESS in 1h 28m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e6566c2417a94f5faedd78268f048491 : SUCCESS in 20m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/adcb718156a44ad9bdf77596a479dcaa : SUCCESS in 1h 12m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f6255b238469456485b7508173ae50dc : SUCCESS in 1h 03m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d365cee4c5d04884baaecc689aeb4c94 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dacdab70aad14b3caa528d64bc8e9dcd : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ba1471365e44aa99dbb0f1be4f2aea : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df946e4010964487b76aa10c4d027a2b : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e97c10e49ca840f8ab107641aa2615a0 : SUCCESS in 3m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/752155bb78bd4f3f8cf4a4d82693083c : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":54},{"id":"657253acc3b69a76978370ccb6a0f2273f9bf86c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:12:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":54},{"id":"cd7b0c2763ed9b992b450e537290db0e4546ac2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:15:01.000000000","message":"Patch Set 54:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e06828f646c94d508ea6ed6af7ae46d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36db83fbf66342649811f028bf2f97c1 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f3931f9d12441f5957133b612deddb2 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"82ee1e6319d8631b2637c69f3413a020f6a27c0a","revisions":{"58a47a619ef03832e328daf2f5d8d09abeeefffd":{"kind":"REWORK","_number":1,"created":"2023-01-12 08:51:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/1"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 08:51:48.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58a47a619ef03832e328daf2f5d8d09abeeefffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58a47a619ef03832e328daf2f5d8d09abeeefffd"}]},"branch":"refs/heads/stable/yoga"},"6dc15c34a80d9e8f88aa6fad6dcb45cf14e6e498":{"kind":"REWORK","_number":2,"created":"2023-01-12 09:02:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/2"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 09:02:11.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dc15c34a80d9e8f88aa6fad6dcb45cf14e6e498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dc15c34a80d9e8f88aa6fad6dcb45cf14e6e498"}]},"branch":"refs/heads/stable/yoga"},"68dbacb102b8273182f4da6da031c2f6629179b2":{"kind":"REWORK","_number":3,"created":"2023-01-12 09:20:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/3"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 09:20:18.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68dbacb102b8273182f4da6da031c2f6629179b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68dbacb102b8273182f4da6da031c2f6629179b2"}]},"branch":"refs/heads/stable/yoga"},"f6bd5847adfcdd21e1051de13d0e0db8f2596474":{"kind":"REWORK","_number":4,"created":"2023-01-12 10:48:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/4"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 10:47:33.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6bd5847adfcdd21e1051de13d0e0db8f2596474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6bd5847adfcdd21e1051de13d0e0db8f2596474"}]},"branch":"refs/heads/stable/yoga"},"c6ef4ecc2e68bb08312d52ac03c56402f7199979":{"kind":"REWORK","_number":5,"created":"2023-01-12 11:24:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/5"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:24:14.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6ef4ecc2e68bb08312d52ac03c56402f7199979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6ef4ecc2e68bb08312d52ac03c56402f7199979"}]},"branch":"refs/heads/stable/yoga"},"251d18133e4a9c7dde0bbebc61ef92f3e1f6cb53":{"kind":"REWORK","_number":6,"created":"2023-01-12 11:34:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/6"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:34:06.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/251d18133e4a9c7dde0bbebc61ef92f3e1f6cb53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/251d18133e4a9c7dde0bbebc61ef92f3e1f6cb53"}]},"branch":"refs/heads/stable/yoga"},"96d26ab8da24cf15bec9a728c4c9200dca8f80bb":{"kind":"REWORK","_number":7,"created":"2023-01-12 12:01:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/7"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 12:01:10.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d26ab8da24cf15bec9a728c4c9200dca8f80bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d26ab8da24cf15bec9a728c4c9200dca8f80bb"}]},"branch":"refs/heads/stable/yoga"},"151365eca5a71caee449527c5eb89d222b5d5651":{"kind":"REWORK","_number":8,"created":"2023-01-12 13:05:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/8"}}},"commit":{"parents":[{"commit":"5d56db08911441a3e3b603b5c09779514ba6ee88","subject":"horizon: Ensure mysqldb client is installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d56db08911441a3e3b603b5c09779514ba6ee88"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 13:04:55.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/151365eca5a71caee449527c5eb89d222b5d5651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/151365eca5a71caee449527c5eb89d222b5d5651"}]},"branch":"refs/heads/stable/yoga"},"e3311afb37bb72a7b571b911c9961e0f0a68d82e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-01-12 13:30:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/9"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 13:30:12.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3311afb37bb72a7b571b911c9961e0f0a68d82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3311afb37bb72a7b571b911c9961e0f0a68d82e"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"77b8c5ea5ecb558363b0b873bcecf76d9bb9bb49":{"kind":"REWORK","_number":10,"created":"2023-01-13 09:14:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/10"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-13 09:14:42.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b8c5ea5ecb558363b0b873bcecf76d9bb9bb49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77b8c5ea5ecb558363b0b873bcecf76d9bb9bb49"}]},"branch":"refs/heads/stable/yoga"},"dcd9af45ebf010fa7c1dc9cdaceedbf952a96686":{"kind":"REWORK","_number":11,"created":"2023-01-13 09:27:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/11"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-13 09:27:10.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd9af45ebf010fa7c1dc9cdaceedbf952a96686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd9af45ebf010fa7c1dc9cdaceedbf952a96686"}]},"branch":"refs/heads/stable/yoga"},"3c9ed1a43f40812c2f05db60d4faf5c3c99c28af":{"kind":"REWORK","_number":12,"created":"2023-01-13 10:03:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/12"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-13 10:03:01.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9ed1a43f40812c2f05db60d4faf5c3c99c28af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9ed1a43f40812c2f05db60d4faf5c3c99c28af"}]},"branch":"refs/heads/stable/yoga"},"20f7bfe008896a7a9e95813893e45d0dd5d297ff":{"kind":"REWORK","_number":13,"created":"2023-01-19 11:52:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/13"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 11:52:11.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f7bfe008896a7a9e95813893e45d0dd5d297ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f7bfe008896a7a9e95813893e45d0dd5d297ff"}]},"branch":"refs/heads/stable/yoga"},"a6b8d166f71b64d0fd0ca16a0ce7b53425e2fabc":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-01-19 11:52:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/14"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 11:52:19.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6b8d166f71b64d0fd0ca16a0ce7b53425e2fabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6b8d166f71b64d0fd0ca16a0ce7b53425e2fabc"}]},"branch":"refs/heads/stable/yoga"},"c292db64df33adc5d575217c00436ee74d3b2036":{"kind":"REWORK","_number":15,"created":"2023-01-19 11:55:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/15"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 11:55:01.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c292db64df33adc5d575217c00436ee74d3b2036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c292db64df33adc5d575217c00436ee74d3b2036"}]},"branch":"refs/heads/stable/yoga"},"fbacae09de28f5731864a178b0d3a5c075620c9d":{"kind":"REWORK","_number":16,"created":"2023-01-25 11:19:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/16"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 11:19:00.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbacae09de28f5731864a178b0d3a5c075620c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbacae09de28f5731864a178b0d3a5c075620c9d"}]},"branch":"refs/heads/stable/yoga"},"11269b6981cad5f2bf93bc299ce8bcb9739d8e21":{"kind":"REWORK","_number":17,"created":"2023-01-25 12:50:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/17"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 12:50:45.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11269b6981cad5f2bf93bc299ce8bcb9739d8e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11269b6981cad5f2bf93bc299ce8bcb9739d8e21"}]},"branch":"refs/heads/stable/yoga"},"0ca40c5438b1548a56f3dd82fd91581d26e36776":{"kind":"REWORK","_number":18,"created":"2023-01-25 13:54:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/18"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 13:54:25.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca40c5438b1548a56f3dd82fd91581d26e36776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca40c5438b1548a56f3dd82fd91581d26e36776"}]},"branch":"refs/heads/stable/yoga"},"9a0b5902f828bfdac470dc872de1184c07d63995":{"kind":"REWORK","_number":19,"created":"2023-01-25 16:32:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/19"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 16:32:43.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a0b5902f828bfdac470dc872de1184c07d63995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a0b5902f828bfdac470dc872de1184c07d63995"}]},"branch":"refs/heads/stable/yoga"},"076356c358f934d2d5eeeb484498e71e18e90f5f":{"kind":"REWORK","_number":20,"created":"2023-01-25 17:44:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/20"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 17:44:35.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/076356c358f934d2d5eeeb484498e71e18e90f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/076356c358f934d2d5eeeb484498e71e18e90f5f"}]},"branch":"refs/heads/stable/yoga"},"d5453aedcd37b608f91a81b5ae7b37675375f297":{"kind":"REWORK","_number":21,"created":"2023-01-26 09:37:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/21"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 09:37:16.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5453aedcd37b608f91a81b5ae7b37675375f297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5453aedcd37b608f91a81b5ae7b37675375f297"}]},"branch":"refs/heads/stable/yoga"},"9afadd4031c50a6b5ef4173861c95fe58a72d1e2":{"kind":"REWORK","_number":22,"created":"2023-01-26 09:49:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/22"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 09:49:15.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afadd4031c50a6b5ef4173861c95fe58a72d1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9afadd4031c50a6b5ef4173861c95fe58a72d1e2"}]},"branch":"refs/heads/stable/yoga"},"53d64f39157740d16e4831fa62d708bddfc853d0":{"kind":"REWORK","_number":23,"created":"2023-01-26 10:07:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/23"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 10:07:19.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d64f39157740d16e4831fa62d708bddfc853d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53d64f39157740d16e4831fa62d708bddfc853d0"}]},"branch":"refs/heads/stable/yoga"},"e48333a16daa0fbd2766c0659bb1107e1f2ef3a7":{"kind":"REWORK","_number":24,"created":"2023-01-26 10:20:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/24"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 10:20:33.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e48333a16daa0fbd2766c0659bb1107e1f2ef3a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e48333a16daa0fbd2766c0659bb1107e1f2ef3a7"}]},"branch":"refs/heads/stable/yoga"},"c19538e8cafcb826e735136e997369eecfd076b2":{"kind":"REWORK","_number":25,"created":"2023-01-26 10:30:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/25"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 10:30:33.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c19538e8cafcb826e735136e997369eecfd076b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c19538e8cafcb826e735136e997369eecfd076b2"}]},"branch":"refs/heads/stable/yoga"},"1863586a6256e9e09efae58ebde49835a433fd22":{"kind":"REWORK","_number":26,"created":"2023-01-26 10:46:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/26"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 10:46:16.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1863586a6256e9e09efae58ebde49835a433fd22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1863586a6256e9e09efae58ebde49835a433fd22"}]},"branch":"refs/heads/stable/yoga"},"45a44d0ceeda7277502493e8b75dc118e52b895a":{"kind":"REWORK","_number":27,"created":"2023-01-26 11:28:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/27"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 11:28:41.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45a44d0ceeda7277502493e8b75dc118e52b895a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45a44d0ceeda7277502493e8b75dc118e52b895a"}]},"branch":"refs/heads/stable/yoga"},"fb6535692fd30d4def63cfc35247162497d5b27a":{"kind":"REWORK","_number":28,"created":"2023-01-26 11:56:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/28"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 11:56:26.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb6535692fd30d4def63cfc35247162497d5b27a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb6535692fd30d4def63cfc35247162497d5b27a"}]},"branch":"refs/heads/stable/yoga"},"071ae6a2d41d86b040bd04659d6a2c38f6eb4110":{"kind":"REWORK","_number":29,"created":"2023-01-26 13:03:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/29"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 13:03:02.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/071ae6a2d41d86b040bd04659d6a2c38f6eb4110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/071ae6a2d41d86b040bd04659d6a2c38f6eb4110"}]},"branch":"refs/heads/stable/yoga"},"6f6c264d09e18e71010f42ec885c657ee26f68b8":{"kind":"REWORK","_number":30,"created":"2023-01-26 14:05:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/30"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 14:05:39.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f6c264d09e18e71010f42ec885c657ee26f68b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f6c264d09e18e71010f42ec885c657ee26f68b8"}]},"branch":"refs/heads/stable/yoga"},"0023333270156b54c10275139dc0624a351b18a2":{"kind":"REWORK","_number":31,"created":"2023-01-26 14:20:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/31"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 14:20:28.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0023333270156b54c10275139dc0624a351b18a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0023333270156b54c10275139dc0624a351b18a2"}]},"branch":"refs/heads/stable/yoga"},"bc82e66074e6ef975af21b41ae46eacdfdfb8c4f":{"kind":"REWORK","_number":32,"created":"2023-01-31 08:29:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/32"}}},"commit":{"parents":[{"commit":"b930a682165a536ca4f9587534591f249cee79cd","subject":"relay non-zero exit codes when creating openvswitch bridges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 08:29:12.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc82e66074e6ef975af21b41ae46eacdfdfb8c4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc82e66074e6ef975af21b41ae46eacdfdfb8c4f"}]},"branch":"refs/heads/stable/yoga"},"e3c60a82cf8f6e40da14b19a51f135b55aeb7bef":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2023-01-31 08:29:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/33"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 08:29:36.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3c60a82cf8f6e40da14b19a51f135b55aeb7bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3c60a82cf8f6e40da14b19a51f135b55aeb7bef"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"a50b95ecf7ccfe29f232b687deaac2e34cd3fabd":{"kind":"REWORK","_number":34,"created":"2023-01-31 08:29:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/34"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 08:29:58.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a50b95ecf7ccfe29f232b687deaac2e34cd3fabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a50b95ecf7ccfe29f232b687deaac2e34cd3fabd"}]},"branch":"refs/heads/stable/yoga"},"cfb362e31966bf7f0035bd1f8a48ae9ec23b8b7f":{"kind":"REWORK","_number":35,"created":"2023-01-31 14:50:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/35"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 14:50:38.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfb362e31966bf7f0035bd1f8a48ae9ec23b8b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfb362e31966bf7f0035bd1f8a48ae9ec23b8b7f"}]},"branch":"refs/heads/stable/yoga"},"fca54253c24128e70dd4ea3c7293c3d15682e749":{"kind":"REWORK","_number":36,"created":"2023-01-31 15:07:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/36"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:07:14.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca54253c24128e70dd4ea3c7293c3d15682e749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fca54253c24128e70dd4ea3c7293c3d15682e749"}]},"branch":"refs/heads/stable/yoga"},"648d706bc21ef6492bc5e11ecf35f82d9e0f7046":{"kind":"REWORK","_number":37,"created":"2023-01-31 15:41:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/37"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 15:41:29.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/648d706bc21ef6492bc5e11ecf35f82d9e0f7046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/648d706bc21ef6492bc5e11ecf35f82d9e0f7046"}]},"branch":"refs/heads/stable/yoga"},"b706d6b3f098aa10e1f15a924fba6f3c48acbcf7":{"kind":"REWORK","_number":38,"created":"2023-01-31 17:24:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/38"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-31 17:24:29.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b706d6b3f098aa10e1f15a924fba6f3c48acbcf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b706d6b3f098aa10e1f15a924fba6f3c48acbcf7"}]},"branch":"refs/heads/stable/yoga"},"2f30c45d16e6426f0e6a09f2966b52f204dfd760":{"kind":"REWORK","_number":39,"created":"2023-02-01 07:57:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/39"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 07:57:38.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f30c45d16e6426f0e6a09f2966b52f204dfd760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f30c45d16e6426f0e6a09f2966b52f204dfd760"}]},"branch":"refs/heads/stable/yoga"},"4e7ba6514f6a849c718b334d6e1fd8cad24e4ccd":{"kind":"REWORK","_number":40,"created":"2023-02-01 08:52:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/40"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 08:51:56.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e7ba6514f6a849c718b334d6e1fd8cad24e4ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e7ba6514f6a849c718b334d6e1fd8cad24e4ccd"}]},"branch":"refs/heads/stable/yoga"},"9af0a04305b59028c6ef7e5bdd52cf9c1474222f":{"kind":"REWORK","_number":41,"created":"2023-02-01 09:46:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/41"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 09:46:31.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9af0a04305b59028c6ef7e5bdd52cf9c1474222f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9af0a04305b59028c6ef7e5bdd52cf9c1474222f"}]},"branch":"refs/heads/stable/yoga"},"eb412e2c4436623c665af7f4f33f6cce6165f4b0":{"kind":"REWORK","_number":42,"created":"2023-02-01 10:09:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/42"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-01 10:09:15.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb412e2c4436623c665af7f4f33f6cce6165f4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb412e2c4436623c665af7f4f33f6cce6165f4b0"}]},"branch":"refs/heads/stable/yoga"},"63ceb829858faea516b7957555cea0b193c32e34":{"kind":"REWORK","_number":43,"created":"2023-02-02 07:30:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/43"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 07:30:58.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63ceb829858faea516b7957555cea0b193c32e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63ceb829858faea516b7957555cea0b193c32e34"}]},"branch":"refs/heads/stable/yoga"},"00e847c300ebf398ab2016918618035a4c0d25a2":{"kind":"REWORK","_number":44,"created":"2023-02-02 10:46:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/44"}}},"commit":{"parents":[{"commit":"f3234d17d655e48208284369a169a9788eda50d4","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3234d17d655e48208284369a169a9788eda50d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-02 10:46:49.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00e847c300ebf398ab2016918618035a4c0d25a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00e847c300ebf398ab2016918618035a4c0d25a2"}]},"branch":"refs/heads/stable/yoga"},"49002fe649dda54497ff984f17b59519ca09f0a3":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2023-02-08 09:14:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/45"}}},"commit":{"parents":[{"commit":"73e4461bd25231dfcd40296e13bf5785b41d801c","subject":"CI: Install Apparmor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73e4461bd25231dfcd40296e13bf5785b41d801c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-08 09:14:16.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49002fe649dda54497ff984f17b59519ca09f0a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49002fe649dda54497ff984f17b59519ca09f0a3"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"159db6021ebe3410e0f28c1bd8d5ec97a9342ebe":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2023-02-09 09:39:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/46"}}},"commit":{"parents":[{"commit":"daedc1d1f4acd8c3494f625b5f1189928f0ee9e4","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daedc1d1f4acd8c3494f625b5f1189928f0ee9e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-09 09:39:25.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/159db6021ebe3410e0f28c1bd8d5ec97a9342ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/159db6021ebe3410e0f28c1bd8d5ec97a9342ebe"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"d67a4660e1a5b3aed9b0ac4c6a8439e7f53b94ca":{"kind":"REWORK","_number":47,"created":"2023-02-09 16:08:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/47"}}},"commit":{"parents":[{"commit":"daedc1d1f4acd8c3494f625b5f1189928f0ee9e4","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daedc1d1f4acd8c3494f625b5f1189928f0ee9e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-09 16:08:18.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67a4660e1a5b3aed9b0ac4c6a8439e7f53b94ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d67a4660e1a5b3aed9b0ac4c6a8439e7f53b94ca"}]},"branch":"refs/heads/stable/yoga"},"af1d76a0cf3d8bae92b49dba8d9c4b0ec650a7bc":{"kind":"REWORK","_number":48,"created":"2023-02-09 17:32:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/48"}}},"commit":{"parents":[{"commit":"daedc1d1f4acd8c3494f625b5f1189928f0ee9e4","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daedc1d1f4acd8c3494f625b5f1189928f0ee9e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-09 17:32:16.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af1d76a0cf3d8bae92b49dba8d9c4b0ec650a7bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af1d76a0cf3d8bae92b49dba8d9c4b0ec650a7bc"}]},"branch":"refs/heads/stable/yoga"},"06977e9b419172adc7968511d4fc4b9fdfd1b4a1":{"kind":"REWORK","_number":49,"created":"2023-02-13 09:49:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/49"}}},"commit":{"parents":[{"commit":"daedc1d1f4acd8c3494f625b5f1189928f0ee9e4","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daedc1d1f4acd8c3494f625b5f1189928f0ee9e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-13 09:49:25.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06977e9b419172adc7968511d4fc4b9fdfd1b4a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06977e9b419172adc7968511d4fc4b9fdfd1b4a1"}]},"branch":"refs/heads/stable/yoga"},"e83eec1bd52184cd31db09a4f88fac853425eab5":{"kind":"REWORK","_number":50,"created":"2023-02-13 10:33:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/50"}}},"commit":{"parents":[{"commit":"daedc1d1f4acd8c3494f625b5f1189928f0ee9e4","subject":"CI: store only templated out Dockerfiles in work_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daedc1d1f4acd8c3494f625b5f1189928f0ee9e4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-13 10:33:54.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e83eec1bd52184cd31db09a4f88fac853425eab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e83eec1bd52184cd31db09a4f88fac853425eab5"}]},"branch":"refs/heads/stable/yoga"},"ba010992326736f7ecb692557e303d1b894b6524":{"kind":"REWORK","_number":51,"created":"2023-02-13 10:42:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/51"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-13 10:42:21.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba010992326736f7ecb692557e303d1b894b6524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba010992326736f7ecb692557e303d1b894b6524"}]},"branch":"refs/heads/stable/yoga"},"265abe01da1beb436d45d6575629483de0e3488c":{"kind":"REWORK","_number":52,"created":"2023-02-13 10:44:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/52"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-13 10:44:29.000000000","tz":60},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/265abe01da1beb436d45d6575629483de0e3488c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/265abe01da1beb436d45d6575629483de0e3488c"}]},"branch":"refs/heads/stable/yoga"},"bff3685a67aac5de779e73d436f2fe6689ce3264":{"kind":"REWORK","_number":53,"created":"2023-02-13 12:35:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/53"}}},"commit":{"parents":[{"commit":"48308c7bdf1e3059f3557dd7560efe2e2ea4971f","subject":"Merge \"Add util-linux to centos base packages\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48308c7bdf1e3059f3557dd7560efe2e2ea4971f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-13 12:35:32.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bff3685a67aac5de779e73d436f2fe6689ce3264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bff3685a67aac5de779e73d436f2fe6689ce3264"}]},"branch":"refs/heads/stable/yoga"},"82ee1e6319d8631b2637c69f3413a020f6a27c0a":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2023-02-17 06:28:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/06/869906/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/869906/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/869906/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/869906/54"}}},"commit":{"parents":[{"commit":"9050b67252ff199b8fcfebc2af8203053c7a452d","subject":"Add Opensearch image(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9050b67252ff199b8fcfebc2af8203053c7a452d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-04 09:29:14.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-17 06:28:58.000000000","tz":0},"subject":"Add support for CentOS Stream 9","message":"Add support for CentOS Stream 9\n\nThis changeset adds support for CentOS Stream 9.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/864920\n\nChange-Id: I113b9984294cf8663d3fc0c8840320e1d40ea731\n(cherry picked from commit 878b00013b01e3d02a704ce7c9fac1629cd1a705)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ee1e6319d8631b2637c69f3413a020f6a27c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/82ee1e6319d8631b2637c69f3413a020f6a27c0a"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
