)]}'
{"id":"openstack%2Fkolla~870664","triplet_id":"openstack%2Fkolla~stable%2Fzed~Iad83132b61efadbf09aa9aa2edf96235085764c6","project":"openstack/kolla","branch":"stable/zed","topic":"bug/1999778","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-01-17 10:47:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad83132b61efadbf09aa9aa2edf96235085764c6","subject":"relay non-zero exit codes when creating openvswitch bridges","status":"MERGED","created":"2023-01-16 18:15:02.000000000","updated":"2023-01-17 10:48:32.000000000","submitted":"2023-01-17 10:47:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870664-bug/1999778","meta_rev_id":"5a93e08cacfaf9b356e8a897d74029bbdeced1c8","_number":870664,"virtual_id_number":870664,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-17 10:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-17 08:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-17 09:31:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-17 09:31:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 18:15:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"state":"CC"},{"updated":"2023-01-16 19:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 08:35:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-17 09:31:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b0657ff58a383e9c9cbb0b95fb1482c3c53a387d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-16 18:15:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a26d9d94f1cc17b9cc12298f701ea5368eae6c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 19:52:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43f98ddc9f334d9fb15d139364b1f873\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a032d4816e3d4efb9efad9164ce34fea : SUCCESS in 24m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/ad02e1ee3e864c79ad840c79c79ef3b2 : SUCCESS in 1h 19m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b7ff18a7d41d44ceb56aae7379724e22 : SUCCESS in 22m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bb6b68a589314c96be1effa20bbaa669 : SUCCESS in 1h 08m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1be76d372bc3471b90b98bbf598fe361 : SUCCESS in 1h 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1dba0a6ae5d4e89b4c2606b9eb35b52 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b6486f2d69644afa652afe23e166f72 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7748ad84928d47a58479e2cf963678b2 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/530ddf8d0c104ddf929e04b13c1d066b : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949f865294954626bede345907fcf166 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d78732528a34e04abc82c0512cd0c7d : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/11517693ca354ca6bd974fda16333f0f : SUCCESS in 25m 25s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/0b9cf1b010db4119ac333c203c28107d : SUCCESS in 54m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d641de9b336442b7b699612f1b28f152 : SUCCESS in 21m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77abc7cc92674dcebc8dde361fb6fb5a : SUCCESS in 47m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee411f548bd14bf6ba4e8fd1661b07e0 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dfce8ab9eab79be17aedeb794a0cc6806381551f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 22:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae72515fe4714f2aa2c9e56bc0b0cebe\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7ade7ec2018843439ca8502452e8b326 : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30cf55a9b671489d83d65d77b19de12e : SUCCESS in 1h 43m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa1fb2f6fe6e4e988778060cae78f32f : SUCCESS in 1h 47m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97086cf0df384e4cb95432385dd9ef61 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c589a1bbabb44c428861047edcaba996 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fa803f496d8c4c5084f0d175154296c3 : SUCCESS in 2h 06m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/29f0e2f8602147f5b43ecba70e478687 : SUCCESS in 1h 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f4062f9630d632a3388ed6faa004f6854138323","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-17 08:35:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98e0ea42090cbdce1755ceac0ee43e5c49c8e1e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-17 09:31:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78c55e46994f28361dc3b863d4cc0239cf21a6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 09:31:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b5491e7cf46611247d3348f887ff031fca65b23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:47:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458248f8a76545a0acf4bbce5b35a60d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/62a5db6ffcdb486ba8f5622ce4fdec88 : SUCCESS in 20m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/47eb1f12f4d54dab9a5c0aba609f5a7b : SUCCESS in 1h 05m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ca76a71d9b7f48d68a8624ac9fa7a76c : SUCCESS in 1h 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236f1d0c45f546089c8294678cca425d : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7f6b2b52dc4151b7c21eafb65c6273 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddf784affe064f2db2d862d26dee4b68 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f24bc019b141e987ccb1cbc599d959 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebdb57c992f54aadb4e391f7b82e1b65 : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ebcd32c973a4d4e8a38d79904cb540e : SUCCESS in 23m 39s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/df62f1032234445b801cf1aedf705335 : SUCCESS in 1h 04m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b435b1b9bbd40f7a7effe48ffcc1f45 : SUCCESS in 19m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02992cf8c1fa4076bac462640ba44a44 : SUCCESS in 45m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/81412cc5ac5a4e5dbb7c67af4dfa78f4 : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"83356073ecc7138f92b6c70f0597e2b91d74fe1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:47:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a93e08cacfaf9b356e8a897d74029bbdeced1c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:48:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7269c269fc14311babec1c562d401e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69575432ec2345a2b1f85fcc411980f2 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40c924d1dbd84d0db899d8e6093e52eb : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e4ab0287352a623e275fe703ed76305e1fbfd82","revisions":{"2e4ab0287352a623e275fe703ed76305e1fbfd82":{"kind":"REWORK","_number":1,"created":"2023-01-16 18:15:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/64/870664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/870664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/870664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/870664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/870664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/870664/1"}}},"commit":{"parents":[{"commit":"b5b8d5198e2e8f68dc8f6c4947caede64149d202","subject":"CI: do not check for error in binary images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5b8d5198e2e8f68dc8f6c4947caede64149d202"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-01-16 18:14:26.000000000","tz":180},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/870540\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n(cherry picked from commit a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e4ab0287352a623e275fe703ed76305e1fbfd82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e4ab0287352a623e275fe703ed76305e1fbfd82"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
