)]}'
{"id":"openstack%2Fkolla~870665","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Iad83132b61efadbf09aa9aa2edf96235085764c6","project":"openstack/kolla","branch":"stable/yoga","topic":"bug/1999778","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-01-18 19:25:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad83132b61efadbf09aa9aa2edf96235085764c6","subject":"relay non-zero exit codes when creating openvswitch bridges","status":"MERGED","created":"2023-01-16 18:15:35.000000000","updated":"2023-01-18 19:26:47.000000000","submitted":"2023-01-18 19:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870665-bug/1999778","meta_rev_id":"4cbe7a351354f3856b793c6fa841930021a82336","_number":870665,"virtual_id_number":870665,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 19:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-18 17:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-17 09:31:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-18 17:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 18:15:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"state":"CC"},{"updated":"2023-01-16 21:21:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 08:36:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-01-17 09:31:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-18 17:39:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"62b9180802e37d8b8dba689c1e9f2bf4aac4b129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-16 18:15:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eceb38c4f97953334b6477e273d0662bec8fcb13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:21:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75b32470dc924262a9b2227d1d859793\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/21d39cc14aa74b7b8b36c74a7176a5d3 : TIMED_OUT in 3h 01m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/efdf2e96fe204783b20f4a711a627cec : TIMED_OUT in 3h 01m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/005aaba426ff4a9da61ada56d5e6c643 : SUCCESS in 1h 01m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5502e846a2e845708cca9b62e896f440 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e81bf1d1aea6474ca02575e80a797ffb : SUCCESS in 1h 27m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/731396648f8d4fe181e14844bb3548f0 : SUCCESS in 22m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bbdcead5933a4834979d271c9a4b1e93 : SUCCESS in 20m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5c00f595f0bb4f6aacf045074dca245d : SUCCESS in 55m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5bb152a966694c21a92120f500452f07 : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17eb277ca95c4c72b366b54d1272be9e : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2c574c9dd824eb4abf533449c682b2f : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/046291fcc5a243b1b68ef31f0e1bf820 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c521882009742e4bcbe207b64aeaaa1 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34edaa173d4543dd8958fd5741ac90b0 : SUCCESS in 3m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9bc71c689ba34d498feffc3dc38dee1b : SUCCESS in 16m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/39a7241a72584f5baef87a2b8a40048c : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"69e1b5f31d3bae648e8381d51f976b3be2ca7961","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 00:35:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4528c14988c4c598c3e718e97eced89\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/adc7a455477a4fd08a2c55493b3d17c5 : SUCCESS in 2h 34m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab6c1d46a88949c0a25bf8aad2ed74fe : SUCCESS in 2h 46m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20f7a14281084f5ca7fcb2aa94824532 : SUCCESS in 2h 42m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df17aa836cec43bbb6df257509b22959 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37fbe561192e4e289c364b73dc4100fc : SUCCESS in 5m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/76c4d3b796774f0bb76d8a5a68df8455 : SUCCESS in 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8509faf72bf815734edf9a8a79558e64b772878c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-17 08:36:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80effd708899edf6b05150657833ed1f21aaaaa1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-17 09:31:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61f5a7fea201b01ad8cf31288eb1fa2cd582ec5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:28:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7484093ed949492d97391888f2ebe894\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a8b06059e34a46f99341539af405d221 : SUCCESS in 55m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d2b73c98e7c44e5ea3844299f65e0855 : SUCCESS in 1h 06m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8cd2f389b2404ccfb5d720191a4a1606 : SUCCESS in 1h 33m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0e8cc281e9c9457eb563e33386492d03 : SUCCESS in 1h 30m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/37d7ad0423264c8cbd842664057a3b47 : SUCCESS in 1h 50m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c6a30b9a5a67483e90c4853fbb9130ef : SUCCESS in 27m 32s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f4f5a05f11041bc86a9ecd3bc0b415d : SUCCESS in 17m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43e9d01422a046f084818d18783acaf5 : SUCCESS in 51m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b0d5a8a9d1c840a0bec802aa5ac9e827 : SUCCESS in 1h 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99560a47e1744d3ab43113e4e2d75dc2 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616983aef40f4ab186b836a0bb7bace3 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52fb9e5bbe0443c9981556f8c283ef2d : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0fc634e7a4455b93ec3982a0210f70 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf13d3d2c6f47b79dc2732b27ac85f9 : SUCCESS in 2m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f85015a034043b8bdd0b6ef5427a37b : SUCCESS in 17m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33b600ec11e44ccf935ee9c39dca1bb8 : SUCCESS in 23m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0300779fe1e9da615bf47dc21259af768e11e2ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 11:10:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8457182d957740f6bff18dbad1d3e919\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b86b32c9b0434924800bd73d62ebc93c : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/707d65c71c7843549f8aa9fe0a9b1eca : SUCCESS in 2h 00m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5abf7269557549a7a1292e758ac6641b : SUCCESS in 2h 30m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e32613c8983b4f54818f83513b4afd87 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26f42a58ad1f47c0a7e3393bd8d145bc : SUCCESS in 6m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2873d5442181498dbb7cdabef50ec9af : SUCCESS in 1h 44m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da8cd204aa5a8a760ab2d7fd5f085888503b92a9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-18 17:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8f11e4d8095b7f041c6e0188717d1bc5d9ea1b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 17:39:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3c1e3510a2b983cf7c671406dd59783f8b007ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:25:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d65d0e640b476696c1a1bfb44e3d7a\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7c6e7512ebce4db8b4b837fd5cec35c7 : SUCCESS in 28m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/73678a57b4154a5aad063f5c5d491c4e : SUCCESS in 1h 17m 17s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/01dd1042bb7e44aa9c7881da124613bb : SUCCESS in 1h 25m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0eebc2de1ca843789bd9e7438080bd91 : SUCCESS in 22m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/756177996fa9414ea192b69842a92f54 : SUCCESS in 1h 00m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/43e6ce33ccc7457094c775bfb3441651 : SUCCESS in 1h 09m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a944dda098641e9a7eede6a32e6a5e3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56138a426b74492cb57af6e9c462ed72 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9367b75036df4a20bef8c74dbe62b8fd : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88fd72e8c354b95b96ace4423c45cb1 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c49d4de481044f4b3435be30caf9327 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cda7ef5446bb454faa6cae209c605013 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8814fd5954bac4971c2cd8ed8807296d8d7d61c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4cbe7a351354f3856b793c6fa841930021a82336","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:26:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfce515ca8fc4484afda51fd3db1fec8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af9da5448c34d958244906971c13f67 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14a2595f75c74ba1bd144ab7b4203b22 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b930a682165a536ca4f9587534591f249cee79cd","revisions":{"b930a682165a536ca4f9587534591f249cee79cd":{"kind":"REWORK","_number":1,"created":"2023-01-16 18:15:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/65/870665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/870665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/870665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/870665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/870665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/870665/1"}}},"commit":{"parents":[{"commit":"8a9def6c8fca38979f578e0d21aedd534c4aab4e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a9def6c8fca38979f578e0d21aedd534c4aab4e"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-01-16 18:15:13.000000000","tz":180},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/870540\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n(cherry picked from commit a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b930a682165a536ca4f9587534591f249cee79cd"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
