)]}'
{"id":"openstack%2Fkolla~870666","triplet_id":"openstack%2Fkolla~stable%2Fxena~Iad83132b61efadbf09aa9aa2edf96235085764c6","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1999778","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-01-18 22:00:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad83132b61efadbf09aa9aa2edf96235085764c6","subject":"relay non-zero exit codes when creating openvswitch bridges","status":"MERGED","created":"2023-01-16 18:16:08.000000000","updated":"2023-01-18 22:01:35.000000000","submitted":"2023-01-18 22:00:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870666-bug/1999778","meta_rev_id":"56d8bedba493b1d90cbf5bdd1ee87b868b2dac16","_number":870666,"virtual_id_number":870666,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 22:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-17 08:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-17 09:31:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-17 09:31:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 18:16:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32814,"name":"Erik Panter","email":"e.panter@mittwald.de","username":"epanter"},"state":"CC"},{"updated":"2023-01-16 19:43:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 08:35:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-17 09:31:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e10f9ba075dda8cb1c2edec2ec7f07be9a441f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-16 18:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea0c9bc33b25423c790123362f1422d16c0ed7e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 19:43:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a43e35b1ceb450a8387a803dc77d730\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5dce1a0f859e49b0b33e7a1f9dcf17ce : SUCCESS in 26m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0bccc1dbdae84f388a16f550ef4ec06f : SUCCESS in 28m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/069b55d20a164f22ac743c746834958a : SUCCESS in 1h 17m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/38d1dfdfd50944ee95358e0eb9665b21 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f47cf55f49241da84fa1d260feb7ea7 : SUCCESS in 15m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83f6cb14f4fb420faf40df0c9c1a328a : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1dc3d40ff99b46528730a74283dc7b0c : SUCCESS in 57m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8473b5851d8349e9970d4199f5b4ed15 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f524dde58a147e18842335d9919c158 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7de00668a7c4c60840d47723e697352 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00c60e11ddd84083a9af0f870595512d : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdaf165d682540b2bda93f7ce3d2207f : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97f3e862425a47dea83cf1ce51baf59e : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2bbbf2d06f6436b896cba4f74b2b7c8 : SUCCESS in 20m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ac56c6039cc4444bc89485327f74440 : SUCCESS in 18m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ffe10c369de43e8adb973cde497a296 : SUCCESS in 47m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94cce984c53c4cefabe74ee15a10a2fb : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ecc81a305ba4964a72daf24895eb507 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3e5a0fffaf196824d48d7661b415ce94a70b523d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 00:47:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fbb0514ae4546049453229dfa7fb2dc\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/207da73fcbca4431a53bb5f2f1d0295b : SUCCESS in 2h 46m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d04e3fbb0b24b19a49eebed992db78b : SUCCESS in 2h 13m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8f3acbbbed64048b4ca715bdb065ea8 : SUCCESS in 2h 41m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28dfd22585204415bdc9ac1fc472625e : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6767e1dfe2054f579ac5f0ad6168acfc : SUCCESS in 5m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e183442a3b4a49c586cf2764d907526d : SUCCESS in 2h 17m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a76b2554dcda258dc006063808dd30d7b28df75","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-17 08:35:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b9bb795ce7e4132e225a9b1f56bc1898c5c65d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-17 09:31:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4dd734033a83bd808bbcc9fb49fa44162765815b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 09:32:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29b39ac6243e32e291c4c069e2da4fc94a2ed60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 10:52:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3c24a123354f5498099dc254365b63\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d3efa4c257884089a0304ebec7ca2e51 : FAILURE in 34m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a7bc61f43b4f4bf7b41bc4c89a07c609 : SUCCESS in 1h 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52d9e5b50d743199d3b9907dcad33f6 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ce0916e531649069c90926e2f302834 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aacc9af7f404a6b96a19fe8ec1e3254 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade9df291fa54d8d910785a9a1eb72dc : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe2164ac6a04bf7ac4ec7fd4f89f60f : SUCCESS in 3m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fe91a409ca343b9b96dc2e0be577ffb : SUCCESS in 24m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/707f0f73b0184840bfe72db71c06dd08 : SUCCESS in 50m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4885f61b45b8454aabd35f486d30d7ad : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"46fecffbb226f7d8c8c5873c177155aea0ce277c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-17 12:23:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"139050156481360bb9fc9ece0d619dbbed231fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4de228f1f4f64f8c9a7b056803c2cb19\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e05e7df52beb49efbdce9947725555ae : SUCCESS in 28m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2348f15959b14bc2b338ae0eb2e1ef59 : SUCCESS in 32m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c663bb1a2aa844eb9d80ac0fbf78b79e : SUCCESS in 1h 18m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d9e47d5bce54713888e65954c1f4bbe : SUCCESS in 1h 10m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/20270d94e4254e22a1fcc33f112a54e2 : SUCCESS in 13m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/38e36218371848208d5261404e63b1e1 : SUCCESS in 20m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/039976ea597f44e5a6e067c2ee4b8622 : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20e4167a00a400790abff2b415fbd56 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9283160499844b080ac182acb0cb0a3 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4228526124c34d66a20eabe9f955869a : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97c625ca50d44e1ebfbcc314d2577dcf : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5431cab8aa04b27b10ac2179f939c90 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25de41a3cc674c36a8d7608bf5b7f733 : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d16c6740fcd421084a87d68820243a3 : SUCCESS in 16m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/62f42385ec3345efbb30f82267ba91e4 : SUCCESS in 18m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/117f1ae080ee45058a55265e9fe42e1d : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c2f7c6a335f4c0faf087bcebe7043c0 : SUCCESS in 55m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51d238e420844fff9cd4014afdaad3fe : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0780c9ea5dc2d972a27ac21ebc7492ddbe58ba2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a564886a8cf3e7ed64f5589b7ff253661fb738d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 15:10:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/297e9e79d7714b86a60b8c0804db76c4\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/41e705fef89147179f95a2dc272ed517 : FAILURE in 29m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/77386cf18f7e4aacb25cfdc062b065cf : SUCCESS in 1h 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbc937d8a7940ba89b54f86d1e1fea2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e1fa1a0ab7e43959c6a6db54f5e57a4 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee870c567d64c158cc76e49d431fc97 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09f35b932e64da380433c4e5168ecf1 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/872db45f80c8415cb7a59eeb756e6b7d : SUCCESS in 3m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a54321d411484ed5813654111fd897eb : SUCCESS in 20m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91d7061c7dfd4fa4bbd024492dcfaa3e : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5243256e29424008821a057595502f90 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f611449cf314e0565d59df1156fa42da152169c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:44:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5fcfecce844eccb0734a7d7b7f4c6f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ec33089f7d348cd984da760838515db : TIMED_OUT in 3h 06m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec0075219cd5487985794acc55550303 : SUCCESS in 2h 38m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8c0005cdbbb4803869a6e9d7e4ed2da : SUCCESS in 2h 58m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfdc4ba866a941b0a16ca934f382396c : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4104c4a942e54f6398e78004b2058d8d : SUCCESS in 5m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e36a94c1217c4aefa0d736340c1647da : SUCCESS in 2h 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9917ccf371e6052a9d35995413419f467868efb8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-18 17:39:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02f17b46af099ee1fcf9509c2917ab2db9f2389a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 20:47:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b830be4419455bb5b3566d348b12be\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/924f770e3532471885d8cface1df84d3 : TIMED_OUT in 3h 01m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/21326335884d41dc9a11a929de6ccec0 : SUCCESS in 28m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f48666243f8b4dfc88f480290713c4bf : SUCCESS in 1h 09m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7185c1e3d5ba416c8f462096a109babe : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0e05b2dc2208485ca4752833019201e9 : SUCCESS in 15m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/03da55cac2dc460699077c15690b9ec1 : SUCCESS in 16m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/02c5998536ac412b9c07e1b94c2dc265 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2253459b38b64e10a01409c3c41e9ef5 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2eb089c58ef4e26a7663cd2f22ef946 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe63d5ce25e4a81b1fefd95253447c2 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae951ad2212c474398c5f714519f9043 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938353c018f544c292cc6e9d2292a02f : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b264b86c767747e784e52d04e90946be : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f41e6278c094503836010f9dfb58558 : SUCCESS in 17m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/663db7a454184123aaecde05fc7099a2 : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b296cfb549441eda2922da8cc4aef2c : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d726105ab1a44e3aa12956b1abef931 : SUCCESS in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67e181b186574aad96aa8529629baf0d : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"037a701c0e14e0402d940a6fd38dc6561f170182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 20:47:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddb1c8821d5dd08df31e69d4db1244332c0974f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 20:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2aa3fe97664c0db3108173c983f17e\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ab4d1fb93294f9484f7a2e23779ade3 : SUCCESS in 2h 52m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70faaced59f84b5e9b8a7143c5fd4677 : SUCCESS in 2h 41m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7144ae580504ea9a2047d96e14c0e17 : SUCCESS in 2h 53m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0caf307805a64e74aec4d26c1f494944 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb6a532ddecf4b579cfeb300dc7308da : SUCCESS in 6m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6c486f5609c4ab7bfa20c0425d99137 : SUCCESS in 2h 46m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c587d2c6952a12eb49df258c184aafe3f97b7fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 22:00:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb6bd8e66394dcf9e11a1c4e7988000\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3be047cf7b3a4a2e82e4aa78fe2941cf : SUCCESS in 29m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b2a7f779844b43738ce08bcf843e9037 : SUCCESS in 1h 08m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b6fc34983f4d97b286b1abe99d8154 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bbc7396070c4d4abef5e3cd4861b970 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d32fa2642bc466fa0d10e9bf21c4e83 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45b5f1c140a4a12959c3abb445e91c3 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88e4723e7d3d4a939df4e7cf30b07650 : SUCCESS in 3m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e19ed4281db42e9b8220c219eca9831 : SUCCESS in 19m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/537da094b5ce4a6494baad0930b99322 : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c4e5afdbc9a408080949a04383cf641 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2697803ad260b3749dcb6704b4b61ebe78f092b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 22:00:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56d8bedba493b1d90cbf5bdd1ee87b868b2dac16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 22:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8684548bd934fc38d81a9282f7fc42d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ab773d3c644c6c930d7047eeec52df : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3744a9d57d1649a5b6c0c2e985fe643e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c0e3d275d1d336bfa9d730e0a96ad84118bb39e","revisions":{"0c0e3d275d1d336bfa9d730e0a96ad84118bb39e":{"kind":"REWORK","_number":1,"created":"2023-01-16 18:16:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/870666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/870666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/870666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/870666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/870666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/870666/1"}}},"commit":{"parents":[{"commit":"00e50f58ae83a0acbf8f58f9814173aea3bf745e","subject":"CI: Fix tox 4 once again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00e50f58ae83a0acbf8f58f9814173aea3bf745e"}]}],"author":{"name":"Erik Panter","email":"e.panter@mittwald.de","date":"2022-12-15 12:59:56.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-01-16 18:15:47.000000000","tz":180},"subject":"relay non-zero exit codes when creating openvswitch bridges","message":"relay non-zero exit codes when creating openvswitch bridges\n\nCalls to `ovs-vsctl` in `ovs_ensure_configured.sh` did not get checked\nfor errors.\nThis can cause false success statuses when\nthe script is run by automation tools such as ansible.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/870540\n\nCloses-bug: #1999778\nChange-Id: Iad83132b61efadbf09aa9aa2edf96235085764c6\n(cherry picked from commit a840ac0d7c8777a8198ae40cbb79b65c5fffd7e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c0e3d275d1d336bfa9d730e0a96ad84118bb39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c0e3d275d1d336bfa9d730e0a96ad84118bb39e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
