)]}'
{"id":"openstack%2Fkolla~870897","triplet_id":"openstack%2Fkolla~stable%2Fzed~I08fe2827bc58baa95d396daf28de1079e3df4f77","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-18 15:05:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I08fe2827bc58baa95d396daf28de1079e3df4f77","subject":"grafana: remove SHA1 workaround","status":"MERGED","created":"2023-01-18 10:41:03.000000000","updated":"2023-01-18 15:06:15.000000000","submitted":"2023-01-18 15:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870897","cherry_pick_of_change":868996,"cherry_pick_of_patch_set":2,"meta_rev_id":"561cc65cc1207be53495a414197c7e24534786db","_number":870897,"virtual_id_number":870897,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 15:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-18 13:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-18 13:53:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-18 13:53:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 10:41:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-01-18 12:30:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 13:52:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-18 13:53:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"71f62513f9b927eda7fd56c1536e54784af0d409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-18 10:41:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a39bbda921cee82994064bef5c4be9e1292885f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:30:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be9369198a44d3db11dd9c96e1705fb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/519d07de683641159341394450550aba : SUCCESS in 21m 34s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/922d9015bb1444afb52ed494b0ded359 : SUCCESS in 1h 12m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/53e33ab42f924ba18be9028142fa5c34 : SUCCESS in 19m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab7ee5ea96f4438389a6cca8c15f358b : SUCCESS in 1h 02m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/31710781e83a4f1ab3b106753fbea18b : SUCCESS in 1h 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7cf7be8f3ce429ea3cda8c309f34625 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291d7e4d05964148b410797c01d40755 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d18fca538f6470cb06077a5c8e92821 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/247a7f560b1c4b6baf0464c60fc415fb : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083ae55008624632a204c02e98197b5b : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/86120e4a5e6740389f1aee0062cd7f66 : SUCCESS in 21m 47s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ea15c9aafae0406a881cca38d63061db : SUCCESS in 1h 06m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c50fe8bc4c444c92bd0f802dd6469288 : SUCCESS in 16m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/655a0dab55a74bbca4fdbf4b0ceaadba : SUCCESS in 46m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e1cb1e7b04ce4d99b6fce7190cc8d559 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a67879da6f616e66a64784bf04d9e4ebb8f27deb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-18 13:52:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"277c89dffe17c4792e8a3b6c876d5112c0a5e87d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-18 13:53:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81e4aa9973183985f57cf34d0b4eb9de8b3f96e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 13:53:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/521def34e04343baa2cc62d0a29e526e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/274e488cb13c4d0ab32ce9ef7250e4cf : SUCCESS in 2h 02m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77f029d05cc8452cb19c518bb0c3c6d1 : SUCCESS in 2h 15m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f3fd81ab5d543beb003dc01e012dde1 : SUCCESS in 2h 27m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ae7a834b5dd4f21824528aaa027a996 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4e2574a05594c7ca83df3fa6fc8c065 : SUCCESS in 6m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4445c5fbe91e425b8259ac9d4e4ad23f : SUCCESS in 2h 11m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1efd333ec162473cb4191a4936a6f1a6 : SUCCESS in 2h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3f37a14733e730b280e790ff2d486ebb000cbdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 13:53:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54bd7f2746a1d97c4ee637e60112432593dc5f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:05:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972a90e50a304933878229da1e81d491\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/395ef90609004ea69074a3a476674fbe : SUCCESS in 19m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc4771b9958940e6986dc01947df89c9 : SUCCESS in 1h 01m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e648e52a28594a0caa45d495e812fd3b : SUCCESS in 1h 02m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3368fdf7c9f43f984aff00c5996e2f6 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b61423082342bd904b06137a48dc9b : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511098d2a62445eea7f29c4d6539b6af : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b55bba382f74797ba66cdb2eb60fa9e : SUCCESS in 5m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/694be5ce63d641fdbc5342f9701dac7a : SUCCESS in 21m 46s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f52d2ecffd1348259884672e9b57a140 : SUCCESS in 1h 03m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ed63e552ba8d4c1892657e71c46fdaa8 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4889a9d6783747adb608dc2d08b97a9b : SUCCESS in 46m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/df9c53196e134ef28904e1440c83ab50 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7f596d46cc25944d2d600845b62c140eb6a5e92d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:05:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"561cc65cc1207be53495a414197c7e24534786db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8915e6a33cb4011b4f228b0c1206c5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba68d0d68d649108f5707e1c9421ff9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff29af45d70bedfd76b30f015e9e4c68ca89bf59","revisions":{"ff29af45d70bedfd76b30f015e9e4c68ca89bf59":{"kind":"REWORK","_number":1,"created":"2023-01-18 10:41:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/97/870897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/870897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/870897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/870897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/870897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/870897/1"}}},"commit":{"parents":[{"commit":"d34ae7edb17eda7e4a2b6c4e0989ce2cebf116a5","subject":"Merge \"relay non-zero exit codes when creating openvswitch bridges\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d34ae7edb17eda7e4a2b6c4e0989ce2cebf116a5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-02 11:14:12.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-01-18 10:41:03.000000000","tz":0},"subject":"grafana: remove SHA1 workaround","message":"grafana: remove SHA1 workaround\n\nChange-Id: I08fe2827bc58baa95d396daf28de1079e3df4f77\n(cherry picked from commit 47a9a818596b27f658fb7b40aac2130fd04b66b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff29af45d70bedfd76b30f015e9e4c68ca89bf59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff29af45d70bedfd76b30f015e9e4c68ca89bf59"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
