)]}'
{"id":"openstack%2Fkolla~871109","triplet_id":"openstack%2Fkolla~master~I9e8140037fa9c676e67e3c7a0434833099a7f593","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-20 08:39:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e8140037fa9c676e67e3c7a0434833099a7f593","subject":"CI: Rename jobs - drop source part","status":"MERGED","created":"2023-01-19 15:14:47.000000000","updated":"2023-01-20 08:41:04.000000000","submitted":"2023-01-20 08:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871109","meta_rev_id":"36cc83c68614bf23929261bfbc4911f4fd983175","_number":871109,"virtual_id_number":871109,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-20 08:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-01-19 18:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-01-20 07:15:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-01-20 07:15:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-01-19 18:26:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 15:15:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 15:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 18:26:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-01-20 07:15:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"ac696a2062694e6408cfe1417558c825d0d7ee30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 15:14:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"042563a071d32f5df7f9b7d14f7cef1295603c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:15:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-ubuntu-prometheus-efk not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu\n          - kolla-ansible-ubuntu-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-ubuntu-aarch64\n      gate:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu\n          - kolla-ansible-ubuntu-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-ubuntu-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-ubuntu-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-ubuntu\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-ubuntu-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-ubuntu-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-ubuntu-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-ubuntu-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-ubuntu-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-ubuntu-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-ubuntu-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e7b209e8d15edda0396eaac4ba1738dc22ebcb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:15:26.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-ubuntu-prometheus-efk not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu\n          - kolla-ansible-ubuntu-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-ubuntu-aarch64\n      gate:\n        jobs:\n          - kolla-build-ubuntu\n          - kolla-ansible-ubuntu\n          - kolla-ansible-ubuntu-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-ubuntu-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-ubuntu-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-ubuntu\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-ubuntu-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-ubuntu-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-ubuntu-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-ubuntu-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-ubuntu-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-ubuntu-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-ubuntu-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-ubuntu-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-ubuntu-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e509ea8e883374325fdf21c1fc8c77a22abef0ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 15:17:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3d7749b36031ae48c29c980185dbbbdd2a650eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:57:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbc8a04fbaf4140a2c21e3794d9e9b6\n\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7865db4f0f1a42ec9358e7c65dcacca1 : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/79b90fa5aaef4a1085196d02555957b6 : SUCCESS in 1h 01m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/816d3c9cda8f4b9c9755a7b22933b7cf : SUCCESS in 1h 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5faeb10598de4f7a81245ce305d19c86 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fb1c803a34b45dc949242d53b5dc620 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3d4bc0c02dc4e5eb3ba19b04c7d9942 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e9d1a48fe14bcc908a91fdc9b6816e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fccab6bc193244eca500e9a65aab2abb : SUCCESS in 4m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c2430a749f6446e1bc73801e2ff0e0ee : SUCCESS in 1h 11m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f7e067e3e0094e6a92e460f8d4bc7ac3 : SUCCESS in 53m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8fbe115168a84bfab574586c7ee43e94 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8e07419aaf20e3ec22a600c1c3191be2b2e3f4dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:24:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d62256ce6b04da883808481e46eda86\n\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d18f374f2e654129937687e023f38e71 : TIMED_OUT in 3h 03m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfacc5caa1a94734bf4a8656cf6b8081 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d3eef22b9d544bc8b232c44f34229044 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a662d1f6118b8411b5084b6c09b3d574f231b20c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-19 18:26:04.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"4eeaf954bd3cefc82d85fa86ec5d86abf8b07ef5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-20 07:15:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"52bdf6afc1a99152253239d9a0e45124cd5873f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 07:16:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e8b0472050c532f0aa95dd300be258f3f73255b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 08:39:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2e3773bfa39475f998f9514348d5723\n\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4b6f738a2ff34fee9952315045f7cd1d : SUCCESS in 59m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/34d121953617496da6585ea993e56520 : SUCCESS in 1h 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd79a876c724c1287cd950a3de961f9 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f21f37bbea4241a11686435e791d14 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a66afa0d7c864cf193ae54ed9c334eb1 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2864a801274040139d8212dfea80c33d : SUCCESS in 7m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0174140d217241428527b9db7c131f29 : SUCCESS in 1h 16m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e0d462b7add043d3a040a1e0e939903a : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/174d610322714066ba93e70848029b74 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1c4af3443d8247e47d3bb5ad3f4c4cd1276395f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 08:39:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36cc83c68614bf23929261bfbc4911f4fd983175","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 08:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458ef2306a7f4d89b66a73f485b5622f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6cd310a61444188d4c6879c61d621e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61594f653a962b59839d1da275a090d0974d69db","revisions":{"bee98470d0100d4e484537f008e0af361c54af19":{"kind":"REWORK","_number":1,"created":"2023-01-19 15:14:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/871109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/871109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/871109/1"}}},"commit":{"parents":[{"commit":"47a9a818596b27f658fb7b40aac2130fd04b66b6","subject":"grafana: remove SHA1 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47a9a818596b27f658fb7b40aac2130fd04b66b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:13:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:14:44.000000000","tz":60},"subject":"CI: Rename jobs - drop source part","message":"CI: Rename jobs - drop source part\n\nAlso stop setting base_distro for rocky - it\u0027s done on k-a side.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/871080\nChange-Id: I9e8140037fa9c676e67e3c7a0434833099a7f593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bee98470d0100d4e484537f008e0af361c54af19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bee98470d0100d4e484537f008e0af361c54af19"}]},"branch":"refs/heads/master"},"61594f653a962b59839d1da275a090d0974d69db":{"kind":"REWORK","_number":2,"created":"2023-01-19 15:17:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/871109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/871109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/871109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/871109/2"}}},"commit":{"parents":[{"commit":"47a9a818596b27f658fb7b40aac2130fd04b66b6","subject":"grafana: remove SHA1 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47a9a818596b27f658fb7b40aac2130fd04b66b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:13:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:16:34.000000000","tz":60},"subject":"CI: Rename jobs - drop source part","message":"CI: Rename jobs - drop source part\n\nAlso stop setting base_distro for rocky - it\u0027s done on k-a side.\nDropping prometheus-efk job as well.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/871080\nChange-Id: I9e8140037fa9c676e67e3c7a0434833099a7f593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
