)]}'
{"id":"openstack%2Fkolla~871110","triplet_id":"openstack%2Fkolla~master~I537edd2a4d5ad7107e6f40022ee99c0f72f18ab2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-01-20 23:56:49.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-20 23:56:49.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-20 17:34:45.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-01-20 11:42:41.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I537edd2a4d5ad7107e6f40022ee99c0f72f18ab2","subject":"CI: Add rocky9 experimental jobs","status":"MERGED","created":"2023-01-19 15:22:31.000000000","updated":"2023-01-20 23:58:33.000000000","submitted":"2023-01-20 23:56:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871110","meta_rev_id":"3c1fd07125e80d8120f22681f57ff4f76e605d0f","_number":871110,"virtual_id_number":871110,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-20 23:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-01-20 17:34:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-20 23:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-20 23:50:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 15:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 15:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-20 11:37:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-01-20 11:37:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-01-20 11:37:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"4c7b9bcdecc543afdf31ae5237a7d3751cc3f960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 15:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7eeda6caf3e05825fe8dbc0ba384cee1c1f82dc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:23:17.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-rocky9-upgrade-cephadm not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-ansible-rocky9\n      check-arm64:\n        jobs:\n          - kolla-build-rocky9-aarch64\n      gate:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-ansible-rocky9\n      periodic:\n        jobs:\n          - kolla-publish-rocky9-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-rocky9-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-rocky9\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-rocky9-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-rocky9-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-rocky9-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-rocky9-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-rocky9-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-rocky9-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-rocky9-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-rocky9-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-rocky9-prometheus-opensearch:\n              files: ^docker/(base|opensearch|fluentd|grafana|prometheus)/\n          - kolla-ansible-rocky9-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9b45fa04793d86b19034a951d5036e0f1bd6002c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:23:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-rocky9-upgrade-cephadm not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-ansible-rocky9\n      check-arm64:\n        jobs:\n          - kolla-build-rocky9-aarch64\n      gate:\n        jobs:\n          - kolla-build-rocky9\n          - kolla-ansible-rocky9\n      periodic:\n        jobs:\n          - kolla-publish-rocky9-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-rocky9-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-rocky9\n          # Test rabbitmq and mariadb in multinode ceph jobs.\n          - kolla-ansible-rocky9-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-rocky9-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-rocky9-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-rocky9-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-rocky9-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-rocky9-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-rocky9-masakari:\n              files: ^docker/(base|masakari|openstack-base)/\n          - kolla-ansible-rocky9-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-rocky9-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-rocky9-prometheus-opensearch:\n              files: ^docker/(base|opensearch|fluentd|grafana|prometheus)/\n          - kolla-ansible-rocky9-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"396c1b846c6a4ca854dd8c3adef8bc1c039d2e1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 15:29:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"65a69ccab80efda7da007256000943bffbb9bf35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:39:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56281b87a13c48d39da385c8e61e3754\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8532860dff2341fcab871c3681e9ce67 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331c905b4ab24a78992c3535ee04b3df : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/110cb92ac4154749b22a7f01730c20a7 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f22d23059b4d4912a7c5f25723b52c4f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be43042a18474a38ae25aa6045758dad : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9524fd6b539053a1815b2c439e8e3a2c9acda83f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:43:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73b811f4d704888b7b1f3bebedc5ea9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8b685a2341642bfae95b93618bd2765 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6220d7a575104c329ef59875adcf4686 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5a8da8c43b250e0dad7b0110fa4154a29f51fe0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 15:57:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31fbd0b4b200c9366a81509b4b26e4f54c3ff2fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 16:00:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13b3d2911fe4cbe60f6ab2258aa86340e4a02460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:19:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896a32d285444f6d9610720373fc22ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c95b7e979a4e77837584b66f7b5759 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc471dd4147942349d1574472a881217 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/076c0d6487524a9b867fff77a9640b5e : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c417354421334e8cbb486a4bc0f2493b : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba58ad936634b11a043821f0e7dfa73 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c6d9429443ec8a6ea5b3ca484b21535a8209760f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:25:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd47610badfc4a879ab59724324ccf8b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f014e7e24b5d412fab8ce8d36d9f1f4c : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc8a760c676049d6a80304abd2c64ebe : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f79425295d72ddca48a3558082d4d79789e5b653","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:10:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d5cd2fe1bf4523aaa7c815db127f32\n\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bdcb4a1a686146aab412ea1868d797a8 : SUCCESS in 1h 55m 23s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/6e26c64faacc410f9b1cad7011a3b298 : SUCCESS in 1h 31m 10s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6702a2b0cc4c41189672bbde33db0ff1 : FAILURE in 1h 03m 57s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/2c9101522e724fc3ad7b163c783aa8a7 : SUCCESS in 1h 28m 13s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e8faf13516e74f988da49844fe69b5c5 : SUCCESS in 34m 39s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2717f40b7d9b4c2595df7d62cc38efeb : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b0c054abb7244a2e8bd866d177cbf1da : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bc97fe1713284e97934e47b6731f650b : FAILURE in 34m 23s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dac710e13e0144d0909289d71e6556eb : SUCCESS in 37m 09s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/d6d2b6d4fce3485f88a417278854559f : SUCCESS in 52m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a89f2024eb9a4dd29f0d5e1013e3b053 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e55299317e1a8e56f27a47c8292431a7f610798","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-19 18:11:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"964dac609c87a35fd492f6934cfb659f44bdaf37","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 20:03:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c271a6284c964256b98c8ad8aaac4d11\n\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8eb4afe75d6f4ef4a46b05aa9991a078 : SUCCESS in 1h 46m 40s (non-voting)\n- kolla-ansible-rocky9-zun https://zuul.opendev.org/t/openstack/build/56eda026f6754870822f9c191a58e8ac : FAILURE in 1h 04m 29s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d78656c55e6d4ffba0bf8d31964525dd : FAILURE in 52m 28s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e5502cb8ffb64fb8b966c12c04376b3b : SUCCESS in 1h 27m 02s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2a43f7ae58d6406dac706dbd884af810 : SUCCESS in 31m 32s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/8f0f9ca3d84a434fa8c0fc0aed6a7838 : SUCCESS in 1h 12m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/262c2985554143648fa5dfd837e9c719 : SUCCESS in 1h 13m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/dda55fd97f75491481c0e911a660b60f : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/41b82bbb80f048308815d95a0b95bd6a : SUCCESS in 33m 22s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/ea755000b2ed4441b210b731944b0bb0 : SUCCESS in 1h 13m 22s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fffe131934b44f80aae9f9024726ff21 : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e24732cb57e86d636d20df748d094ae1f6c94372","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-20 11:42:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6ee4b11a028f22ca12cf43ef50153664d122a3ce","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-20 15:09:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c9cbb8d6c9f68b01d38da8a9a71bb261491e1a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-20 15:45:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6c41fb9e206d23c80c51aaa214a292f1da86eaf5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-20 15:45:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4fefbeb396bfb4ce1983252513efdba313d0440","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 16:07:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9d381ddcdf43cd80dad04d766ca87f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ea0b06144bd4cf4a0b3257f5b3c666f : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/090213b4938d43a8b8fc5c78c78d846d : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f8bbcab0679a859299c7221ac6c24b59b739f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 16:13:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a4f9dd3f61343dc9af62bbb83809cb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17817626adaf4cc6b51c2ae6f6f91fc1 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62d914765c1b468592ab257e8428edb2 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32b26abb99664327b110a5daa4e425cc : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c504bb1d7840319b73017d32c128d5 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236e80167133476d8829f2acc7cc5aae : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"46d58e07e9fb92c0cb996209af205fb446ec55c1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-20 17:34:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6c68610a2c9771573a0dcf171a95635a6bc05c48","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-20 23:50:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c6aacdf304e42e469983a4c2c74be532ddd172d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 23:51:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d1e8afb89f822fd774b8ed578802ba29d10c056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 23:56:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3f2cea68204e279742c73394b27823\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7acec1a019c4b64903a9da61d302b30 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a97d9d3b5054ecfa2ca719c45e9be3b : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b0b293948db434b9b7602b5fd43e6bb : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27e843a198d480c94a8af7a203916c2 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"f34f86f2fa5bfb2153480b1aebfc767e65ece8cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 23:56:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3c1fd07125e80d8120f22681f57ff4f76e605d0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 23:58:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc60fbef0dba430199c72383b42acfc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808c9656447c42a5b32956e6b539e31f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a16331de20ce9f6c0554419d06f4dc1a40bb9d86","revisions":{"6106a2e16f91788eb854e31ced78ef2f4c2ff5b1":{"kind":"REWORK","_number":1,"created":"2023-01-19 15:22:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/871110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/871110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/871110/1"}}},"commit":{"parents":[{"commit":"61594f653a962b59839d1da275a090d0974d69db","subject":"CI: Rename jobs - drop source part","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:22:17.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:22:17.000000000","tz":60},"subject":"CI: Add rocky9 experimental jobs","message":"CI: Add rocky9 experimental jobs\n\nChange-Id: I537edd2a4d5ad7107e6f40022ee99c0f72f18ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6106a2e16f91788eb854e31ced78ef2f4c2ff5b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6106a2e16f91788eb854e31ced78ef2f4c2ff5b1"}]},"branch":"refs/heads/master"},"373117afc5afac08cbf2f5043caa4f83340ac698":{"kind":"REWORK","_number":2,"created":"2023-01-19 15:29:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/871110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/871110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/871110/2"}}},"commit":{"parents":[{"commit":"61594f653a962b59839d1da275a090d0974d69db","subject":"CI: Rename jobs - drop source part","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:22:17.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:29:02.000000000","tz":0},"subject":"CI: Add rocky9 experimental jobs","message":"CI: Add rocky9 experimental jobs\n\nChange-Id: I537edd2a4d5ad7107e6f40022ee99c0f72f18ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/373117afc5afac08cbf2f5043caa4f83340ac698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/373117afc5afac08cbf2f5043caa4f83340ac698"}]},"branch":"refs/heads/master"},"a16331de20ce9f6c0554419d06f4dc1a40bb9d86":{"kind":"REWORK","_number":3,"created":"2023-01-20 15:45:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/871110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/871110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/871110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/871110/3"}}},"commit":{"parents":[{"commit":"61594f653a962b59839d1da275a090d0974d69db","subject":"CI: Rename jobs - drop source part","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-19 15:22:17.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-20 15:45:21.000000000","tz":0},"subject":"CI: Add rocky9 experimental jobs","message":"CI: Add rocky9 experimental jobs\n\nChange-Id: I537edd2a4d5ad7107e6f40022ee99c0f72f18ab2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
