)]}'
{"id":"openstack%2Fkolla~871348","triplet_id":"openstack%2Fkolla~master~I9795fca15aa3e9d59eb686ff749245bf618639cf","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-31 05:42:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9795fca15aa3e9d59eb686ff749245bf618639cf","subject":"sahara: install netcat-openbsd on Debian/Ubuntu","status":"MERGED","created":"2023-01-20 18:09:19.000000000","updated":"2023-01-31 05:43:24.000000000","submitted":"2023-01-31 05:42:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871348","meta_rev_id":"59a003f7c784b47543c974456706b0d579817de4","_number":871348,"virtual_id_number":871348,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 05:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-23 13:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-01-25 14:21:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-01-31 00:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-01-31 00:59:54.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 19:42:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 13:26:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-25 14:21:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-01-31 00:59:54.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"8f44a0eedce1fe3ce818ac489bef5e67373709be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-20 18:09:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a4c0dc37d9b9f5e05b3543ca3ad506e8061a60b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 19:42:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f148f344705f4130a9db5a14a2b6e5b5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1470526983f24ef69b3c516beae3219b : SUCCESS in 19m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d7e135d6f4b346f0b5673596e483aadb : SUCCESS in 1h 19m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb55486ec66f4b3f9b5a7392e71513f6 : SUCCESS in 19m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b99a392d81f0411bb257141e384d7ce3 : SUCCESS in 57m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6387445900eb41b58fe58ba79ea1b6cd : SUCCESS in 1h 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff65dc9be4441188694fcf64ca1f5cd : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f72c4f354c547db80bd6796f7dbb646 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/019c3d26ea174a74ac58131c440dd9c7 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e233b19abf4da4bf78a6eea0fb002a : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077149bfd14d447693d925c1c6028cec : SUCCESS in 4m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85bf37ab39314fa1a4c748f446b48835 : SUCCESS in 23m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f53bdb490d234fafa7a43edaedd67df5 : SUCCESS in 1h 15m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3acbeff6ac054071a3507b89d9096d59 : SUCCESS in 18m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/131c35a872634a74a0d761b889d1882b : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/420f1fa1873946ba8d623bda491d7bc3 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"80169b842b9c8535c7e06f781728e43eace45a63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 20:25:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db920b33d9fc4b3fb99f9abb5d7eb940\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9cc3bde7d2d84fefbaaa5bef935193ef : SUCCESS in 1h 27m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e251aa1a289f4b2babfe474a6eb1f5f7 : SUCCESS in 2h 05m 08s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb77f222f99b43e1873368fffc57e2f9 : SUCCESS in 2h 02m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68b3302e046f40428ba02b58a391e035 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49f0167fc1ea4e0ba6922636b6702ffd : SUCCESS in 6m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/660e077987ad4431bd9132e7e7e0ecce : SUCCESS in 1h 32m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/84d1f4663ec54e219ecbbbfd080bc46f : SUCCESS in 1h 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07de4a055816ec4a2600dd89e43a0d92eb062702","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-23 13:26:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbbfbf9c3f4da4bb0bba07546813358d7e3cb6dd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-01-25 14:21:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dcf05cd5813640aca646121d149d3ed08892a6f6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 00:59:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a84488f42231b7e307a323369377535a7b463a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 01:00:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2576f140c0dca091b16214a17543fa16ff50d6ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:28:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1b90aa7966c4287bb8800e987ab240a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2d5d9e8d6769435ca065d222b95b0cdf : SUCCESS in 11m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1e7fd6c041644ca989fd958da91f9ab4 : FAILURE in 24m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e84b21d3b72c401bb27b5e98634795e0 : SUCCESS in 1h 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c85e17065d3497ab129522d3ddcd1bd : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89cfa68e60474acdbb5240cea086ca8d : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7a219224f2d45448963016710906803 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f68f0fe5b9647c2b2464d72e8852622 : SUCCESS in 4m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9ca74dbd8f674f6d89920e7217c0d2ab : SUCCESS in 22m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dc0d2e84986c45b5a762dd7fb65ddd88 : SUCCESS in 1h 10m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/737e6f4aef154101a00e57c67a32bc5c : SUCCESS in 12m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e2392668284b4d58b9263872150d22f5 : SUCCESS in 49m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cbcbce1344744a1ebd31e643bbbafe70 : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0b3cec6f590bd2c812ca14e00b17cdbbaaa83065","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-31 02:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31d0c2d5982a96948aeaeecf003fcc9e7f057e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 04:07:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dfba0580b334db5ba7e4f24646311ae\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e0404d2177f247a898b3d3146a94350b : SUCCESS in 25m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/26cf8faefc0a423f860a2417b1f630be : SUCCESS in 1h 17m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d244e179bd734f69964e32a8f3b5d396 : SUCCESS in 18m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ccac5cc8937c4f23bbff23ceddb10b82 : SUCCESS in 1h 04m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/021fb75695cc4298b55f85c49f57337a : SUCCESS in 1h 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27795f3915bd4f2b9b74ba3695840fed : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7944b374f72f4a9288e9ad8a54ae8b9c : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6ed938bfb9c4c95852d5886d7155f8f : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ae30b31b4fe4f8b903d611cb2584e76 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d89dd5f668422fb6186e4e9683d6f8 : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8947b7a3d1d240038780a146e8d7322e : SUCCESS in 22m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea4316afe13f440b9f216c837725b028 : SUCCESS in 1h 15m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/38145a69f78c480fbf7a59773febea02 : SUCCESS in 18m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d88f393a251a45d2b06c3634ce7a03c0 : SUCCESS in 58m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/129b128b59ef493ea011948cd75693dc : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4683eb46967f1fba6419b196747b21edf11500c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 04:07:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f9e1cb1a53ebe5537ccf0a0214c5230a79e84ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 05:19:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840d60c489804117863154705c47ff93\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/859d694025c44237ae7d42c9b866b42d : SUCCESS in 26m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aec2e6f16f174117a01f890bdc5b3e1d : SUCCESS in 25m 36s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/82681f641b694a57ba5b4f104c8933f4 : SUCCESS in 2h 27m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94f6fa5793854fdf98606e2817714ad8 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a71945b993334dc6bf58e34e0204b558 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a271ff35951449c59e25e379901e65c7 : FAILURE in 2h 02m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/02ca276a03f846aa9be215b742f31169 : SUCCESS in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"813946884bb774811487c4857ae9eb05dcefc701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 05:42:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911517e223244e07b6fbbc7e3d535ab5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b439964faaa4b03810344ed7bff67fd : SUCCESS in 25m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1cb6afca294c4cc8ae026a77b8645010 : SUCCESS in 1h 21m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/24ac0dec7a294271ba6b15afe1f4ae72 : SUCCESS in 1h 23m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e50eb88a334e49bd2ce4446341b609 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b8a9b1c048f4104a7ea01fce2d87e5b : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d8b3d8c64ec4a9bb563600662c601f9 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1c2f657e254a2395320d5cb77651c1 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/735657521f0a41ac9210a56a17f66844 : SUCCESS in 24m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/625aadc00541492da09bfc25334b6a31 : SUCCESS in 1h 15m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fe30020a365541cf94a9e9a3f7d62030 : SUCCESS in 16m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/efb9c7d4295b476c9fe0bd2ef3a6e8ca : SUCCESS in 58m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0509143981ae4ff4bc9a4fc97ee54595 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8d1ab1e8acb1faad9f09afc9a87f9e52c4f38596","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 05:42:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59a003f7c784b47543c974456706b0d579817de4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 05:43:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e32b8703a8d46ca8a039a6b63294c10\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9cfa3248074eb885262295fc89177b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42cf9c34ac6f6a242e8336ca37bdce7ecfcc00c5","revisions":{"42cf9c34ac6f6a242e8336ca37bdce7ecfcc00c5":{"kind":"REWORK","_number":1,"created":"2023-01-20 18:09:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/871348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/871348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/871348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/871348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/871348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/871348/1"}}},"commit":{"parents":[{"commit":"61594f653a962b59839d1da275a090d0974d69db","subject":"CI: Rename jobs - drop source part","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61594f653a962b59839d1da275a090d0974d69db"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-01-20 18:07:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2023-01-20 18:09:08.000000000","tz":60},"subject":"sahara: install netcat-openbsd on Debian/Ubuntu","message":"sahara: install netcat-openbsd on Debian/Ubuntu\n\nBy default in both Debian:11 and Ubuntu:22.04 \u0027netcat\u0027 means\n\u0027netcat-openbsd\u0027.\n\nDebian \u0027bookworm\u0027 12 will not have such default. So let install proper\npackage.\n\nChange-Id: I9795fca15aa3e9d59eb686ff749245bf618639cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42cf9c34ac6f6a242e8336ca37bdce7ecfcc00c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42cf9c34ac6f6a242e8336ca37bdce7ecfcc00c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
