)]}'
{"id":"openstack%2Fkolla~871738","triplet_id":"openstack%2Fkolla~master~I717b4f2215054ef9135a1ac252ed26820bdcb1aa","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-26 11:18:53.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2023-01-26 09:45:50.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}}},"hashtags":[],"change_id":"I717b4f2215054ef9135a1ac252ed26820bdcb1aa","subject":"Add util-linux to centos base packages","status":"MERGED","created":"2023-01-25 15:44:06.000000000","updated":"2023-01-26 11:19:51.000000000","submitted":"2023-01-26 11:18:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871738","meta_rev_id":"c62375c938a6f78beb21c249927e174d00cfbacb","_number":871738,"virtual_id_number":871738,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 11:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2023-01-26 09:45:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2023-01-26 09:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2023-01-26 09:45:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 16:02:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-01-25 18:00:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 18:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 09:41:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"451d53fb7652a341c639ec031fcfd093c395a0c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 15:44:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a87a6ae45d61abdc2b5ab8ff2c11c536d013b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 15:48:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90f5a2379726dabb907a743d30f3ade6e93ee79b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 15:50:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c42f19880ccef17b1d68d1291c1a9ae22607d9b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-25 16:02:03.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b8eaf47fa743c5a9d8138b37b6b5dfc8caa25fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 16:22:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8380070ebc130f95330d31a0f5a654d584507e3b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-25 16:33:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c01ec5ef77c14cb94fdba945eaf972261e853078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:00:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc724c0fd83f4133b645c738ad94ffd0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8e91c7458385434d847ce68e613da609 : SUCCESS in 16m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8bdc2eee49d44f5c957a96d0f20fb016 : SUCCESS in 15m 02s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d09518638284489e9fc6d1ddb2032bfe : SUCCESS in 1h 05m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4f02e7fb777475f9e2f2b4dce320e01 : SUCCESS in 2m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1dcbfd5ec5142a8a4b93fd83a646f95 : SUCCESS in 2m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b407cd7ffef54870b1b1f3118b6b07a4 : SUCCESS in 1h 35m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2f6f95b4d7984ec8bb7d5e0232481e6f : SUCCESS in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49b32712f5949a3d0aab32c35f8d7e0bcdbe8f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:02:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c3aff448b654c828ad5a985e739f579\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bce4bce1e9b64bc18a46d42c1b5bc72a : SUCCESS in 18m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d112ad2350b4cb18455ac9fdd4caed6 : SUCCESS in 1h 18m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f4ca2446599e4ada90f64e57279e2d74 : SUCCESS in 19m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/80c2e2ede6f94b958cf01b3cea143e0d : SUCCESS in 1h 06m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/15fb46ab3c4d4f3e99f9bc1a463390ac : SUCCESS in 1h 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf45aa41c4d46398125fb0c0faccfa2 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e489099655eb455b87e1d79c59efe6f6 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cabbdcff7da40dda168c8d497d8707a : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92abf81d3ba34a3086e1ce754a06e36f : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5774d1d1c3482c916bbd57104e4842 : SUCCESS in 5m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a61e241f07ce4a788f488bf620160c7d : SUCCESS in 22m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10377ce6b25b4f2cb7ccc933fc93da7d : SUCCESS in 58m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/122bbbca4d2d43bd87ae580247fc538a : SUCCESS in 27m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8492071527544fe78749313d20889020 : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b2560067a7e749598f680c1ca35f9525 : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fc1591e1bf464bb8585b513ed37bcb6836821051","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-26 09:41:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"44cf25531c4c5a292ca7aa1df31aed063711f131","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-26 09:45:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f354cce0a5ecca5c65e6b960266aa6a63580fc92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 09:46:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d1d95f28359692d459a61e5ca996b98cb0edd50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:18:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1839953282fc44c4a15171a89efd4d4d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba648320161e443e8dbc4bf743303321 : SUCCESS in 14m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cce95ac277104af89dd1040764949a60 : SUCCESS in 1h 16m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e51d63fc4ce464794b0128fb7115a20 : SUCCESS in 1h 08m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3683847f129f469e94cd3f6d711b8dfa : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/652877b679994c56bd6f07d0e108a093 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4aa1ca12f0134e4dadb88ca358748ec1 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060c79e4f1f042de908e4e718e3011be : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/73a6e36dd6c746cc8171092230308cf9 : SUCCESS in 20m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4de3733a48c84618820b186b31dba4fe : SUCCESS in 1h 08m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d6c2e55e5a654b4e80a6d1cb70273054 : SUCCESS in 25m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/01a382c8619f423e9463af84ef118ced : SUCCESS in 50m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/041314f22a784ba8a1e71bfc5f4eecd3 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":4},{"id":"888e95bb68a63086252d51ac003baf8d7e6131ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:18:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c62375c938a6f78beb21c249927e174d00cfbacb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:19:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06e9e4d185d14783a84b8d1425e2f927\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b6f7159b2f49e3a9176f55f04e1f68 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1371e4fc5b9d3d55f5bb48540753f740f1d1af02","revisions":{"49e8eb6cbee1a5eaac91cf389a554608f583e00f":{"kind":"REWORK","_number":1,"created":"2023-01-25 15:44:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/871738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/871738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/871738/1"}}},"commit":{"parents":[{"commit":"a16331de20ce9f6c0554419d06f4dc1a40bb9d86","subject":"CI: Add rocky9 experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:43:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:44:04.000000000","tz":60},"subject":"Add util-linux to centos base packages","message":"Add util-linux to centos base packages\n\nCurrently /usr/sbin/nologin is missing, because util-linux\nis not installed in the image.\n\nChange-Id: I717b4f2215054ef9135a1ac252ed26820bdcb1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49e8eb6cbee1a5eaac91cf389a554608f583e00f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49e8eb6cbee1a5eaac91cf389a554608f583e00f"}]},"branch":"refs/heads/master"},"13dff46838f8f26950a6ff6aaca71d68d9997944":{"kind":"REWORK","_number":2,"created":"2023-01-25 15:48:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/871738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/871738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/871738/2"}}},"commit":{"parents":[{"commit":"a16331de20ce9f6c0554419d06f4dc1a40bb9d86","subject":"CI: Add rocky9 experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:43:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:48:26.000000000","tz":60},"subject":"Add util-linux to centos base packages","message":"Add util-linux to centos base packages\n\nCurrently /usr/sbin/nologin is missing, because util-linux\nis not installed in the image.\n\nChange-Id: I717b4f2215054ef9135a1ac252ed26820bdcb1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13dff46838f8f26950a6ff6aaca71d68d9997944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13dff46838f8f26950a6ff6aaca71d68d9997944"}]},"branch":"refs/heads/master"},"54c90000638875cb6663243e78bc5a63f3c14ceb":{"kind":"REWORK","_number":3,"created":"2023-01-25 15:50:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/871738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/871738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/871738/3"}}},"commit":{"parents":[{"commit":"a16331de20ce9f6c0554419d06f4dc1a40bb9d86","subject":"CI: Add rocky9 experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:43:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:50:15.000000000","tz":60},"subject":"Add util-linux to centos base packages","message":"Add util-linux to centos base packages\n\nCurrently /usr/sbin/nologin is missing, because util-linux\nis not installed in the image.\n\nChange-Id: I717b4f2215054ef9135a1ac252ed26820bdcb1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54c90000638875cb6663243e78bc5a63f3c14ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54c90000638875cb6663243e78bc5a63f3c14ceb"}]},"branch":"refs/heads/master"},"1371e4fc5b9d3d55f5bb48540753f740f1d1af02":{"kind":"REWORK","_number":4,"created":"2023-01-25 16:22:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/871738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/871738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/871738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/871738/4"}}},"commit":{"parents":[{"commit":"a16331de20ce9f6c0554419d06f4dc1a40bb9d86","subject":"CI: Add rocky9 experimental jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a16331de20ce9f6c0554419d06f4dc1a40bb9d86"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:43:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 16:22:15.000000000","tz":0},"subject":"Add util-linux to centos base packages","message":"Add util-linux to centos base packages\n\nCurrently /usr/sbin/nologin is missing, because util-linux\nis not installed in the image.\n\nChange-Id: I717b4f2215054ef9135a1ac252ed26820bdcb1aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1371e4fc5b9d3d55f5bb48540753f740f1d1af02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1371e4fc5b9d3d55f5bb48540753f740f1d1af02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
