)]}'
{"id":"openstack%2Fkolla~871757","triplet_id":"openstack%2Fkolla~stable%2Fzed~I717b4f2215054ef9135a1ac252ed26820bdcb1aa","project":"openstack/kolla","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-26 14:13:59.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I717b4f2215054ef9135a1ac252ed26820bdcb1aa","subject":"Add util-linux to centos base packages","status":"MERGED","created":"2023-01-26 11:26:06.000000000","updated":"2023-01-26 15:53:38.000000000","submitted":"2023-01-26 15:52:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871757","cherry_pick_of_change":871738,"cherry_pick_of_patch_set":4,"meta_rev_id":"59b45cbb50acc1d46f4c39de41c2127deacb0c5f","_number":871757,"virtual_id_number":871757,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 15:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-26 14:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-26 11:30:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-26 14:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-26 11:30:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2023-01-26 12:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7ecb1afc8cb1e6ce1c3a810a44cef248fcb7768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 11:26:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b2f663eedfd46712ec9f9a537948257a750cd9d2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-26 11:30:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"753497e971cf0b02f78a983247f7e169f48caba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 12:45:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d759393f0954fc0ad3ae54be9ffb9bb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a84fc9f2c2204e2bb5d30b4d0da7c581 : SUCCESS in 27m 19s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/5f5632680f0541cf808ebfb7556e243d : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5728bad1cfe43318bfc1055fcd5996b : SUCCESS in 18m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b63823dc25142748978cb012c35420e : SUCCESS in 57m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0bdb2ed1421f4be994e32ea83a728193 : SUCCESS in 1h 08m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c63ea21cc94f5390d79185aeb4985a : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a8e09c174b4eb4bea3333d9a62cffe : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b49beeed6344b12bc50850ba65e3cac : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/415fd90b1730462b8938f1af32a257d5 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06340c76accf418495b026b896fd0765 : SUCCESS in 5m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5aa274c2b68c4c4ebe4b65f76db6e76c : SUCCESS in 22m 37s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f1171211799345b790eb1f54f55b681d : SUCCESS in 1h 08m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/10365a1d00e34a279771535c2dd48b40 : SUCCESS in 21m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7455de9899a54b6b9bfe07b30145c87d : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3521ce0b02104ce2aa2c74bbf312aec8 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":1},{"id":"735a3cbf4c19bc58e90e479078df53b5489fb742","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:50:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b999d491f7e41e0890534d182f16d8b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e8f6ca552bcb477599454f50a155690d : SUCCESS in 2h 09m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9a1cf231d40a4fc9964eed6976866a50 : SUCCESS in 2h 22m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a97f7f2ec9f54edfa23471606fcee6f5 : SUCCESS in 2h 08m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/310df81e306246b6b6dbd6aac62b594a : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a018be3cb4874ca4a6cc1c1b9e9a5589 : SUCCESS in 4m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/34290ca0c11740619eb3e2460ca22253 : SUCCESS in 2h 13m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/722bb112a2824130b04dbe3cbd82ae64 : SUCCESS in 2h 04m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c6ff01a34a24cbb48d75a8601c9bb89e92b6ca1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-26 14:13:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea90e1b9ff323749ac3d858084dbdbc6869ca0d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:14:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78d7bfe5b21b6be5ce56c9a926a586081315e7ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:52:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f787bb1f4a84922bc4ba4466d362a2f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/70b03c16db6242b5a348f153e705c2eb : SUCCESS in 16m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7613dd5e0ef4086bf7555d78213150c : SUCCESS in 55m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b8b5b5ff481f418b83fac0fcc3bca3b6 : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f325ca65464a1da90a32eb57025229 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/255c8ed6a99e44398b1ce918f7c3deb1 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70d4a01d72244539823be69d62a7a73e : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed83ad9f6ca41e6ae9b7d03270acaef : SUCCESS in 4m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/464b15098ca74e198375257ccb7ecd43 : SUCCESS in 32m 06s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/376a179467e54b57b8613fb92f4c7f01 : SUCCESS in 1h 06m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3615afa5dbc4f3ea81b6abd0698c681 : SUCCESS in 16m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df58b4ce029447419eda7f064083e002 : SUCCESS in 36m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e131976746b249649fd367db9571aeef : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d6643cd920aac5fe9c083ab5a0078fc322452094","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:52:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59b45cbb50acc1d46f4c39de41c2127deacb0c5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:53:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876f57c61dd648ca99200bd8a685c6bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13da76eb2bf44c1fbd87e25ca941f2da : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"055cbf6876216d63f3f3cb8f3460d2409f9855d5","revisions":{"055cbf6876216d63f3f3cb8f3460d2409f9855d5":{"kind":"REWORK","_number":1,"created":"2023-01-26 11:26:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/57/871757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/871757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/871757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/871757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/871757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/871757/1"}}},"commit":{"parents":[{"commit":"ff29af45d70bedfd76b30f015e9e4c68ca89bf59","subject":"grafana: remove SHA1 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff29af45d70bedfd76b30f015e9e4c68ca89bf59"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-25 15:43:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-26 11:26:06.000000000","tz":0},"subject":"Add util-linux to centos base packages","message":"Add util-linux to centos base packages\n\nCurrently /usr/sbin/nologin is missing, because util-linux\nis not installed in the image.\n\nChange-Id: I717b4f2215054ef9135a1ac252ed26820bdcb1aa\n(cherry picked from commit 1371e4fc5b9d3d55f5bb48540753f740f1d1af02)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/055cbf6876216d63f3f3cb8f3460d2409f9855d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/055cbf6876216d63f3f3cb8f3460d2409f9855d5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
